)]}'
{"id":"openstack%2Fdesignate~803241","triplet_id":"openstack%2Fdesignate~stable%2Frocky~If8738f5005e60cf46ed93edbefa272bc2611b53f","project":"openstack/designate","branch":"stable/rocky","topic":"cap-bandit-1.6.2","hashtags":[],"change_id":"If8738f5005e60cf46ed93edbefa272bc2611b53f","subject":"[stable-only] Cap bandit and fix lowwer-constraints","status":"ABANDONED","created":"2021-08-02 17:54:27.000000000","updated":"2022-09-24 05:53:31.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8efa358a23df5dc2f305c3e5c311556539e5a47f","_number":803241,"virtual_id_number":803241,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-02 19:18:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-02 19:18:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"00f3f3134a3049cefca9d1e217fbcca1b579c178","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-02 17:54:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12e926fdda1aebbc554922db92487d895611eead","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 19:18:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/55945dbc483c492a8d530913be9120e3 : SUCCESS in 43m 40s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/358b95d1427744dba6759920f4c9103a : SUCCESS in 45m 30s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/4603cbb336d54610996f96d460b2fe32 : TIMED_OUT in 1h 13m 59s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/04fc8b3eb28146ea91625adb81d807d8 : TIMED_OUT in 1h 14m 35s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/3c93cc6b05b94843b291177124673481 : FAILURE in 11m 31s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/35ad3ec2f7f6416aac0a2fd3710740e7 : FAILURE in 11m 44s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5f905cfd0d1d4da9b1f9bf54dce83b1b : FAILURE in 15m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/772967178b2649a59840221592ab0a8c : FAILURE in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d43986d5ac1f4b6f8ca9dfe37fa0b851 : SUCCESS in 8m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/55c35eb8234e4b5a871c1d057a6724d9 : SUCCESS in 7m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/66296739aeb44c0eb18a05a64742617b : FAILURE in 7m 38s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/2375212810fb44abb38c200d16fbf144 : SUCCESS in 6m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/71b3c502582a4f4a81755e21dfad963d : FAILURE in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b9be1867f3b4db199e06c7d61a04570 : SUCCESS in 6m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03665ddaf4f34db8991177bcbe34c233 : SUCCESS in 7m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d51ad3ad20f24bfc9fc37582f63ad2a9 : FAILURE in 9m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3815590ef40b34f5c0819c75758386e5d3f2523","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-05 15:27:38.000000000","message":"Patch Set 1:\n\nrecheck - devstack-gate based jobs are fixed with Ia62245bbeaf4f1a4b50c07efe14b58b550a9817f","accounts_in_message":[],"_revision_number":1},{"id":"eed4f88005680b9f983d4e2adcf93c4ec69c952b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 16:45:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/356ff1e0471146e6a76837ccd359e724 : SUCCESS in 41m 06s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/85d05f3ce14943178dd29a73535504d9 : SUCCESS in 41m 34s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/fc4165070ecb46fb878432f421908b35 : TIMED_OUT in 1h 15m 18s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/f8a91a336464431588dfa3c0e44cea7c : TIMED_OUT in 1h 15m 05s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/17f34857fe0d4a5ea3b03c99defd7a3f : SUCCESS in 39m 26s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/451dced35eea40ab829c80f29497a446 : SUCCESS in 36m 48s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9b83b1bcd8e44170a790e81c6d3a7fd6 : SUCCESS in 51m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5048cf9157da44d0a4481593b237aa49 : FAILURE in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a21efd5da0c04b6680e79909cd9f8fd2 : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d602d6bf1811490ca0b7514c91196fdf : SUCCESS in 5m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c0ef6c7a663b4cc9a8d727898a0c4067 : FAILURE in 7m 25s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3fa09f8b3540403fa088acd3a0f222f2 : SUCCESS in 5m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/aae826b303dd4e3a8cbd12315ee1460b : FAILURE in 2m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64af4188f9e84f05b6741e31d887562f : SUCCESS in 4m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f1856040a3641389aa1b8b9d9178e8b : SUCCESS in 5m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48178274abcc4fdf81f2eaccd92e0f80 : FAILURE in 6m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8efa358a23df5dc2f305c3e5c311556539e5a47f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-09-24 05:53:31.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0672ec0612bf534c3d97447327abc63f8522db21","revisions":{"0672ec0612bf534c3d97447327abc63f8522db21":{"kind":"REWORK","_number":1,"created":"2021-08-02 17:54:27.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/41/803241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/41/803241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/41/803241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/41/803241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/41/803241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/41/803241/1"}}},"commit":{"parents":[{"commit":"e4ee680f24d00b9633284c061edebaf0f62d7b3c","subject":"Worker should send NOTIFY also to all servers in \u0027also_notifies\u0027 pool settings.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e4ee680f24d00b9633284c061edebaf0f62d7b3c"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-15 18:32:16.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-08-02 17:53:23.000000000","tz":120},"subject":"[stable-only] Cap bandit and fix lowwer-constraints","message":"[stable-only] Cap bandit and fix lowwer-constraints\n\nbandit 1.6.3 [1] release has dropped support for py2 [2] but the\nrelease is faulty and pip still picks it up for py2 [3][4], so cap to\n1.6.2 when using py2.\n\nWith the new pip dependency resolver (introduced in pip 20.3) the\nlower-constraints job started to fail. Problem is here with the\n\u0027install_command\u0027 in tox.ini, which uses both the upper- and lower-\nconstraints files, causing the job to fail. This patch adds separate\ninstall_command without the upper constraints, so that only the\nlower-constraints.txt is used.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n[3] https://github.com/PyCQA/bandit/issues/663\n[4] https://github.com/PyCQA/bandit/issues/665\n\nChange-Id: If8738f5005e60cf46ed93edbefa272bc2611b53f\n(cherry picked from commit 86544150913bcd819b4ee8d77e83d2566142fb41)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0672ec0612bf534c3d97447327abc63f8522db21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0672ec0612bf534c3d97447327abc63f8522db21"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
