)]}'
{"id":"openstack%2Fdesignate~807671","triplet_id":"openstack%2Fdesignate~master~I54f6b49554833a2c7feef1865c2cdadcc0b3b014","project":"openstack/designate","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I54f6b49554833a2c7feef1865c2cdadcc0b3b014","subject":"Imported Translations from Zanata","status":"MERGED","created":"2021-09-07 06:12:56.000000000","updated":"2021-09-07 19:22:14.000000000","submitted":"2021-09-07 19:20:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807671-zanata/translations","meta_rev_id":"c94124d9265458462d086ab9910bdbe148544232","_number":807671,"virtual_id_number":807671,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-07 19:20:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2021-09-07 14:26:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-07 18:08:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-07 18:08:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-07 07:47:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-07 14:26:23.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-09-07 18:08:34.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"6cfb7450e4f9c951e66aecf7c7d32551bb9b268f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-07 06:12:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee844a4df7515af4b867529690a093bb7eb89106","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 07:47:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5bd8fee9eefa46089b9b9596f0a85eef : SUCCESS in 41m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/838226366df44001be905a7a631963bb : SUCCESS in 48m 49s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/855dd062ba99497f86d56043f57d5668 : SUCCESS in 1h 10m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3abcb1fc983845bda8072c1492fa058b : SUCCESS in 42m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/592ad592750245f581b74b809e8cc15f : FAILURE in 41m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e10720102c504222a4d692e28dc7b018 : SUCCESS in 6m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7fcd93242de47b4a93859601c16fd29 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49aa061dbc274de7a9c78f3e0c1a6594 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfd9ca1e11d0492c98835cd8819569c8 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d29de69358846f0aaef52249457f954 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94ca0f22188b4a91b18531a16ed18c33 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbe126069e504ee3893c66977ccdd2c2 : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e312c3afc9b487f9d13bb4f8fe1ad9b : SUCCESS in 5m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/21d1621d76c94a7fa1b54afd8f1a8f6c : SUCCESS in 34m 48s","accounts_in_message":[],"_revision_number":1},{"id":"621109b99704154917b169c6440ceece3db8fb47","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-09-07 14:26:23.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"36effbf4948e05d0a023358fa3c864a0351a6db0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 15:34:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b8aebfd47a9547ba85692b6e8e7a516f : SUCCESS in 41m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3ceb7f8fd4c641099d2c8697a801179f : SUCCESS in 38m 21s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f70fa5112ae9481687a422cdd969ff01 : FAILURE in 1h 00m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d3fc2d8f556c4b0c8c1970e526416614 : SUCCESS in 41m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8687acaba2cb425486093e013733981d : SUCCESS in 35m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0acd9b2611bc4a5d982de69393eb0ffe : SUCCESS in 6m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f0ac26ba3704376891cd1a356076e37 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bc09f7352f840869dafd2003204ff4c : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f446798a4e1e44d987935bdf4b4300fb : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bbeddfd20f9400186f3805498dccf12 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9090c5f44e844b5882d4478208cc117 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fec172d4d2ce40ce9ca0f20a14de3274 : SUCCESS in 9m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27fd11c2278f4021a9b6b47be4bf456d : SUCCESS in 4m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/15747c660cc14fb8b2888690c79e02a2 : SUCCESS in 28m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6a99bf0d2e8b6c4b222eb1ee13e571ecc4403336","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-09-07 16:31:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"199fff003945334210c42bbfdcf64b342d99ec2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 17:31:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c26c3f2319de4006952f77c48a04e760 : SUCCESS in 33m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/aeb766b1094b47b6b09d91b49cb8e0e6 : SUCCESS in 35m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1f58c462cf8547118d632de165a4d495 : SUCCESS in 58m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1a7818b325ee4f18a2ca929f1f682e25 : SUCCESS in 34m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/11e6f6e496a04c23b65b1c45468f20e5 : SUCCESS in 35m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a15ce8eec70c4cbfb257b86962f8dc0a : SUCCESS in 6m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e7d78207caa48f390e87b098f890435 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7b0f8ab1a3048a09a33558a0402bfcd : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8acdee4df79c44f18ca4dcb442c81b5e : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64a56360369b49f8a2967d5ca9ef125c : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c9c7794193d4222a1079207c31db199 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/990d3e67df0c4e18a54187a1a0019071 : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/911031b740e3414f966390004423292b : SUCCESS in 5m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/31626e1fd27842498b0f447a0717ba39 : SUCCESS in 28m 34s","accounts_in_message":[],"_revision_number":1},{"id":"eb1b0066a6268a4c99013878407151d86b620b90","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-09-07 18:08:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"559a3b8ae2114bf755df603aee2e3a1019e9e84c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 18:09:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1978b64cb8037044012bbfbe67dd52f646b5982f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 19:20:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/261bbfb4977f4e46a2b7f16a65d9c539 : SUCCESS in 36m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/99f9f86bfc8e4e65a0a60c2def9b91b0 : SUCCESS in 45m 32s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ba4ea4a6764a4f88a4c4860853ae9155 : SUCCESS in 1h 07m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/893e4596c06742bba6781675abf1c51e : SUCCESS in 45m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/63d38a950b7e43b1abb84b808aaf7c6f : SUCCESS in 41m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f0cfdb2e30940679e4451843582cf37 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a5c60b4aa4a443aa636f4f16799666d : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a46a260cbc04ef58d9baa558525202c : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f4ee38f6b0d4136916c587800ba213b : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cd91f156e164f3284e8b1881fc34a4d : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1affcd26ef5240c6904f448570de6746 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"c5793d22b0796089f9a7a565b72cd7ec8e6a4491","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 19:20:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c94124d9265458462d086ab9910bdbe148544232","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 19:22:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd7af9829fc24d31a7b9e029f289d3b8 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8d7b1b0645840bc8d078e8e647a812b : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"173578848a5bccf470e01e332038a44d1c7a4854","revisions":{"173578848a5bccf470e01e332038a44d1c7a4854":{"kind":"REWORK","_number":1,"created":"2021-09-07 06:12:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/807671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/807671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/807671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/807671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/807671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/807671/1"}}},"commit":{"parents":[{"commit":"76bb79dd0d151e0c7b9d820d8d93da51c32fd037","subject":"Merge \"CERT DNS records\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/76bb79dd0d151e0c7b9d820d8d93da51c32fd037"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-07 06:12:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-07 06:12:55.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I54f6b49554833a2c7feef1865c2cdadcc0b3b014\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/173578848a5bccf470e01e332038a44d1c7a4854"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/173578848a5bccf470e01e332038a44d1c7a4854"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
