)]}'
{"id":"openstack%2Fdesignate~807784","triplet_id":"openstack%2Fdesignate~master~Ic686d35c25436a1bca83c4bb9c62ebf28a445f28","project":"openstack/designate","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"Ic686d35c25436a1bca83c4bb9c62ebf28a445f28","subject":"Fix count_record policy to use the right deprecation variable","status":"MERGED","created":"2021-09-07 17:53:47.000000000","updated":"2021-09-17 02:12:03.000000000","submitted":"2021-09-17 02:10:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807784-secure-rbac","meta_rev_id":"15daf2800a04870b1f71eefff81fbf431e4d6c02","_number":807784,"virtual_id_number":807784,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-17 02:10:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2021-09-07 19:05:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2021-09-07 18:50:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-17 00:20:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-17 00:20:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-07 17:57:36.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-09-07 19:02:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-07 19:05:36.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2021-09-17 00:20:56.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"aaea62bbbd59501b182882b1b6fb5495ca4a4539","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-09-07 17:53:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"beb123200bcb318f07738c294dc81c2c2a7ecb76","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-09-07 17:54:22.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"f7d18aaeaa6a9b2e4d1480640dbd3b6a25621d63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-09-07 17:55:54.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"fe6cceb8e56bf22a668bde2d2c42c3ba37b1af47","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-09-07 18:50:01.000000000","message":"Patch Set 2: Code-Review+2\n\nYep, fixed recordsets, but not records.","accounts_in_message":[],"_revision_number":2},{"id":"52780baae00c94a0fdc1f963eefc2c70bab05b18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 19:02:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ef9102a320e34074845acf218fd66147 : SUCCESS in 41m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6b34fd893f3a44f69f925e5a319c7771 : SUCCESS in 36m 20s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/727d0a5066854682a1725ff25fd1b779 : SUCCESS in 1h 00m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/57a5071c8a49442ca60122b606c49ed4 : SUCCESS in 34m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/83795b2eb0384e8a91ed73dd672346c3 : SUCCESS in 35m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e874e3d4881e414082c65793f3991dc6 : SUCCESS in 5m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/267146f6d47146a99f35a4e8c0bbf87a : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/075529c7ba6e4abaad1cc3f5bab92ee6 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ac52d2a13964b02a986041db4ac2868 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7542ee2dda34941855088e66a0ce0a4 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/272d55ae032f4fe6a5845553b7d13eae : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/454fbbaf4a9e42bbb1aca92886b1674a : SUCCESS in 8m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d1f42a1bd1e44bf2ae00eecfb20c7361 : SUCCESS in 31m 33s","accounts_in_message":[],"_revision_number":2},{"id":"a0f83e5fb6e5ce6ed449074df6177c20f38d4247","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-09-07 19:05:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0e3d84426cae2e5ded7aa5029af09f808b743e2b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-09-17 00:20:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6b701641e419f521e72cd9d39b9d44893cce9824","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 00:21:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5d5da0b929e4d8b5f719184468517d0d25d261c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 02:10:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4e49d7398d3f48f986f084da57bda661 : SUCCESS in 35m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/991bfc9107294cfd834494360186b97d : SUCCESS in 35m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/df51a06cfdf341a9b76d63f68f4c1e47 : SUCCESS in 1h 01m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8fe17d06bbd44a2d898b77419c25e17c : SUCCESS in 41m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7a0ee87d62d64d62b396866f16211e36 : SUCCESS in 38m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de5cef96a9734c8d8735b7072b506975 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcf53c8a371b4cdca1efaef5092ed34b : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb87c3a37fe344189ec8fe99ebdfe3f9 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b225ab34cae646a481f30d9afd838c04 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d99873d179054a559850ca521c65bb33 : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":2},{"id":"a9d1da487bec95aebb5fc9116c59d54df80909b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 02:10:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"15daf2800a04870b1f71eefff81fbf431e4d6c02","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 02:12:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d2dbf53092f4df3a2e7f7a1d136ac2c : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f67ae19888c738aba4f829782a92018a8e475950","revisions":{"bfe68265cfddf4f6151a2b8790a1999e0ec18ebe":{"kind":"REWORK","_number":1,"created":"2021-09-07 17:53:47.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/84/807784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/84/807784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/84/807784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/84/807784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/84/807784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/84/807784/1"}}},"commit":{"parents":[{"commit":"246ec25d6ee19b407449296f125c00ee16fbf637","subject":"Fix typo in the find_pool deprecated rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/246ec25d6ee19b407449296f125c00ee16fbf637"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-09-07 17:51:00.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-09-07 17:51:00.000000000","tz":0},"subject":"Fix count_record policy to use the right deprecation variable","message":"Fix count_record policy to use the right deprecation variable\n\nThe cound record policy was referencing the wrong variable name for the\ndeprecated rule. This causes oslo.policy to generate a deprecation\nwarning that didn\u0027t make sense since it thought the policy name was\nchanging.\n\nThis commit updates the rule to use the proper deprecation.\n\nChange-Id: Ic686d35c25436a1bca83c4bb9c62ebf28a445f28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bfe68265cfddf4f6151a2b8790a1999e0ec18ebe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bfe68265cfddf4f6151a2b8790a1999e0ec18ebe"}]},"branch":"refs/heads/master"},"f67ae19888c738aba4f829782a92018a8e475950":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-09-07 17:55:54.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/84/807784/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/84/807784/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/84/807784/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/84/807784/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/84/807784/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/84/807784/2"}}},"commit":{"parents":[{"commit":"76bb79dd0d151e0c7b9d820d8d93da51c32fd037","subject":"Merge \"CERT DNS records\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/76bb79dd0d151e0c7b9d820d8d93da51c32fd037"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-09-07 17:51:00.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@redhat.com","date":"2021-09-07 17:55:54.000000000","tz":0},"subject":"Fix count_record policy to use the right deprecation variable","message":"Fix count_record policy to use the right deprecation variable\n\nThe cound record policy was referencing the wrong variable name for the\ndeprecated rule. This causes oslo.policy to generate a deprecation\nwarning that didn\u0027t make sense since it thought the policy name was\nchanging.\n\nThis commit updates the rule to use the proper deprecation.\n\nChange-Id: Ic686d35c25436a1bca83c4bb9c62ebf28a445f28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f67ae19888c738aba4f829782a92018a8e475950"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f67ae19888c738aba4f829782a92018a8e475950"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
