)]}'
{"id":"openstack%2Fdesignate~813176","triplet_id":"openstack%2Fdesignate~master~I0a1ffdc92c054183cb6e720eb53cc98f99bbc6ab","project":"openstack/designate","branch":"master","topic":"fix-dnspython","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2021-10-12 04:10:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0a1ffdc92c054183cb6e720eb53cc98f99bbc6ab","subject":"Fix test failures with dnspython\u003e\u003d2","status":"MERGED","created":"2021-10-08 11:58:12.000000000","updated":"2021-10-12 04:11:55.000000000","submitted":"2021-10-12 04:10:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813176-fix-dnspython","meta_rev_id":"caaa0d4ae2937e1d116039686bcd4917b4b03426","_number":813176,"virtual_id_number":813176,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-12 04:10:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2021-10-11 22:59:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-12 00:21:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2021-10-12 00:30:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-08 13:14:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-11 22:59:36.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-10-12 00:21:17.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"2092f09960b74301b9b2a863e74feb2ff98495f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-08 11:58:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1244501891243f06dbc95a4cf0adedb6a6058cc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 13:14:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/86f52a1064254206bd70fad8730fec3f : FAILURE in 38m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bd654b3e91774317b1e5a3d350c6f717 : FAILURE in 1h 14m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ac8c5f174d254a37bb9351ece17909c5 : FAILURE in 54m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1b66ad30a22d4d92b90826da8fa94b19 : FAILURE in 36m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/535e6614151c4244bfe56e0a37560e16 : FAILURE in 41m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1480a461a156417daa92bbc2ef52a8d0 : SUCCESS in 6m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2ea9ddad2464ef7a567d37373028439 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e8490e1f5c3473f91c46c7eda389fb8 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da41967d751f474a8239aa2fbbd2f102 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bd6312cf8fe46a1aab2b62a82111caa : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b483a1d6cba48d193d860978cb11c3e : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39f38ff47d5741d5b4b093c7aa6b4dca : SUCCESS in 8m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/984d7db6f3bc4f2780ff4ae5a1477288 : SUCCESS in 22m 45s","accounts_in_message":[],"_revision_number":1},{"id":"271a9bd7f7616f73b2d675dfdfc31cc07a2549ab","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-08 19:16:59.000000000","message":"Patch Set 1:\n\nrecheck holding a node for further analysis","accounts_in_message":[],"_revision_number":1},{"id":"d60874bf13c3a8c04567bce6b49ef01092120dc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 20:20:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d69eb495cb4343afb3a2b0de757187ce : FAILURE in 48m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0ab9c0d824a54ab097c13e64ab9fb533 : FAILURE in 55m 07s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a97f5a975c20484b8c3a26f406b73cab : FAILURE in 43m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3b6f31f9cd704774b404a322bef2714f : FAILURE in 29m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3d7b8e245b7d4c1eb32275a79980ef7e : FAILURE in 29m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3974d2c9ad4848309a8af2f7c2938f6a : SUCCESS in 6m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c76f375c3a44b479bba1ae718f88185 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/debf62bb76834b10b54523ebb5989f7b : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d265cbdd45bc4501aa0ef7ede36ee6f8 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e864aca47b1346589dd1dab2a8e53696 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0e52c3f7f9c48d381760335783e4c8e : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b22983d9d994b319d6af0a64825d342 : SUCCESS in 9m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/81536e6e8c584d6f873192a3938332c2 : SUCCESS in 28m 08s","accounts_in_message":[],"_revision_number":1},{"id":"3c96a638ddc1033cb6bcb623e36132dd0d06da68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-10 17:16:29.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c7401fb84dabdb1e8c5ca3dad4b78f95fe7cee8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-10 18:20:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/66f1c365bbf748169e6b0eabe05ae846 : SUCCESS in 41m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1d4aa24033024d6dbb887a62a571383c : SUCCESS in 33m 49s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/23e6a7edcb1d40049a1529595c0c2177 : SUCCESS in 58m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/444a2f842e80437d94777daf00fe59be : SUCCESS in 35m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c0f06163b7a44cefb73a8279e315c73c : SUCCESS in 43m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cba2069105e4475bf87e15fccdf02bb : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c23de9444194ee9a90f2f16624967be : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5dc490fda9d4f799e92108a3155cdbe : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03968d80ba304779b8f5f6afd4dab410 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef640536bfa5452dbbabcc55a24d8bbb : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1a54bb8b3c74d129a21e3fad6ebaeb2 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1072657b816455cbf51e25672d1033b : SUCCESS in 7m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/02e2a0ced1284f5bb5431edd8040c5db : SUCCESS in 21m 15s","accounts_in_message":[],"_revision_number":2},{"id":"a6f0567db8b5273def1e8b85e704c9dcf11984e8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-11 22:59:36.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6e44f80cfe90a1e38e8c03be4b5b48f1e58569e7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-10-12 00:21:17.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d4a18563a9b1545a4d12aa58718910b6e75526bc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-12 00:30:52.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b75001a34a26dacc9dec316bef99b2408c8b78df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 00:36:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d9d1f090c02454e2d0721600e9bb60ed1d9c72a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 02:14:19.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/facc2bee01df4612b665fa32555a5112 : SUCCESS in 40m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/154bdab7dbf342fe997e7073c592ab2d : SUCCESS in 46m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9fad6cf9a8074b33b4b4f6719948108f : FAILURE in 1h 34m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/133a265202ac4945950639fb4a1d0b08 : SUCCESS in 39m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1df1debcb21c43f2a21d245e55bc6919 : SUCCESS in 44m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8caf8a80a0c14daea888cde8091be283 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d98bdd9c25a1456ea30145e511cb5437 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/669452f3aa0b41e5aa02c545a001d4bc : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abec257b63ca4289a8bbf84d3c099aca : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84bfde38b82940ccaafdb0527de88020 : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":2},{"id":"1c59f6c374f8d91b338e51c93f3ee43e1b8e0587","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-12 02:17:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aee8f8beccbe8d7aea6e0b5d6ef4e4304e00a0d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 03:13:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9bfc3d1249144ae4b0b4be76b0e3f800 : SUCCESS in 40m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/74a9b31b9ebd40eeb69334508655aac8 : SUCCESS in 39m 48s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c6271ca3223d4db5b47f6a9b09f382f2 : SUCCESS in 54m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cc4f13701db1489fb2b608b960d5e97d : SUCCESS in 40m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3653cf90816f41f88e94ac9a8a043b0f : SUCCESS in 47m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ef286602204420fbd613b57ee6f54fb : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8941d7bd6eca4a8c9ae8bcd854a38927 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76f015658d0244d18915bcd1294b8440 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f99c7ccad1dd4b248a9523a2e634c090 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/211882d87b3b42149878b2a7e0503890 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b15ea89bd024c4a8ff663dd5d21cd39 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f66881c02ad7420f8f3773d67cf84086 : SUCCESS in 8m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f18ff28e4f854e569dba5f011d9a6f34 : SUCCESS in 40m 34s","accounts_in_message":[],"_revision_number":2},{"id":"42733e5041b167a22602889528a15ec80c375d1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 03:14:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ec541a40f732b82a2e41f3d8a41af8d7ef4369d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 04:10:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ea54282754d044fd82b5200fafe21160 : SUCCESS in 37m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ad11fd6336504356b781636454edcbee : SUCCESS in 33m 18s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/51ba11693f7c434d856e4c0eb8ad5f36 : SUCCESS in 55m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/48bd118ccc3441d08c4165b50bc2955d : SUCCESS in 34m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ecbbfea5fe6745d0b2faf8f15aa4bab2 : SUCCESS in 40m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77a52247845d49daa448ae66e31dc065 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ec5dc987e134868bdd03a7bc0fff7e7 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/586118544f6944f089cbd9488485caa5 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c15526207a6d42c4a48b45c03919a98b : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eef39f2ea19e459e955faef0a4c7fc84 : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":2},{"id":"150dfbf59b4cf7a2548522d817f3975b56373647","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 04:10:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"caaa0d4ae2937e1d116039686bcd4917b4b03426","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 04:11:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a660d6fd572e4d87ae15cb2d1dc4b011 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"897fc7925bd7c821dcc13df577c0ec75974f5458","revisions":{"f3127e7ba2735ae5da28af6819c7dbfe4b5a1ed1":{"kind":"REWORK","_number":1,"created":"2021-10-08 11:58:12.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/76/813176/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/813176/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/813176/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/813176/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/813176/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/813176/1"}}},"commit":{"parents":[{"commit":"2f17dd20e260c3b8a36a5034077275c796abb60b","subject":"Update Quotas documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2f17dd20e260c3b8a36a5034077275c796abb60b"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-10-08 11:54:14.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-10-08 11:57:24.000000000","tz":120},"subject":"Fix test failures with dnspython\u003e\u003d2","message":"Fix test failures with dnspython\u003e\u003d2\n\nNewer dnspython has some stricter type checks, in particular we need\nto ensure that zone names end with a \".\". Also some objects are now\ndicts instead of lists.\n\nTwo checks try to create seemingly invalid messages, those will need\nfurther work, marked as expected failure for now to unblock gate.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I0a1ffdc92c054183cb6e720eb53cc98f99bbc6ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f3127e7ba2735ae5da28af6819c7dbfe4b5a1ed1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f3127e7ba2735ae5da28af6819c7dbfe4b5a1ed1"}]},"branch":"refs/heads/master"},"897fc7925bd7c821dcc13df577c0ec75974f5458":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-10-10 17:16:29.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/76/813176/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/813176/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/813176/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/813176/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/813176/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/813176/2"}}},"commit":{"parents":[{"commit":"2f17dd20e260c3b8a36a5034077275c796abb60b","subject":"Update Quotas documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2f17dd20e260c3b8a36a5034077275c796abb60b"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-10-08 11:54:14.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-10-10 17:12:52.000000000","tz":120},"subject":"Fix test failures with dnspython\u003e\u003d2","message":"Fix test failures with dnspython\u003e\u003d2\n\nNewer dnspython has some stricter type checks, in particular we need\nto ensure that zone names end with a \".\". Also some objects are now\ndicts instead of lists.\n\nTwo checks try to create seemingly invalid messages, those will need\nfurther work, marked as expected failure for now to unblock gate.\n\nDepends-On: https://review.opendev.org/813299\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I0a1ffdc92c054183cb6e720eb53cc98f99bbc6ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/897fc7925bd7c821dcc13df577c0ec75974f5458"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/897fc7925bd7c821dcc13df577c0ec75974f5458"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
