)]}'
{"id":"openstack%2Fdesignate~814290","triplet_id":"openstack%2Fdesignate~master~I867600eb48a3e30a4d17471ab794ca717706823d","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2021-10-19 20:25:02.000000000","reason":"Dr. Jens Harbott (frickler) replied on the change"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2021-10-21 21:50:50.000000000","reason":"Michael Johnson replied on the change"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2021-10-21 22:59:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I867600eb48a3e30a4d17471ab794ca717706823d","subject":"Fix race condition in the sink when deleting records","status":"MERGED","created":"2021-10-17 10:12:43.000000000","updated":"2021-10-21 23:01:05.000000000","submitted":"2021-10-21 22:59:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814290","meta_rev_id":"b892092890b69d21f3f4fefa423cf67da33b5bee","_number":814290,"virtual_id_number":814290,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-21 22:59:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2021-10-19 20:25:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-21 21:50:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-21 21:50:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2021-10-19 20:25:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-17 11:22:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-18 18:13:14.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-10-18 20:36:46.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2021-10-21 21:50:50.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"8ebaa7a321f97c03f65479413f71e8766a210c53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-10-17 10:12:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a7b939f97566271fbed4af3da1fb25fd0887904","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-10-17 10:14:38.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2a971d7826255f0ed7ad7fea24e1eb741644aaa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-17 11:22:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/947692650f3945f1892e4126a3d2f16a : SUCCESS in 41m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8aad6abccaea4c36ba7782f5366ec7b4 : SUCCESS in 42m 34s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5dc46dcfcc634592a8586f4ba9b14948 : SUCCESS in 1h 01m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e00eaad81d634c19a669932c66afc853 : SUCCESS in 41m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/05790d3273924055b93d0daa64ed9dca : SUCCESS in 44m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc119cdfc0934aa897120ae67d08292a : SUCCESS in 6m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1498768a9c14b8f99b4de435b4dc970 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0eba929d68a74e038969fb8ad8aa073d : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/714021e761b24f6381f158286dbd436a : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4be87f2b314b4650b8f682b2d2f06c2b : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b98a39e13f204f4c8cca7d2822006360 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d243b0e3b564948a413de3d04c25de8 : SUCCESS in 9m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d90a9c39e98f42d2978bb4b62d204b37 : SUCCESS in 38m 03s","accounts_in_message":[],"_revision_number":2},{"id":"b8d9c20db6159768fc9d65ee02e2a38eabb8ce91","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-18 18:13:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b38ecd90c565a54e2e221f35d688e35ad47876a1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-10-18 18:23:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"354796d607ce62ef9132c1f8cc6c6ef727d196f2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-18 20:36:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"41bb0f73fd2e58875b99d0e53d33e4e7885717c2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-10-18 20:39:48.000000000","message":"Patch Set 2: Code-Review-2","accounts_in_message":[],"_revision_number":2},{"id":"35cbef4747e90a5d644b8a6454272008cce33ee7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-10-18 20:40:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b80d98ae7ed9a616802784810df33132d3c8750d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-10-19 01:49:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cf60740d16836a59369376fbb6efce05ca0dcea5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-10-19 01:49:37.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"d8d4c9d4748974e15b08e490b8fb3cac7dec496c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-10-19 01:52:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1ef712ff59d296fd1a5c2bb761653f578c6f97d9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-10-19 01:54:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"406f3e926390ab81b5687217147bc46cd723933e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-10-19 02:34:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ba2765c4842c063385622223887d7dcfcb6242e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-10-19 02:37:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"cb57bcf799870392d7be4bf66680b64759c211d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 03:43:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c1dacfaa09ec405a8e456e3c54c2b5b2 : SUCCESS in 43m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/79f78055f88343c59950296ebca6a782 : SUCCESS in 41m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a0ae9d0dfae84dba8e3acde07530d895 : SUCCESS in 59m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cb5b79cfd09143998945c5c8ef89bf5b : SUCCESS in 37m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1aece29c979c4be3a3bebae6418a4eb0 : SUCCESS in 40m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73b124dcdc2540338abd7c4d3c72f236 : SUCCESS in 6m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8414e6709ae544a38e732dd0fec44bd2 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/795cb5ad9e6549788dfb4c150a095016 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17e62079a56a4b94b7cfed723524b824 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0cb2d0cb0134ae4a1e3bab5ce95c146 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18b9b0d42b024b4f81b6cd7bccff1005 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c60d539e9b974e1caebcc2bd679dff7b : SUCCESS in 7m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/efc5bdd03c974af5ab43bffaf2454a24 : SUCCESS in 41m 32s","accounts_in_message":[],"_revision_number":6},{"id":"1df86196b35acfb33846bb5a5fa61f853c8efc79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-10-19 03:44:53.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"034b7cce8c2288b859151fa295373d9283746740","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-10-19 03:45:55.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"41473cd2aeff8e563ce5d70ff330ebdaf6c7922a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 04:35:57.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7919ab134ff143819cec458127ae6685 : SUCCESS in 37m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/db13696ae8a84dc0b1111df2b89e3190 : SUCCESS in 36m 22s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/369937b657cf4034885f1e86fc0e3cd2 : SUCCESS in 49m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cdf055d4da844a6e82432f6d7956c6f5 : SUCCESS in 37m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/87ea1e3efd874e8e852f021d629a3ce4 : SUCCESS in 37m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/256a6e5bbf4e457bba39e4b6d124c7a7 : SUCCESS in 6m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffd8355bb33b46ddb6bb5ad87ddaef30 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d75ee757e81424885460788be73f708 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94da5a2c7f0a4513be94b3b6b47e85d4 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8047c8b6f2674d708d0cf3632db8b1e5 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3ac715f27e34a98aea2866eb8da6492 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31e1ae3c36d941f998d1c00ae702d3d9 : SUCCESS in 8m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0dfc315d23c544cf9673fade86f94828 : SUCCESS in 34m 53s","accounts_in_message":[],"_revision_number":8},{"id":"c6cd06a1b673bc17a1de0b503c9481b63f810ddc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-10-19 05:35:04.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"dcafd1d2fd523c5a50b8ef04c14460e408e92bf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-10-19 05:36:58.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"dea31eea3c738ef8a79162a5813e6390bf2e771a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 06:36:38.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4684e0f493b04b488fd26e2291c6f0bd : SUCCESS in 36m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/96353fbc21984d7da16055909ff4b715 : SUCCESS in 42m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7fda9dc429a14ed49407353c75b157f5 : SUCCESS in 56m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cf29c1fc0f4845dc88f89475e2cf13fd : SUCCESS in 35m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0cf47d2a199a4c4d90f15a4b69d900af : SUCCESS in 38m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d298163085fd47288c27ce0bcd35520c : SUCCESS in 6m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f1b3b9883b74f52be4ddbc00f222f3b : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a09ebabf8b4466f9ad0d03e535f358a : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d0bbe5f70134d5883f1c7b9777ca52b : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79e55c688ad846b295d88623ba7a5117 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b5d3683209a4486be729acd6315277c : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30b13f5baf694b89b7045d024e55068b : SUCCESS in 8m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/15126c02a52648aebc828deff2b86552 : SUCCESS in 34m 23s","accounts_in_message":[],"_revision_number":10},{"id":"639053a220b524d91ce609e82fa4813e9ed6f955","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-10-19 19:39:03.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"f13a894a5127298c67d3b1f836ecefa80f8bd290","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-19 20:25:02.000000000","message":"Patch Set 11: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":11},{"id":"1954ee77ba9418bc56ebead07396524a3811ab54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 20:31:44.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0beb5882bf8240459e40188496918afd : SUCCESS in 36m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b96fe3a4fd344eb09803be4c793f1944 : SUCCESS in 39m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/93ddfd81046843299a53dbe34cda9cf3 : SUCCESS in 42m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f1609f491a5a4ebab02cf2e96dcdd42d : SUCCESS in 36m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e269e51f3d9d48fbb3b507c2c216ac5f : SUCCESS in 33m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59cc5418adc449138595f64480208bfa : SUCCESS in 6m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84e9338950d24be09ba67543cc35a018 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/344ebeb6477440bbb3ae19de0b6ee9d2 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7db89a866a054935a51899c689d72d26 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ec6ba87699c4475b0f64525db1da078 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6041967b4f2f4ca6983e0bacde99d3f1 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b09daee9ff54c648c0a50a01c866689 : SUCCESS in 9m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8e9d96fa75914020a01157d3b3765383 : SUCCESS in 45m 30s","accounts_in_message":[],"_revision_number":11},{"id":"c1d507a2bf9d2ba9aa51d8e95e91f66eecfb7ce9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-21 21:50:50.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"3b244fb3b80a51977956cd0d0e18b5e0c33c780b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 21:51:28.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"9673cae33bae6433d24d6d289ee46090311c437c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 22:59:37.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4cb42438f3974b52bbf3ca8cac2d136c : SUCCESS in 40m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7adc3533b3124371b01f7b1cce99aa06 : SUCCESS in 39m 57s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/19c68fdfeac64f489a286eaab05cf295 : SUCCESS in 1h 02m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/61b1c412b1e04bd4808061ec3ee404a7 : SUCCESS in 37m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8ac3e81334c149e5bc0bc05eed942a7a : SUCCESS in 36m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7636b80ca74d45a494fce20c2ef86914 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b52ecfda0513433e9e93f56cafa13199 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f72ed22b9dc48128b4ff95d4ac729eb : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fad6d1fd8a8042999b64de6f3861319b : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/912b3ab116cd4e1e9b1af07dc6e16ca8 : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":11},{"id":"f68d3f4ae3e4ec1d34da6bd8296fe347d44e7eba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 22:59:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"b892092890b69d21f3f4fefa423cf67da33b5bee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 23:01:05.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/010fb98cdfb74adc99fefd3a3b47cba3 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"4807c23228fbb23af3087ee072bf73d7fc43aff5","revisions":{"94d0d0ca7f2052022f272a893e55083bcaa8841c":{"kind":"REWORK","_number":1,"created":"2021-10-17 10:12:43.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/90/814290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/814290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/814290/1"}}},"commit":{"parents":[{"commit":"771197c2f31209fe135888aef6f2828af4889653","subject":"Remove Akamai (eDNS SOAP API) backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/771197c2f31209fe135888aef6f2828af4889653"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-17 10:05:28.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-17 10:12:19.000000000","tz":-420},"subject":"Fix race condition in the sink","message":"Fix race condition in the sink\n\nWe no longer use delete_record, and should\nbe using delete_recordset instead. This is\nwhat the api would use.\n\nChange-Id: I867600eb48a3e30a4d17471ab794ca717706823d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/94d0d0ca7f2052022f272a893e55083bcaa8841c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/94d0d0ca7f2052022f272a893e55083bcaa8841c"}]},"branch":"refs/heads/master"},"655d641f806ddac12b257d9b8d2d4b14c3bf791e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-10-17 10:14:38.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/90/814290/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/814290/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/814290/2"}}},"commit":{"parents":[{"commit":"771197c2f31209fe135888aef6f2828af4889653","subject":"Remove Akamai (eDNS SOAP API) backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/771197c2f31209fe135888aef6f2828af4889653"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-17 10:05:28.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-17 10:13:03.000000000","tz":-420},"subject":"Fix race condition in the sink","message":"Fix race condition in the sink\n\nWe no longer use delete_record, and should\nbe using delete_recordset instead. This is\nwhat the api implementation uses today.\n\nChange-Id: I867600eb48a3e30a4d17471ab794ca717706823d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/655d641f806ddac12b257d9b8d2d4b14c3bf791e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/655d641f806ddac12b257d9b8d2d4b14c3bf791e"}]},"branch":"refs/heads/master"},"571896c2ca06a1a541f477444079c9a83928ae26":{"kind":"REWORK","_number":3,"created":"2021-10-19 01:49:23.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/90/814290/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/814290/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/814290/3"}}},"commit":{"parents":[{"commit":"771197c2f31209fe135888aef6f2828af4889653","subject":"Remove Akamai (eDNS SOAP API) backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/771197c2f31209fe135888aef6f2828af4889653"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-17 10:05:28.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-19 01:48:16.000000000","tz":-420},"subject":"Fix race condition in the sink","message":"Fix race condition in the sink\n\nUpdated the sink to behave closer to how these calls are implemented\nin the the designate v2 api today.\n\n- Use recordset update / delete instead of just record delete.\n- Added object validation to all requests.\n\nChange-Id: I867600eb48a3e30a4d17471ab794ca717706823d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/571896c2ca06a1a541f477444079c9a83928ae26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/571896c2ca06a1a541f477444079c9a83928ae26"}]},"branch":"refs/heads/master"},"28f934c7d27e199581e2715940d65e11e36dd718":{"kind":"REWORK","_number":4,"created":"2021-10-19 01:52:55.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/90/814290/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/814290/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/814290/4"}}},"commit":{"parents":[{"commit":"771197c2f31209fe135888aef6f2828af4889653","subject":"Remove Akamai (eDNS SOAP API) backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/771197c2f31209fe135888aef6f2828af4889653"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-17 10:05:28.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-19 01:52:34.000000000","tz":-420},"subject":"Fix race condition in the sink","message":"Fix race condition in the sink\n\nUpdated the sink to behave closer to how these calls are implemented\nin the the designate v2 api today.\n\n- Use recordset update / delete instead of just record delete.\n- Added object validation to all requests.\n\nChange-Id: I867600eb48a3e30a4d17471ab794ca717706823d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28f934c7d27e199581e2715940d65e11e36dd718"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28f934c7d27e199581e2715940d65e11e36dd718"}]},"branch":"refs/heads/master"},"c91c1d42aa290dcd65c1c439f2f82dede192a36b":{"kind":"REWORK","_number":5,"created":"2021-10-19 02:34:46.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/90/814290/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/814290/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/814290/5"}}},"commit":{"parents":[{"commit":"771197c2f31209fe135888aef6f2828af4889653","subject":"Remove Akamai (eDNS SOAP API) backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/771197c2f31209fe135888aef6f2828af4889653"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-17 10:05:28.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-19 02:34:15.000000000","tz":-420},"subject":"Fix race condition in the sink","message":"Fix race condition in the sink\n\nUpdated the sink to behave closer to how these calls are implemented\nin the the designate v2 api today.\n\n- Use recordset update / delete instead of just record delete.\n- Added object validation to all requests.\n- Better unit testing.\n\nChange-Id: I867600eb48a3e30a4d17471ab794ca717706823d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c91c1d42aa290dcd65c1c439f2f82dede192a36b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c91c1d42aa290dcd65c1c439f2f82dede192a36b"}]},"branch":"refs/heads/master"},"d6538d77140424a25b9060b0d32e6f257bd6e7eb":{"kind":"REWORK","_number":6,"created":"2021-10-19 02:37:23.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/90/814290/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/814290/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/814290/6"}}},"commit":{"parents":[{"commit":"771197c2f31209fe135888aef6f2828af4889653","subject":"Remove Akamai (eDNS SOAP API) backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/771197c2f31209fe135888aef6f2828af4889653"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-17 10:05:28.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-19 02:37:17.000000000","tz":-420},"subject":"Fix race condition in the sink","message":"Fix race condition in the sink\n\nUpdated the sink to behave closer to how these calls are implemented\nin the the designate v2 api today.\n\n- Use recordset update / delete instead of just record delete.\n- Added object validation to all requests.\n- Better unit testing.\n\nChange-Id: I867600eb48a3e30a4d17471ab794ca717706823d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d6538d77140424a25b9060b0d32e6f257bd6e7eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d6538d77140424a25b9060b0d32e6f257bd6e7eb"}]},"branch":"refs/heads/master"},"ac3fadfd93bb496707b8359333ffd7910ec18188":{"kind":"REWORK","_number":7,"created":"2021-10-19 03:44:53.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/90/814290/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/814290/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/814290/7"}}},"commit":{"parents":[{"commit":"771197c2f31209fe135888aef6f2828af4889653","subject":"Remove Akamai (eDNS SOAP API) backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/771197c2f31209fe135888aef6f2828af4889653"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-17 10:05:28.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-19 03:43:17.000000000","tz":-420},"subject":"Fix race condition in the sink when deleting records","message":"Fix race condition in the sink when deleting records\n\nUpdated the sink to behave closer to how we handle this type\nof operations in designate.api.v2.\n\n- Added object validation to all requests.\n- Better test coverage.\n- Use recordset update / delete instead of just record delete.\n\nChange-Id: I867600eb48a3e30a4d17471ab794ca717706823d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ac3fadfd93bb496707b8359333ffd7910ec18188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ac3fadfd93bb496707b8359333ffd7910ec18188"}]},"branch":"refs/heads/master"},"b1ec09575104ba19ce78767ab53ecb81b3db3db4":{"kind":"REWORK","_number":8,"created":"2021-10-19 03:45:55.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/90/814290/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/814290/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/814290/8"}}},"commit":{"parents":[{"commit":"771197c2f31209fe135888aef6f2828af4889653","subject":"Remove Akamai (eDNS SOAP API) backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/771197c2f31209fe135888aef6f2828af4889653"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-17 10:05:28.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-19 03:45:48.000000000","tz":-420},"subject":"Fix race condition in the sink when deleting records","message":"Fix race condition in the sink when deleting records\n\nUpdated the sink to behave closer to how we handle this type\nof operations in designate.api.v2.\n\n- Added object validation to all requests.\n- Better test coverage.\n- Use recordset update / delete instead of just record delete.\n\nChange-Id: I867600eb48a3e30a4d17471ab794ca717706823d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b1ec09575104ba19ce78767ab53ecb81b3db3db4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b1ec09575104ba19ce78767ab53ecb81b3db3db4"}]},"branch":"refs/heads/master"},"b142cc7ae0f19013022d09553408165a358d1670":{"kind":"REWORK","_number":9,"created":"2021-10-19 05:35:04.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/90/814290/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/814290/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/814290/9"}}},"commit":{"parents":[{"commit":"771197c2f31209fe135888aef6f2828af4889653","subject":"Remove Akamai (eDNS SOAP API) backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/771197c2f31209fe135888aef6f2828af4889653"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-17 10:05:28.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-19 05:34:58.000000000","tz":-420},"subject":"Fix race condition in the sink when deleting records","message":"Fix race condition in the sink when deleting records\n\nUpdated the sink to behave closer to how we handle this type\nof operations in designate.api.v2.\n\n- Added object validation to all requests.\n- Better test coverage.\n- Use recordset update / delete instead of just record delete.\n\nChange-Id: I867600eb48a3e30a4d17471ab794ca717706823d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b142cc7ae0f19013022d09553408165a358d1670"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b142cc7ae0f19013022d09553408165a358d1670"}]},"branch":"refs/heads/master"},"33e70f4a6c2623c4da0d3cebe925d61237a8e82b":{"kind":"REWORK","_number":10,"created":"2021-10-19 05:36:58.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/90/814290/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/814290/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/814290/10"}}},"commit":{"parents":[{"commit":"771197c2f31209fe135888aef6f2828af4889653","subject":"Remove Akamai (eDNS SOAP API) backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/771197c2f31209fe135888aef6f2828af4889653"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-17 10:05:28.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-19 05:36:50.000000000","tz":-420},"subject":"Fix race condition in the sink when deleting records","message":"Fix race condition in the sink when deleting records\n\nUpdated the sink to behave closer to how we handle this type\nof operations in designate.api.v2.\n\n- Added object validation to all requests.\n- Better test coverage.\n- Use recordset update / delete instead of just record delete.\n\nChange-Id: I867600eb48a3e30a4d17471ab794ca717706823d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/33e70f4a6c2623c4da0d3cebe925d61237a8e82b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/33e70f4a6c2623c4da0d3cebe925d61237a8e82b"}]},"branch":"refs/heads/master"},"4807c23228fbb23af3087ee072bf73d7fc43aff5":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2021-10-19 19:39:03.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/90/814290/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/814290/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/814290/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/814290/11"}}},"commit":{"parents":[{"commit":"771197c2f31209fe135888aef6f2828af4889653","subject":"Remove Akamai (eDNS SOAP API) backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/771197c2f31209fe135888aef6f2828af4889653"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-17 10:05:28.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-19 19:39:03.000000000","tz":0},"subject":"Fix race condition in the sink when deleting records","message":"Fix race condition in the sink when deleting records\n\nUpdated the sink to behave closer to how we handle this type\nof operations in designate.api.v2.\n\n- Added object validation to all requests.\n- Better test coverage.\n- Use recordset update / delete instead of just record delete.\n\nCloses-Bug: #1947765\nChange-Id: I867600eb48a3e30a4d17471ab794ca717706823d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4807c23228fbb23af3087ee072bf73d7fc43aff5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4807c23228fbb23af3087ee072bf73d7fc43aff5"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
