)]}'
{"id":"openstack%2Fdesignate~817123","triplet_id":"openstack%2Fdesignate~master~I3ae4a8c9c964290cd790df186ae45bc9f2ee38ec","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-01-25 18:04:15.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"I3ae4a8c9c964290cd790df186ae45bc9f2ee38ec","subject":"Add a BIND9 job on CentOS 9 Stream","status":"MERGED","created":"2021-11-08 23:19:33.000000000","updated":"2022-01-25 20:32:49.000000000","submitted":"2022-01-25 20:26:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"817123","meta_rev_id":"87edb7427c143de51525ff09cf3d2eacf76847ec","_number":817123,"virtual_id_number":817123,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-25 20:26:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-01-25 06:39:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-01-25 06:43:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-08 23:20:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-25 06:39:58.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-01-25 09:56:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"df79a7470921598465d3f6d78664c06282e6af62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-08 23:19:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31fc4b892b05c5172830eb3f23846bc19dfbc08e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 23:20:53.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"5f5ba5f8989ced8e6694697d2951fbaa7e316c4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-08 23:27:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c1bb971280b03469cf801ed1d7e50cb37cc5c26f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 23:36:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9-centos9stream https://zuul.opendev.org/t/openstack/build/d5ac6d8fb80a486e8b2ae724330473af : FAILURE in 7m 26s","accounts_in_message":[],"_revision_number":2},{"id":"f166b984c6cbc09c9f8821b4cbf950e497890093","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-20 01:11:10.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"fa89f80c816183a1e6d2cd5a1f64aa2b9559e3e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-20 01:34:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9-centos9stream https://zuul.opendev.org/t/openstack/build/400f0f5b7eba45698aa5736fd72817a1 : FAILURE in 14m 20s","accounts_in_message":[],"_revision_number":3},{"id":"ec9a4f7cecbd7846c31feda69262b3938c5fa5da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-12-07 17:21:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"571e28b775add13186344b4e070c1dfcdbaf3c19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 18:33:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9-centos9stream https://zuul.opendev.org/t/openstack/build/bb06fbbb0fb34f27883847ded2fcc252 : SUCCESS in 52m 13s","accounts_in_message":[],"_revision_number":4},{"id":"d22037bf90bf2513373174feef9c1457eb77bf24","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-12-07 23:34:51.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"fb37d850fd762201ef18d31bf932e9a92899be90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-24 23:29:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d7ad1b7bade9f531128ccc56f2310bfe65a93a7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 23:30:06.000000000","message":"Patch Set 5: Verified-1\n\n(23 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/designate on branch master.  The error was:\n\n  Job designate-bind9-centos-9-stream not defined\n\nThe error appears in the following project-template stanza:\n\n  project-template:\n      name: designate-devstack-jobs\n      check:\n        jobs:\n          - designate-bind9\n          - designate-bind9-centos8stream-fips:\n              voting: false\n          - designate-bind9-centos-9-stream\n          - designate-bind9-scoped-tokens\n          - designate-pdns4\n          - designate-grenade-pdns4\n          - designate-ipv6-only-pdns4\n          - designate-ipv6-only-bind9\n  \n      gate:\n        queue: designate\n        jobs:\n          - designate-bind9\n          - designate-bind9-centos-9-stream\n          - designate-bind9-scoped-tokens\n          - designate-pdns4\n          - designate-grenade-pdns4\n          - designate-ipv6-only-pdns4\n          - designate-ipv6-only-bind9\n\n  in \"openstack/designate/.zuul.yaml@master\", line 165, column 3","accounts_in_message":[],"_revision_number":5},{"id":"fb15694d9b93f689f476d22eba6b8b63ac8929cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-24 23:31:37.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a3683769f8585aba0c7d155c379bcef9cd3a04ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 01:11:29.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bb2269c448464f718d9392b9fc5d83a3 : TIMED_OUT in 1h 13m 39s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/1e6a4712676b43daa546de34dcbe1d79 : TIMED_OUT in 1h 14m 41s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e85806a2b1946f6b9ab5045f70f6d49 : TIMED_OUT in 1h 13m 36s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/668e08344a7647c7991ba95f0bfb2d5e : TIMED_OUT in 1h 13m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ef11467a36ee477b857735ba4a75d312 : TIMED_OUT in 1h 13m 30s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7328e3a03ab747aa874ab5a2852b3f55 : FAILURE in 1h 37m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/144b58dc14a6482b91f8e52e39a6af54 : FAILURE in 1h 24m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2b281d3d63c844c09f6de356336c0832 : FAILURE in 1h 15m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6eb54938ea94b638e4c7bff34f45996 : FAILURE in 5m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92a65220bcb1478d92e144bc2203e037 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d7b0c8e7fc34b4eaf64f2065237d929 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a201818bee54db7a523fa4a91e771b6 : FAILURE in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2e77ebe913b42ceb32907b656a9ac7d : FAILURE in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5830f6f3b53c444782faa2ac9045270b : SUCCESS in 7m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a8efc3df3db948b7a63b688d5c3ef46c : SUCCESS in 28m 17s","accounts_in_message":[],"_revision_number":6},{"id":"6de0c989d0b15f5aae7e18a8acbd35cf1b53094c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-25 01:58:55.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"50dec63dcf1d1b259f03ab94965da06e9a00aefd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 03:07:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/18657bfc02f6446cadd613e4afc53fb7 : SUCCESS in 45m 27s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/fd77dd74769d46d9b2425366bdfeae0b : SUCCESS in 1h 01m 39s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/28c8c3bb7e25459dbdcf49326c99c2a3 : SUCCESS in 38m 05s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ac4eb14bee404ccd944946b9d3c1c0b8 : SUCCESS in 33m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/410ce3f78bf9479394b81dfbdbdb72cf : SUCCESS in 44m 42s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9d1f69e237a849819b8dcad9db97360d : SUCCESS in 57m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/372e5585238041c3974df19366935a5e : SUCCESS in 35m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/31f8428066654a77b5a8e914c5d09855 : SUCCESS in 37m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5865e8a225a4c3489d5edabb3e340c0 : SUCCESS in 7m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/162642a2505b4b1d86dab9210259856d : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b7a462ba6a647d4bd6b22ae9966aeb6 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2947e08c6874b598ded950d10fb3f89 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aed9e8e1c5a845ebbf5332833a8b3008 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d26faaeb6e074268806c90fd1972bd8b : SUCCESS in 7m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/500c89e934a141b096bbbbe043a5216a : SUCCESS in 31m 35s","accounts_in_message":[],"_revision_number":6},{"id":"22bee7f5c52fec6b40a84e27be60179251a5b5ae","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-25 06:39:58.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"26ac807403cce5678acb022ec491d140646e7a42","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-25 06:43:01.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"58f6da4d36d04f02d538acb6087656b6b778b852","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 06:43:46.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9ad7204c1218d943d9d86680a47014cc025bf11e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 08:02:25.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f0e2bb45905144879ad1ed607abecd35 : SUCCESS in 33m 02s\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f788610fdd340c38cfcd68596e5c2db : SUCCESS in 42m 46s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/b129bc72d03f43e19fa6369eb7171d78 : SUCCESS in 38m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4eb627d47e5d49ffaf04a457e7cc3d75 : SUCCESS in 36m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/90d1367cffb54940825206628bbedef8 : SUCCESS in 57m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d420ac0c86c0450b900c665e7e4379c3 : FAILURE in 21m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4481a3f536754a98b02e5d7f6b669b84 : SUCCESS in 36m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1930289463e7472d9eb83809f23c543d : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4ca06042ce140118871d1052d0d9de3 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb02ed950e97474d8a186ef17683e36a : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b82d678d474b4414935a59412ea326df : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adb9a3c80b0e49258cabb851145e01bc : SUCCESS in 8m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3578f35c448947c7afc0ff2306c2ff67 : SUCCESS in 27m 03s","accounts_in_message":[],"_revision_number":6},{"id":"85b125ab318677a0ed57e0efbcfc7395f72d4d88","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-25 08:41:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"491a99350c80e129a1fb6563cf58e4420e7b57c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 09:30:35.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b000e6b9b29544dc9dd5bcf3e46595f8 : SUCCESS in 29m 28s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/2fc10e6f5e6245f49be13833afe6c434 : SUCCESS in 41m 40s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1858d3d5d11747c2b4aa66aa6bd9b048 : FAILURE in 36m 19s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1b84bdc4bd0f4b67be25c5f03c585260 : SUCCESS in 33m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2aab3433fb024dd4af95bfa6c69ee481 : SUCCESS in 28m 45s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1ab85802b8a147d9b8f7ba386151eb24 : SUCCESS in 46m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6875a51c39444f43b0892b4b0e4a4264 : FAILURE in 31m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/02ef598c19c940bdbcb54a430b2620db : SUCCESS in 36m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d5a77a2420f4750ac373104dcf672f1 : SUCCESS in 5m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f72cfeba64d4c7a940f9fadec4f848a : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7823fa63448a45268b49e9c8ea4e9e5b : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f4ad1c1d22d4ccc8ab6f32ae716fb0b : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f820814b9a141afa668b5691106b6be : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ec60446ad9842a7a9ed2ca79ece00a7 : SUCCESS in 7m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4fad58ae57b84c7691edd8e5d3103aab : SUCCESS in 29m 24s","accounts_in_message":[],"_revision_number":6},{"id":"755e31a43e9f8c26e861c080b9e519427fab1ffc","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-01-25 09:56:43.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3d005fc352796b45c48de31d85baf96011af2a5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 10:58:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/67ebcb83d9b640df814211de970f7adb : SUCCESS in 43m 53s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/2f031b4135604f7e873fa4b7c6b183dc : FAILURE in 56m 36s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d6f500e7dff427e91ded3d71e4bb755 : SUCCESS in 44m 31s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/c154f10c4ef04abea0f67e67e8cfbf52 : SUCCESS in 45m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/88378b69efaf4f278e356d3ce2b5a248 : SUCCESS in 42m 52s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e4f15feec576458fbd74f8522dec13a7 : SUCCESS in 56m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8b3cad31fc054fe69d9a617924036644 : SUCCESS in 40m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5515e4b8e0ef4bb8af88b1e78c47bb96 : SUCCESS in 41m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2e1fd8ed51c4e8aab5eb4e0b08a8f1b : SUCCESS in 5m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a699dd48a7874368a22ac8a27d57a4d8 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a053cd92823f4138bf266c068d00f115 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe0387c3303e4a6f83c56fc33918d6a7 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df67182da1404e44b7f5f525030ea828 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61971d9cc7d1421fb50f03092fae1b21 : SUCCESS in 7m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1b3ebb0f85a941109daeb9cef53e0b86 : SUCCESS in 31m 35s","accounts_in_message":[],"_revision_number":6},{"id":"5fc8a727d2bbc8bff101fa09ca44a0dcc0f7da8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 10:59:43.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"a09a0d22731e010fb340a3c410398d6235b9430e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 11:49:20.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a1ccc34889c141c29af7c4bb5f100321 : SUCCESS in 32m 37s\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/62184fdc5aae4f4fa6bed5e2c02dd831 : SUCCESS in 33m 50s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/32492bd025dd4456b3406ab5f98053d4 : SUCCESS in 41m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8c29d180c42d41af991291c5f9e39e92 : SUCCESS in 29m 48s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c065b7e3bd1045c4a9e461def3fae6c4 : FAILURE in 47m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1e2a60b8617c4388b915558b7b97d2f3 : SUCCESS in 28m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1fb69f855b764939968ba1a3f1f741c5 : SUCCESS in 35m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73cd53e1bba34d95873bb3a05c7830a4 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0ad20bd9a5b4e499583379806fc0701 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7693191526394c4c81cfe676ae59a4e8 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6e749967e3b48df8f74900bf32a5ea2 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5667173fc66543e09eae7534db16c94a : SUCCESS in 10m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9f8e047976fa4572b4ee2079abdd799a : SUCCESS in 27m 44s","accounts_in_message":[],"_revision_number":6},{"id":"6132846f1fb90f9651b3225e18ef87c51570b71e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-25 18:04:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"26d38ca695f449c2c4d759fa4e15c939dd555f97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 19:11:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4994c2cee1864f789d4e351900c26b43 : SUCCESS in 42m 52s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/c2f793d673104162a080c47589b1748e : SUCCESS in 38m 24s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/12d40e65ef91413493fae06417784404 : SUCCESS in 36m 02s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3877e83e26cc4408b0a27aeeb6abd588 : SUCCESS in 47m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/443b002fe10548df8651eeac7c0858e6 : SUCCESS in 36m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/73de3717acfa4d69a459b9d1c8961174 : SUCCESS in 58m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f2e4af8da51d4bdfacaa97ddd6e14ab9 : SUCCESS in 44m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/31bb647fa6ec497b90df40b489b306fe : SUCCESS in 47m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b85682b15f64be599993d1b7aee1b3c : SUCCESS in 7m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d73bc4447a124698978b90d3039c0033 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94f5481f19024b12a0dc864d8af76adc : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b493c9bdc02c4f87a3c4b3498ef060ae : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc7f2c6c5f954e7da864cfdc5dc61050 : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/156cb951140c4800830fd50f281cc88a : SUCCESS in 10m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/beb5295d2db74082aaddc7aec7beebc2 : SUCCESS in 30m 45s","accounts_in_message":[],"_revision_number":6},{"id":"df7a38055575221a75e850fc59de8d6625d49724","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 19:12:59.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"671902f3f88f577d120fce4c24d7f5562a574237","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 20:26:42.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/51eef5a32359466286743053b5de9fb4 : SUCCESS in 32m 15s\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/efab140dbe424c4c921b940fc098f886 : SUCCESS in 49m 33s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f178c8a9f8cd410eaf83fdc9d3a87588 : SUCCESS in 38m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/17a7d66b1e5f4de8a5ff720fe0ffd52d : SUCCESS in 43m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ed04e110901147c18f9084916e9459b2 : SUCCESS in 1h 12m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d489b07d746f420c8b5de200c5449396 : SUCCESS in 40m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4b62753f1a4e41c487c424cb8466721a : SUCCESS in 30m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4dc1a2b7ad8741419d9286708b50db14 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96b8ab07cc994b8997388109a631098e : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/824df5ab00ed42b78d116e49ddf008bc : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fce0f3b40f6842ffab810b463352a222 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28456788131a4de1939372b42efdbcd2 : SUCCESS in 8m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7f309f984cae4b9d80afd71dd4fac5b9 : SUCCESS in 27m 16s","accounts_in_message":[],"_revision_number":6},{"id":"939d1539a08b217778edfbf5fe11249ec32edec1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 20:26:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"87edb7427c143de51525ff09cf3d2eacf76847ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 20:32:49.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f395a3aeb9b340d7bc6ac9c87e26829d : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"0275aaad556a6b20dc8666270cbda6ae7bf084fa","revisions":{"76702de4cc55c5b2edfb7b9b034313d62752a633":{"kind":"REWORK","_number":1,"created":"2021-11-08 23:19:33.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/23/817123/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/817123/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/817123/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/817123/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/817123/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/817123/1"}}},"commit":{"parents":[{"commit":"604fd8cf7468a54125486e45959f717cf182ece8","subject":"Add fips jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/604fd8cf7468a54125486e45959f717cf182ece8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-08 23:18:34.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-08 23:19:24.000000000","tz":0},"subject":"DNM - Testing BIND9 job on centos 9 stream","message":"DNM - Testing BIND9 job on centos 9 stream\n\nChange-Id: I3ae4a8c9c964290cd790df186ae45bc9f2ee38ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/76702de4cc55c5b2edfb7b9b034313d62752a633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/76702de4cc55c5b2edfb7b9b034313d62752a633"}]},"branch":"refs/heads/master"},"12ec5b9c25d57c254a0f0d55939b81d36bda26b8":{"kind":"REWORK","_number":2,"created":"2021-11-08 23:27:13.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/23/817123/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/817123/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/817123/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/817123/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/817123/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/817123/2"}}},"commit":{"parents":[{"commit":"77f239634242315e1533d8c969fd1bcc59cc8a12","subject":"Merge \"Allow TXT record over 255 characters if split\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/77f239634242315e1533d8c969fd1bcc59cc8a12"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-08 23:18:34.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-08 23:27:07.000000000","tz":0},"subject":"DNM - Testing BIND9 job on centos 9 stream","message":"DNM - Testing BIND9 job on centos 9 stream\n\nChange-Id: I3ae4a8c9c964290cd790df186ae45bc9f2ee38ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/12ec5b9c25d57c254a0f0d55939b81d36bda26b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/12ec5b9c25d57c254a0f0d55939b81d36bda26b8"}]},"branch":"refs/heads/master"},"f5a2ecef24d67cff5634593968b31dbfa61b08b3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-11-20 01:11:10.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/23/817123/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/817123/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/817123/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/817123/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/817123/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/817123/3"}}},"commit":{"parents":[{"commit":"77f239634242315e1533d8c969fd1bcc59cc8a12","subject":"Merge \"Allow TXT record over 255 characters if split\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/77f239634242315e1533d8c969fd1bcc59cc8a12"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-08 23:18:34.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-20 01:11:10.000000000","tz":0},"subject":"DNM - Testing BIND9 job on centos 9 stream","message":"DNM - Testing BIND9 job on centos 9 stream\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/800903\n\nChange-Id: I3ae4a8c9c964290cd790df186ae45bc9f2ee38ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5a2ecef24d67cff5634593968b31dbfa61b08b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5a2ecef24d67cff5634593968b31dbfa61b08b3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8f5d1c0a274600ca7a3bf6ced548bf902873b819":{"kind":"REWORK","_number":4,"created":"2021-12-07 17:21:21.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/23/817123/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/817123/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/817123/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/817123/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/817123/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/817123/4"}}},"commit":{"parents":[{"commit":"aa798fc006e0d4bce38751af7b0c14d97b0fecfb","subject":"Merge \"Add hacking check D710 for LOG.warn()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aa798fc006e0d4bce38751af7b0c14d97b0fecfb"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-08 23:18:34.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-12-07 17:21:20.000000000","tz":0},"subject":"DNM - Testing BIND9 job on centos 9 stream","message":"DNM - Testing BIND9 job on centos 9 stream\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/800903\n\nChange-Id: I3ae4a8c9c964290cd790df186ae45bc9f2ee38ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8f5d1c0a274600ca7a3bf6ced548bf902873b819"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8f5d1c0a274600ca7a3bf6ced548bf902873b819"}]},"branch":"refs/heads/master"},"aec243142a855d0ae48d511d524c553f8aeeb3fd":{"kind":"REWORK","_number":5,"created":"2022-01-24 23:29:15.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/23/817123/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/817123/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/817123/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/817123/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/817123/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/817123/5"}}},"commit":{"parents":[{"commit":"bff3d5f6e31fe595a77143ec4ac779c187bf72a8","subject":"Fix designate-manage pool update bugs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bff3d5f6e31fe595a77143ec4ac779c187bf72a8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-08 23:18:34.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-01-24 23:29:13.000000000","tz":0},"subject":"Add a BIND9 job on CentOS 9 Stream","message":"Add a BIND9 job on CentOS 9 Stream\n\nCentOS 9 Stream is part of the yoga release PTI[1] so we should have a job that tests with it.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: I3ae4a8c9c964290cd790df186ae45bc9f2ee38ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aec243142a855d0ae48d511d524c553f8aeeb3fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aec243142a855d0ae48d511d524c553f8aeeb3fd"}]},"branch":"refs/heads/master"},"0275aaad556a6b20dc8666270cbda6ae7bf084fa":{"kind":"REWORK","_number":6,"created":"2022-01-24 23:31:37.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/23/817123/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/817123/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/817123/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/817123/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/817123/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/817123/6"}}},"commit":{"parents":[{"commit":"bff3d5f6e31fe595a77143ec4ac779c187bf72a8","subject":"Fix designate-manage pool update bugs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bff3d5f6e31fe595a77143ec4ac779c187bf72a8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-08 23:18:34.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-01-24 23:31:31.000000000","tz":0},"subject":"Add a BIND9 job on CentOS 9 Stream","message":"Add a BIND9 job on CentOS 9 Stream\n\nCentOS 9 Stream is part of the yoga release PTI[1] so we should have a job that tests with it.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: I3ae4a8c9c964290cd790df186ae45bc9f2ee38ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0275aaad556a6b20dc8666270cbda6ae7bf084fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0275aaad556a6b20dc8666270cbda6ae7bf084fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
