)]}'
{"id":"openstack%2Fdesignate~817542","triplet_id":"openstack%2Fdesignate~stable%2Fxena~I867600eb48a3e30a4d17471ab794ca717706823d","project":"openstack/designate","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2021-11-11 21:11:16.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2021-11-11 20:07:58.000000000","reason":"Erik Olof Gunnar Andersson replied on the change"}},"hashtags":[],"change_id":"I867600eb48a3e30a4d17471ab794ca717706823d","subject":"Fix race condition in the sink when deleting records","status":"MERGED","created":"2021-11-11 14:27:00.000000000","updated":"2021-11-11 21:13:37.000000000","submitted":"2021-11-11 21:11:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"817542","cherry_pick_of_change":814290,"cherry_pick_of_patch_set":11,"meta_rev_id":"fb3028e756a626ba5e93f89b416788b60564dbb7","_number":817542,"virtual_id_number":817542,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-11 21:11:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2021-11-11 18:36:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-11 20:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-11-11 20:07:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-11 14:27:00.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2021-11-11 17:24:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-11 18:36:39.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"34566b0071549c4f78de4f4629d2e0d1d395219f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-11-11 14:27:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b31ab9601045055893ee02e14469cfd50fdb1ef6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 17:24:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/80b87c2feb574d2db33e720ea0f9eb09 : SUCCESS in 40m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6338c5742d9444f9a1935b520d1a12c8 : SUCCESS in 42m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/555138c618bf4dd4bc626fcde399b056 : FAILURE in 1h 05m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f3ad35a001f04807bdd4d67679d39265 : SUCCESS in 46m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fe7f5728d3b74ca2a2b72bf997bc217c : SUCCESS in 44m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01322a20c4614cb6b76bb38e0a51f7a8 : SUCCESS in 8m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1cc3845183546c2bb2f02c91f3b988e : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d1619edf20a48a9938417c9060b6b1a : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac28f668e6ae44539b264925a044f843 : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be6c988686c24c648c8849c64ee3c356 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a46884227034f5d808adfdaf07fce24 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c673207f9414d67a10216951d86836e : SUCCESS in 9m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/991bafd2eae641fab4a97c57bfdb2f1d : SUCCESS in 43m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f1f06176e395c0874378241d58a91e2500b0454c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-11 18:36:39.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"137646417d7e4e0f11cd1cafd548dbb343a6be56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 20:01:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b1328a6372694c2184b70f08b6d36c0a : SUCCESS in 43m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2452c141013c43b380382d5c112279cc : SUCCESS in 38m 29s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/87a40ed9ca0341988e54f5a5666bcdbd : SUCCESS in 1h 16m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/78ebe1d89d744633921c6d3005108f87 : SUCCESS in 40m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/eb96008b68cf4b79a78388d21179d309 : SUCCESS in 39m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c08d75812bb42078af071fe09fc0dc8 : SUCCESS in 6m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4d0dea8605448739975b2e9e676ab13 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ed4738353e54760ad64d324338b4fef : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b0efb1f3f4141e89792ac7179e312ab : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e96ac74bb3684b5eb053b6e1313f1758 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e8fcf6c6b6c4ec8b9474449613db167 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c120aac3811b49b790d73f25f0a2807a : SUCCESS in 8m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cbc552d875d940628c117dca190174af : SUCCESS in 39m 16s","accounts_in_message":[],"_revision_number":1},{"id":"f76d64f7e33eb4ac121c149b2b5173d31c8263e9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-11-11 20:07:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e3ed98300d77b549d8c9559a4c0805c99a0b0a67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 20:09:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"123b120de8bdd4e414bd06aa0d4ce803dcc25540","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 21:11:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/60e11edb0d164093b3b698e8aadcb637 : SUCCESS in 44m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b1f26f0d81424b53a581f0e49d628e06 : SUCCESS in 34m 03s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a57084d1c329431fae9b9b853aa51292 : SUCCESS in 1h 00m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/06d4f3e623894b8db02bf19f71dda788 : SUCCESS in 38m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e59ccaa14fc9450c8cd09f8d29ded3b2 : SUCCESS in 43m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b46d11be93043e5b36edbfd92e24e04 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6cd776b5c5d4495b09dbd9355dbf045 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b50f1175a8264ff39eabcc21206afe4f : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af2c4fb06b0f43d18fc49eefbf9bd997 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7542397f7a724722867bb5829592ebae : SUCCESS in 9m 04s","accounts_in_message":[],"_revision_number":1},{"id":"2a3cec67966b65d35f3fe6c72eae2a5926fd7d77","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 21:11:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fb3028e756a626ba5e93f89b416788b60564dbb7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 21:13:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e109a0347ba44e6ea46dbf4082942f3e : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8634d531f2d993901527ecc67ddc40eb9dc0a575","revisions":{"8634d531f2d993901527ecc67ddc40eb9dc0a575":{"kind":"REWORK","_number":1,"created":"2021-11-11 14:27:00.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/42/817542/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/42/817542/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/42/817542/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/42/817542/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/42/817542/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/42/817542/1"}}},"commit":{"parents":[{"commit":"86db7954fc122e0326156f9e6a205ea8453bf73a","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/86db7954fc122e0326156f9e6a205ea8453bf73a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-10-17 10:05:28.000000000","tz":-420},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-11-11 14:27:00.000000000","tz":0},"subject":"Fix race condition in the sink when deleting records","message":"Fix race condition in the sink when deleting records\n\nUpdated the sink to behave closer to how we handle this type\nof operations in designate.api.v2.\n\n- Added object validation to all requests.\n- Better test coverage.\n- Use recordset update / delete instead of just record delete.\n\nCloses-Bug: #1947765\nChange-Id: I867600eb48a3e30a4d17471ab794ca717706823d\n(cherry picked from commit 4807c23228fbb23af3087ee072bf73d7fc43aff5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8634d531f2d993901527ecc67ddc40eb9dc0a575"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8634d531f2d993901527ecc67ddc40eb9dc0a575"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
