)]}'
{"id":"openstack%2Fdesignate~818649","triplet_id":"openstack%2Fdesignate~stable%2Fussuri~I9dcf92a3bff579a963adea2bedc9a5e66207b8c7","project":"openstack/designate","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2021-11-19 23:38:37.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9dcf92a3bff579a963adea2bedc9a5e66207b8c7","subject":"Switch stable/ussuri to use ussuri neutron job","status":"ABANDONED","created":"2021-11-19 22:32:44.000000000","updated":"2021-11-19 23:38:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6d780c5cee741a9f1c4220135e946fa40e327b80","_number":818649,"virtual_id_number":818649,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-11-19 23:28:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-19 23:28:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"24a5556c5b7034bb6300526a4e28e93ab33f41ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-19 22:32:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1de3b8537a2f201039055b09b1c09203f4647ca1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 23:28:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7213bf3efac54a50bf4c164e2ae363cc : FAILURE in 30m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c3385da5923b458db78eeabfe6f26562 : FAILURE in 37m 44s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d2fecb358dc1421a9b14d57241700431 : FAILURE in 50m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6e009201ffdb4deaa9df4686dd7047f3 : FAILURE in 29m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/07d8a167e80742039959ee43f37aa1dd : FAILURE in 51m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc25d7b48dc74870a819acb05a189d66 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/277c8b54e1c84af3aba0d436c7a36725 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef796a8af9984b178ce96d1f35634db1 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9294923c87b64923b4ef2bb1391cbdda : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf46d8bebd33472fa73cf91ccfb09cec : SUCCESS in 8m 21s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/a6508f5a6b6244ffad9a59924aed9738 : SUCCESS in 35m 14s","accounts_in_message":[],"_revision_number":1},{"id":"6d780c5cee741a9f1c4220135e946fa40e327b80","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-19 23:38:37.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c91a5a2d68e96b42abef6a3c4d9fab40b893ffed","revisions":{"c91a5a2d68e96b42abef6a3c4d9fab40b893ffed":{"kind":"REWORK","_number":1,"created":"2021-11-19 22:32:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/49/818649/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/49/818649/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/49/818649/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/49/818649/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/49/818649/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/49/818649/1"}}},"commit":{"parents":[{"commit":"b967e9f706373f1aad6db882c2295fbbe1fadfc9","subject":"Update zones masters using pool target masters.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b967e9f706373f1aad6db882c2295fbbe1fadfc9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-19 22:29:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-19 22:32:43.000000000","tz":0},"subject":"Switch stable/ussuri to use ussuri neutron job","message":"Switch stable/ussuri to use ussuri neutron job\n\nNeutron has branch specific jobs to account for different extensions being\nloaded via devstack.\nCurrently the stable/ussuri branch is running the neutron master branch\nneutron-tempest-plugin-designate-scenario job which loads an extension,\ndns-integration-domain-keywords, which does not exist in ussuri.\nThis causes some new tests in neutron-tempest-plugin to not skip and fail.\nThis patch changes the job to the stable branch specific version of the\ntest to allow the check gate to pass.\n\nChange-Id: I9dcf92a3bff579a963adea2bedc9a5e66207b8c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c91a5a2d68e96b42abef6a3c4d9fab40b893ffed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c91a5a2d68e96b42abef6a3c4d9fab40b893ffed"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
