)]}'
{"id":"openstack%2Fdesignate~818650","triplet_id":"openstack%2Fdesignate~stable%2Fvictoria~I9128ddf7e62d05c1f55810edfdfd1bac3f7a48ff","project":"openstack/designate","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2021-11-20 01:05:48.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9128ddf7e62d05c1f55810edfdfd1bac3f7a48ff","subject":"Switch stable/victoria to use victoria neutron job","status":"ABANDONED","created":"2021-11-19 22:35:49.000000000","updated":"2021-11-20 01:05:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b434a8628b4dc6f34cfa1336e9c196844f1e9e3c","_number":818650,"virtual_id_number":818650,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-11-19 23:44:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-19 23:44:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8dac0526e5a108fd4c039c5c2472991e6f356f45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-19 22:35:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6df6b75d24751b8a1b5319edeaf32e45adefe9f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 23:44:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/efabf2aa4ae04653b6c262df7b863017 : FAILURE in 49m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ba436e8b94f6497dbbf1d85ef8296061 : FAILURE in 49m 56s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ceefca0a0dd84a2d9913604e1a63c8e6 : FAILURE in 1h 06m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/daf5ffac38014a50a571c28f6e3e7b57 : FAILURE in 47m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/59df1c134a734f8d9286255fa3d03921 : FAILURE in 42m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e4b395c0dae4788aec3a3c0ddc6cb87 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c49af412f59646a3b6d5a6f3535974f4 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4a198ae571f4544bd21fbc0a03b975c : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/548bbac7f4634ea2b0d25b25ad19b9a4 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/949e2f7e25d44e34a2e447955c5f301f : SUCCESS in 7m 32s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/c4688a81959d4de7aff0d6cd25f53f91 : SUCCESS in 49m 32s","accounts_in_message":[],"_revision_number":1},{"id":"b434a8628b4dc6f34cfa1336e9c196844f1e9e3c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-20 01:05:48.000000000","message":"Abandoned\n\nsquashed into another patch to unstick the gate","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dd2383acaa2d701f7894517643b8537098ed2284","revisions":{"dd2383acaa2d701f7894517643b8537098ed2284":{"kind":"REWORK","_number":1,"created":"2021-11-19 22:35:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/50/818650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/50/818650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/818650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/818650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/818650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/50/818650/1"}}},"commit":{"parents":[{"commit":"7cbbd3a18ffedc9dbeaddaab189643b59a224878","subject":"Merge \"Fix pool update with duplicate masters\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7cbbd3a18ffedc9dbeaddaab189643b59a224878"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-19 22:35:21.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-11-19 22:35:48.000000000","tz":0},"subject":"Switch stable/victoria to use victoria neutron job","message":"Switch stable/victoria to use victoria neutron job\n\nNeutron has branch specific jobs to account for different extensions being\nloaded via devstack.\nCurrently the stable/victoria branch is running the neutron master branch\nneutron-tempest-plugin-designate-scenario job which loads an extension,\ndns-integration-domain-keywords, which does not exist in victoria.\nThis causes some new tests in neutron-tempest-plugin to not skip and fail.\nThis patch changes the job to the stable branch specific version of the\ntest to allow the check gate to pass.\n\nChange-Id: I9128ddf7e62d05c1f55810edfdfd1bac3f7a48ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dd2383acaa2d701f7894517643b8537098ed2284"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dd2383acaa2d701f7894517643b8537098ed2284"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
