)]}'
{"id":"openstack%2Fdesignate~822371","triplet_id":"openstack%2Fdesignate~master~I4f024ebeac001986cc56384b1f77083926d146d5","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"32291":{"account":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"last_update":"2022-01-05 22:25:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4f024ebeac001986cc56384b1f77083926d146d5","subject":"Updating python testing classifier as per Yoga testing runtime","status":"MERGED","created":"2021-12-21 02:08:45.000000000","updated":"2022-01-05 22:27:45.000000000","submitted":"2022-01-05 22:25:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822371","meta_rev_id":"312d877a02ff180497ee452fa62f163c9bbf55b8","_number":822371,"virtual_id_number":822371,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-05 22:25:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-01-05 20:51:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-05 20:59:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-05 20:59:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-21 03:52:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-05 20:51:18.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-01-05 20:59:01.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"205fa2ba676df3ac8e927ffed6e02bb9cb9795e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-12-21 02:08:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6effad2f26061a02153d1e7617ddff14c8c94882","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 03:52:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b2f2fc878505478c8c09fee56717aa95 : SUCCESS in 3m 30s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a41fb7f1096f423eb85a1e117e129f66 : SUCCESS in 38m 11s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/b488cb091ad349bbbafe67bef279707e : SUCCESS in 1h 00m 39s (non-voting)\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7bc9c6e633954f74b2c3dd377e32bc8f : SUCCESS in 42m 44s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d1e0893c9b3642ceaf072bc6f3dde0dd : SUCCESS in 1h 03m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9b293e6a6eef4e119dcae9934284aace : SUCCESS in 43m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fab782e13d9a46b78e9786bab556cbdd : FAILURE in 44m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8a269e6b9c7497ba7add7a76505507c : SUCCESS in 6m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b06cd0a406564edca04f5ea45b34a6e8 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce38d3ca90dc4b42b2e6e4096e5ae177 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ce697e2b2bb4ee98092010de28e5816 : SUCCESS in 7m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63cf2a07257b4900afb4ce766216c1ff : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8ffa534677d4362aa79cc61d00e0f97 : SUCCESS in 11m 10s","accounts_in_message":[],"_revision_number":1},{"id":"28cc7db271b3a121e532b12e3daa325643cc194e","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-12-22 01:06:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"37a4aac43ac9a2e38587d9a7baadd13f17defb81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 02:03:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3eb2a8aa0bfc46bbbb1cb96efca042e9 : SUCCESS in 2m 26s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2919e341b9354372bf874b8d20e710e7 : SUCCESS in 32m 11s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/9bd52c6a466f46efa8e8989fafae20af : SUCCESS in 49m 56s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/a055fcef1f66475c9d8b488a0bd58c80 : SUCCESS in 38m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/35cf0c77ca674dac9257edd0679dc079 : SUCCESS in 29m 50s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7c2a4dd498004a15b41c2571db2db607 : SUCCESS in 55m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d22634b44d454a759d91f05e6d65f8fe : SUCCESS in 28m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3371bc5238e04849a1853e8183883094 : SUCCESS in 38m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b97f9b188cf43b58facb6835cb9be64 : SUCCESS in 5m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a13332a18ca94334a3f6f33a446df6ed : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdb885dd09fc47cdbf5fbbb2af148af7 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b795c25094a42b093f82a76ffe6fb6c : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbab5061b9244c5aab2049791c515b6e : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19449a5f26b7410d99b0a3aa58e9458d : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":1},{"id":"17b1dc4746a60af0277d4ae27311430ff9bf9456","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-05 20:51:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d2989fcdf1cbc392dd50f2b8afbfe485e686687","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-05 20:59:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9147b8151f4fda17364fbac4b6852a193c8a7a39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 21:00:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8007168ce8b7c095ff77bbf440be2a93373c5152","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 22:25:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5340099809834295bc934bc6fee4dd3e : SUCCESS in 2m 28s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/900b036efd7b4e17bde943f6fe0f301a : SUCCESS in 50m 58s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2251176841e04c03bb97b0590d7f7b4f : SUCCESS in 42m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ee38b0dfa59b44e48d77e5b2632ce3eb : SUCCESS in 40m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d1ab0e4b38434e2f99417b88efbe95ab : SUCCESS in 1h 20m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e449678f09534ce68de95024bf81d4e3 : SUCCESS in 45m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bbd6fcd84a374485b05787559421307e : SUCCESS in 41m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69a12da8d7c24ba1bcc6a1fc58eaf01e : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54dd328f75544926bb911ef5ef3cd882 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abfd2514b8ef4ef2b88b8c6eaa4c91a5 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bec536e978c4444ab8a955149db9ae47 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6db3d16c53a3459fb40390e601880637 : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d65e870b39eccaac6d8f20cc2bc39dd67a204908","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 22:25:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"312d877a02ff180497ee452fa62f163c9bbf55b8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 22:27:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c80bda09c1914dd8aeb3ee9622db26ff : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1504fbee26951af423f77b2c10f45d9e7879e103","revisions":{"1504fbee26951af423f77b2c10f45d9e7879e103":{"kind":"REWORK","_number":1,"created":"2021-12-21 02:08:45.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/71/822371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/71/822371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/822371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/822371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/71/822371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/71/822371/1"}}},"commit":{"parents":[{"commit":"aa798fc006e0d4bce38751af7b0c14d97b0fecfb","subject":"Merge \"Add hacking check D710 for LOG.warn()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aa798fc006e0d4bce38751af7b0c14d97b0fecfb"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-12-21 02:07:36.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-12-21 02:07:36.000000000","tz":480},"subject":"Updating python testing classifier as per Yoga testing runtime","message":"Updating python testing classifier as per Yoga testing runtime\n\nYoga testing runtime[1] has been updated to add py39\ntesting as voting. Unit tests update are handled by the\njob template change in openstack-zuul-job\n\n- https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/820286\n\nthis commit updates the classifier in setup.cfg file.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: I4f024ebeac001986cc56384b1f77083926d146d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1504fbee26951af423f77b2c10f45d9e7879e103"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1504fbee26951af423f77b2c10f45d9e7879e103"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
