)]}'
{"id":"openstack%2Fdesignate~822891","triplet_id":"openstack%2Fdesignate~master~I4e6ea0b6b717d2a1b5cc420874d1bb8fb290e04b","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2021-12-28 06:41:49.000000000","reason":"Dr. Jens Harbott (frickler) replied on the change"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-01-08 00:48:51.000000000","reason":"Erik Olof Gunnar Andersson replied on the change"}},"hashtags":[],"change_id":"I4e6ea0b6b717d2a1b5cc420874d1bb8fb290e04b","subject":"Fix designate-manage pool update bugs","status":"MERGED","created":"2021-12-24 07:15:02.000000000","updated":"2022-01-08 01:10:17.000000000","submitted":"2022-01-08 01:08:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822891","meta_rev_id":"6971ace74641b9e6db5ba886c8fb0f5848d98b8b","_number":822891,"virtual_id_number":822891,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-08 01:08:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-01-08 01:10:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-08 00:03:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-08 00:03:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-24 08:46:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-27 15:52:30.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2021-12-28 06:41:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-01-08 00:03:48.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"4b07964dfd82936d60940e35f327452067a35ccd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-24 07:15:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"432d4e574af27394ac89cc53b8ded2088624a5c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-24 07:19:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bff0833ca4b7a054203a22a0847feca30f38015c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-24 07:22:10.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ae1a71a12e63d5c43adb7cf96336d4070768b563","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-24 07:36:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"98805177680401ea2dda551702c43652cb8c6c23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-24 07:45:26.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6ff24ffe76e8b897205818d766a895b7ae8cc440","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-24 07:49:18.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"ab3d8a6daff578a2756fdf84a111079b79e40b80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 08:46:25.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/26120e620f5f41a585a95488b6717a09 : SUCCESS in 41m 03s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/c983fce27f22485f95b7263f83db0a66 : SUCCESS in 50m 10s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/97be2905bc6746288d1e309eacd1a846 : SUCCESS in 38m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4c8b8282047546f2ad90e63a2bdf200f : SUCCESS in 39m 42s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/36d9c79c605a41d59d7c421edeb120bb : SUCCESS in 54m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3c94f25c65e54c33b1e24c1bb16bd0b4 : SUCCESS in 47m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1ceb8189df19418986286f54432dded5 : FAILURE in 23m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d7238fccba741fc8ded67386c0d1fde : SUCCESS in 5m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71eca90f3b484613b1c4d73dfc28e636 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a53109acf1cb457f92055d0eb0bfb08b : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57a7d1b9f83842e7985d5140952fd307 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9272615fbe694cd081c93115426f6900 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e1a98e9695401ab6f13298c3ee4a15 : SUCCESS in 7m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0aff707e204d404f9ef8d5103f377e36 : SUCCESS in 27m 53s","accounts_in_message":[],"_revision_number":6},{"id":"ce51a02ed2d20dd16eb38ea0645014430f842b5f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-24 08:53:32.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ebb14c61b2e7ca50e5701f12fc32c09547d5e010","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 10:08:15.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/eef8e526709a4dd5888e228ab6ffbdc3 : SUCCESS in 38m 16s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/9c7180d205a2478ab354c0354af2af9d : SUCCESS in 41m 16s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/80e4ddd133134894a83e87a627b7bc9a : SUCCESS in 31m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bc200580077d45ce87ae65b6f5da5bf6 : SUCCESS in 30m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/62f7d7fe065e4c8cbe7672be7eaaf975 : FAILURE in 1h 13m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2abff35bc9e54e0088aefb0fe3f919d8 : SUCCESS in 32m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d95fb7cee569478ea2f66193143734cc : SUCCESS in 32m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/624209cf2db04c92b905d5ffd41244a5 : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce7892bc0aab4686ae09fbcca460f78a : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61fd3e8e4a18473d8bd4b9598f95fdd5 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0440ff4163744d2c98262a5d582406cb : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5586d35fbf914b2fbe116b0476308710 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/739d66d72b0347e79a1e1d2b764753b8 : SUCCESS in 8m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7e661ed8146141dc8c7330241e0deef6 : SUCCESS in 41m 03s","accounts_in_message":[],"_revision_number":6},{"id":"816fbe55470ab6074d621063ac108cdfc27c17b1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-24 17:45:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0633fb7d6ac16ac474332221000dd58970888480","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 18:43:58.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e67e5c026022446bb4fde9a3afd2b202 : SUCCESS in 34m 37s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/4450f2ae383043779b2d181ad0414dad : SUCCESS in 41m 51s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/168fcd7fe4584e8e9b36dcb2000ef142 : SUCCESS in 33m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c1a4d74d9eaa498388853489f52d7435 : SUCCESS in 28m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c8151f889c364aaa9c45c93ffb02aaa4 : FAILURE in 56m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7d0e1c387354472fa5cd10ef32a61a59 : SUCCESS in 36m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7420718876ad46f8940d71cd8534f164 : SUCCESS in 34m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93bb913427c64a409e6e64ed925b73e7 : SUCCESS in 7m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb5d5c65a7c44a749e0e4b63c344fe96 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2504a6cf47024bbea6c946161038b0a5 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eff096ad6ffb4607bb30940016644c8a : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a25089ba0ab426e9985a53594089407 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56f6b483b5e14fc492909ba27272eebd : SUCCESS in 7m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a531794e219e44c7b16288456bc95154 : SUCCESS in 45m 45s","accounts_in_message":[],"_revision_number":6},{"id":"0e1091b13c581c1d44510e6525c36a1b3b89248e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-24 19:05:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"97a069a1b67adbc7a50725c32bfdb8d6794fa5eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 20:07:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/47154a57b6254247b71c44b22570ee4d : SUCCESS in 42m 03s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/dad0b76a034a4949b25849f2d7a4622d : SUCCESS in 57m 01s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1b730fd604d340c4917fa2fc66718387 : SUCCESS in 26m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7fe1c6e84f8d4094bf08c3acf5a2bbbe : SUCCESS in 26m 50s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5d74b293991b467197a92322634e327c : SUCCESS in 43m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d25457156d524eabb526144747097234 : SUCCESS in 27m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cbad00ef6b1c4974aef35c823c0a746e : SUCCESS in 31m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcdcdb5de56343a2bd10a132a8af88be : SUCCESS in 5m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53640dc39c684589938bae485e627f86 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e6afc058633465cb239b6d0d9b7f565 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f5221c209c644dfa3f7236acef7c1a4 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ac8d5b01b4e4f93bb423f8e2afdcab8 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8847f06e9a8a47c09076f4fa5ff22a5c : SUCCESS in 6m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/31a6f3e962d44a1abe6f5686e70f6f06 : SUCCESS in 30m 44s","accounts_in_message":[],"_revision_number":6},{"id":"3067eb5b0989bc8bdd966d45e32b51e2a5244344","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-27 00:13:28.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"1f4f9a7eb7fd36bfe7dfe9bf83e801889f93ece1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-27 00:23:25.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"170b8a3efec006cf5aa4cb4cffa46a58f863c261","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-27 00:24:53.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bb51e919f0d534d651c68a11fa9ac68e2def93bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 01:24:02.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5d931245f42949f8b03552127b20e8e5 : SUCCESS in 44m 56s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/91c8a4e693584464b2ca7133ca76b85f : SUCCESS in 1h 00m 39s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3307c55e8f2946588d300fdc0136885e : SUCCESS in 45m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/10971797f1484719ade0430af207da5d : SUCCESS in 47m 13s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8c51f79d0d2e47d3a2719b540478d90e : SUCCESS in 1h 05m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8e51fafc22c440c2bd7f8a4134ad3089 : SUCCESS in 29m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7705e9c0571a433cbcfd9fc5d8ae130c : SUCCESS in 31m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89c63fe5aeda4fd6b2058101b85d79e9 : SUCCESS in 5m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf186f31685340b4bc00ba2d05f02878 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c846cb5a785e451192ecd719ef9969b8 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/743b2fdac1794af1ba6ccebc3bab2346 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e1e043182d44902a83441b6b9598f45 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16f1dad7b70e4299a81e833c07ae8b4f : SUCCESS in 7m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5fedcee926624cbfb240b08f95a1e984 : SUCCESS in 24m 40s","accounts_in_message":[],"_revision_number":7},{"id":"16ae93f81cf9f874f1cf74e46c29d94aa47d9670","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-27 15:52:30.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"608e2aa05e56caedf69d4b1c5824071a394bc3d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-27 19:16:28.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d07647470a93c9fde5e00a3b66bac03b96db02dd","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-27 19:17:03.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"a1b827f27564cd74dc87ffa32279315d5172f9ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 19:58:10.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3e5c49a834544f89b05d1420ab8995ca : FAILURE in 32m 46s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/18a272ef67b84425b3fe8f4fd9ed29ec : FAILURE in 13m 44s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/d0a8ac3d03f441b4ad0d17ef9c0be1fb : SUCCESS in 38m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/033168abe48247a9b21e407bcf259a11 : SUCCESS in 39m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9d3e1d96baf841ea905c9da35b109646 : SUCCESS in 39m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/006f960bd4674abdbf155718ec9c06d7 : SUCCESS in 33m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/32724b5f77e540998d067ca39e64fab4 : SUCCESS in 34m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e683ba134b24416ebef71bd86939ecb7 : SUCCESS in 6m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2181aa8ccd564e9aa7bc262a8c19a135 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45fdc4320c294648bb9527b0d910aa49 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba17ae72e7454edcb095493d07df7d1a : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd548b0de5ae469497659745cd9db2ea : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07922ae5e8d6477383e99e9b467c0faf : SUCCESS in 8m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/68553067a119435db6a6281d04e557e1 : SUCCESS in 26m 05s","accounts_in_message":[],"_revision_number":8},{"id":"d446ad2aa7a36a3861bc740a8949da080b6431c5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-27 20:25:39.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"00f55b002d9c933f41cb2a8cf8e479a55cd6c787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 21:20:03.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/68825f83e1d04eceb1f3e5539dafea4f : SUCCESS in 46m 34s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/c575892091a44cb29b682b4001d19254 : FAILURE in 18m 26s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/cff9ca9f2b5840b5b5304911569aa6d6 : SUCCESS in 48m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/db3c3628c8c44ac69977516a0caf36a3 : SUCCESS in 40m 27s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a48f38996f354c829b81b0273f89af69 : SUCCESS in 49m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8c93181e4d8447a9a9b1ace9b8e1242b : SUCCESS in 46m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/dbf2b75063de4c3582b94cbfdac29cc7 : SUCCESS in 46m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90346748329845b093d2ef7ecef8f514 : SUCCESS in 4m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7db9e3269a824c3e97af2ed8fb070b5f : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0503e0b9ed645c4a9a52d3006917526 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4162ff2dbb2a4b2da201995bc57f79ac : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7d64fba37e44faf9ae2f4cdbc5e88b1 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62f8e9b267264d36a4570040bdbd2600 : POST_FAILURE in 26m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5e2c6e7441b84594bab2f725f9cecc3f : SUCCESS in 28m 21s","accounts_in_message":[],"_revision_number":8},{"id":"dfaed52f42740e6fa9756112ad30593f3717fdc9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-27 21:40:04.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6d610b90a5262167026cb83e2152478035b57bfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 22:24:52.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/773b3268356e40698164041882ef2872 : SUCCESS in 31m 04s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/7ffcbf1040a14771a32f39d3de782511 : FAILURE in 25m 43s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/48d2940b418a42838e6eedd4dd588033 : FAILURE in 33m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/08b9d0549c5f43e8a203b177bb210c62 : FAILURE in 35m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2e27a585f1764d07b1e959d55dd70a4f : FAILURE in 42m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bc56ec0d1a504f8ba5651f9a2c8f5da9 : SUCCESS in 32m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c894c077f69e4cf4a5c62a40047a565c : SUCCESS in 40m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65ff7645158b438aa10f55a2c1ba58c7 : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73ee40c184b44d8c966d6ac236817fdd : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e862f8b2a07a4b95849d54963075f361 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f11307da0cc3444d849b80927b34c8f9 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b734108755c40bea1a8b7b26e6f1103 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/266ab3393c67481393d13c0e872afa58 : SUCCESS in 6m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6e8d52ea6fa24193bca9d612f4e1ca5a : SUCCESS in 30m 56s","accounts_in_message":[],"_revision_number":8},{"id":"283412f18f1d10e651a18b2039cd226bb029557c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-28 00:37:12.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6d8e61421302bc1b4a6af0ea3dd910e33c59e0aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 01:29:59.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fdd03b68248541679ecf2f9ee4de864b : SUCCESS in 35m 20s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/f9a3d3cc04fe46da96a4fcc0cd103fc7 : FAILURE in 14m 38s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e155d92848f64625a6ed570b2f1eb72d : SUCCESS in 37m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e7f1e35e35d644769fcd2f42e7630d74 : SUCCESS in 33m 39s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/032086a1aaad4796b46a3f45393f8d1e : SUCCESS in 50m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b4dbb5f4cf914907bae04cd9e0c681a9 : SUCCESS in 32m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fca209913ff64c4f82f4e6b60e515a34 : SUCCESS in 36m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d9fb6584fe449b1b1a3c02b1a6e4312 : SUCCESS in 7m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e065119446b54f51a15ef776b86181f8 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0753ef4455d4309aa45b43dcd27906f : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3d166d7ae7b4b988d2f4aaf24dcd6f8 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84f16bdd06e34ba3ab98b276622c82c7 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9724c7271a14d2dbc8ae380159b86ca : SUCCESS in 8m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9714394e3279434498e57063afda1acd : SUCCESS in 28m 31s","accounts_in_message":[],"_revision_number":8},{"id":"f6dec3ac7b2783c85df165790a42c62df20492dd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-28 06:41:49.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"3ebf3e46b12f3a5536587adc3458e7d76fbce935","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-06 03:51:17.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"a6a08e019c96d051839ba6a363c4ecd452bb75d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 04:55:57.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/58deaf9861fe41c8b69c3fe023e8fafc : SUCCESS in 42m 12s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/7fbc452618cb4fd28aef4af82c0ce90e : SUCCESS in 1h 03m 35s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2cad7b8110a94bf4b20d3ec87c2d6e7d : SUCCESS in 40m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/12611268d5464985a8053e18d3144613 : SUCCESS in 46m 57s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/91cd25b5a14d4aeb94f4e564f714777f : SUCCESS in 58m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cfbfc020966343ff9f2ee175c4a22cf6 : SUCCESS in 43m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/85bf7469ab664f6296238545f38b4113 : FAILURE in 44m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5a16fe7b0384156a36131833696d077 : SUCCESS in 5m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d5a9c757c434565ab7fc12f913df878 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53d99a3c5a8a4c5cbf0705d66fcb7ae2 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fddd0d7693c3465581273d59c2605848 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be599e78d19a4a338813c5ebff16c42e : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1490769cf3949eeb1694c9c50838bb5 : SUCCESS in 7m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6d99c42f67314566a5afb7ed7d967277 : SUCCESS in 40m 36s","accounts_in_message":[],"_revision_number":9},{"id":"2ef427e61c6160610ece04a5961eb8e9dc779a58","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-06 06:01:00.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"bb020b42d256f331b54aa49fcb7ba1e2e44f1d20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 07:15:25.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e5982637ba794ce3b3b53cf39e10f373 : SUCCESS in 45m 41s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/6d1343f842ff4b7086be4b149a94c04d : SUCCESS in 1h 09m 25s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/b616275eae874893bb81177c43914757 : SUCCESS in 45m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c550001af0c5467380f1248dca2eba08 : SUCCESS in 39m 18s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/969b26b947fc4d658d1dbe62b7717706 : SUCCESS in 1h 04m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bfd85ba28e93481299535119a09c7eb6 : SUCCESS in 45m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cea09ab9e7c2493784de3dd1d548ca44 : SUCCESS in 42m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45e4f4c5794e4e569cfe15f154efed6f : SUCCESS in 5m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59481787efcf44148b004f89662bf71c : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/888fe9a559c847ad82ca84784ec00bb8 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e8d4436361a4095ab53695c485db6c9 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c10d151b4064f6d9e49a9016aa749c5 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed772898f40546b38bb096036aa19b39 : SUCCESS in 8m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8a952fdd1bb0405eab39090d8a404a21 : SUCCESS in 35m 25s","accounts_in_message":[],"_revision_number":9},{"id":"8a2946f4fea0aab899f2fc7b0b561e9e42a17f89","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-08 00:03:48.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f219685975f2e45dece181d3a97ccc60ac4af22f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-08 00:04:35.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"6d4081c26035f654933b7c1697f5c5ce249b481e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-08 00:48:51.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c80c16ce6dbc1fb66f4c22a14c45622f416d1735","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-08 01:08:47.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/90083b961a484d8592a6a36fb0f29902 : SUCCESS in 55m 26s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/360d139f32f841ebafd2c09abd3626bc : SUCCESS in 43m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/92921d35b3f34ac28b7ef3e3534b643c : SUCCESS in 40m 43s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4b68f973b7184bbcad026932e77b8a97 : SUCCESS in 1h 00m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c546f3d86c24443db55d3a7c2d9fc809 : SUCCESS in 40m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e4673c32c0a0491283a402a67548dab5 : SUCCESS in 43m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a2162f4d19f446882a50b37ceed8f3d : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab5f494f282d41fea967810f0766b292 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/987d403c9a1c49b8952852bfb79c9879 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6cf4c05b15014c37bbe38b0652c37daf : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d2ae77ec05c49cda14581d79880be01 : SUCCESS in 7m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/215b9e56d460431e8806a28088f76bab : SUCCESS in 30m 29s","accounts_in_message":[],"_revision_number":9},{"id":"2ba31fb1f04608ffb3778c4e00e88b75d5afb7ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-08 01:08:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"46648c7bdb731b51396b9dc3a8decfe678a46460","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-08 01:10:17.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbeb35d632e742539094d4910fa0314e : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"bff3d5f6e31fe595a77143ec4ac779c187bf72a8","revisions":{"b5eacb7d87a4a3aaa885612ed605b8f4c518473a":{"kind":"REWORK","_number":1,"created":"2021-12-24 07:15:02.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/91/822891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/822891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/822891/1"}}},"commit":{"parents":[{"commit":"aa798fc006e0d4bce38751af7b0c14d97b0fecfb","subject":"Merge \"Add hacking check D710 for LOG.warn()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aa798fc006e0d4bce38751af7b0c14d97b0fecfb"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-24 07:14:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-24 07:14:46.000000000","tz":-480},"subject":"[WIP] Fix designate-manage not handling NS records properly","message":"[WIP] Fix designate-manage not handling NS records properly\n\nChange-Id: I4e6ea0b6b717d2a1b5cc420874d1bb8fb290e04b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b5eacb7d87a4a3aaa885612ed605b8f4c518473a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b5eacb7d87a4a3aaa885612ed605b8f4c518473a"}]},"branch":"refs/heads/master"},"dbcb652e17dadb575562129e651ec602c6062677":{"kind":"REWORK","_number":2,"created":"2021-12-24 07:19:25.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/91/822891/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/822891/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/822891/2"}}},"commit":{"parents":[{"commit":"aa798fc006e0d4bce38751af7b0c14d97b0fecfb","subject":"Merge \"Add hacking check D710 for LOG.warn()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aa798fc006e0d4bce38751af7b0c14d97b0fecfb"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-24 07:14:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-24 07:19:11.000000000","tz":-480},"subject":"[WIP] Fix designate-manage not handling NS records properly","message":"[WIP] Fix designate-manage not handling NS records properly\n\nChange-Id: I4e6ea0b6b717d2a1b5cc420874d1bb8fb290e04b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dbcb652e17dadb575562129e651ec602c6062677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dbcb652e17dadb575562129e651ec602c6062677"}]},"branch":"refs/heads/master"},"28080d117b48c276dcdb22356d3c4104125d3c48":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-12-24 07:22:10.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/91/822891/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/822891/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/822891/3"}}},"commit":{"parents":[{"commit":"aa798fc006e0d4bce38751af7b0c14d97b0fecfb","subject":"Merge \"Add hacking check D710 for LOG.warn()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aa798fc006e0d4bce38751af7b0c14d97b0fecfb"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-24 07:14:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-24 07:20:15.000000000","tz":-480},"subject":"Fix designate-manage not handling NS records properly","message":"Fix designate-manage not handling NS records properly\n\nThis patch addresses a few problems with the manage\ncommand and NS records. First of all it\u0027s possible\nto add NS records to a zone and that would break\nthe command.\n\nIn addition this bug fixes a scenario where the\nNS record at some point was manually edited and\nno longer a being a managed record.\n\nChange-Id: I4e6ea0b6b717d2a1b5cc420874d1bb8fb290e04b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28080d117b48c276dcdb22356d3c4104125d3c48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28080d117b48c276dcdb22356d3c4104125d3c48"}]},"branch":"refs/heads/master"},"cb3ab95dd8535a9bb18c6d1626f55d268003f06d":{"kind":"REWORK","_number":4,"created":"2021-12-24 07:36:47.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/91/822891/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/822891/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/822891/4"}}},"commit":{"parents":[{"commit":"aa798fc006e0d4bce38751af7b0c14d97b0fecfb","subject":"Merge \"Add hacking check D710 for LOG.warn()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aa798fc006e0d4bce38751af7b0c14d97b0fecfb"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-24 07:14:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-24 07:36:41.000000000","tz":-480},"subject":"Fix designate-manage not handling NS records properly","message":"Fix designate-manage not handling NS records properly\n\nThis patch addresses a few problems with the manage\ncommand and NS records. First of all it\u0027s possible\nto add NS records to a zone and that would break\nthe command.\n\nIn addition this bug fixes a scenario where the\nNS record at some point was manually edited and\nno longer a being a managed record.\n\nChange-Id: I4e6ea0b6b717d2a1b5cc420874d1bb8fb290e04b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cb3ab95dd8535a9bb18c6d1626f55d268003f06d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cb3ab95dd8535a9bb18c6d1626f55d268003f06d"}]},"branch":"refs/heads/master"},"ed6edb4403b7744ed0e6e080de916c64901ce0e6":{"kind":"REWORK","_number":5,"created":"2021-12-24 07:45:26.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/91/822891/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/822891/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/822891/5"}}},"commit":{"parents":[{"commit":"aa798fc006e0d4bce38751af7b0c14d97b0fecfb","subject":"Merge \"Add hacking check D710 for LOG.warn()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aa798fc006e0d4bce38751af7b0c14d97b0fecfb"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-24 07:14:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-24 07:44:36.000000000","tz":-480},"subject":"Fix designate-manage not handling NS records properly","message":"Fix designate-manage not handling NS records properly\n\nThis patch addresses a few problems with the manage\ncommand and NS records. First of all it\u0027s possible\nto add NS records to a zone and that would break\nthe command.\n\nIn addition this bug fixes a scenario where the\nNS record at some point was manually edited and\nno longer being a managed record, cauing the command\nto fail because the recordset already exists.\n\nChange-Id: I4e6ea0b6b717d2a1b5cc420874d1bb8fb290e04b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ed6edb4403b7744ed0e6e080de916c64901ce0e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ed6edb4403b7744ed0e6e080de916c64901ce0e6"}]},"branch":"refs/heads/master"},"5804d76d7b9b211e19fbbb9d437b627f6c10ec77":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2021-12-24 07:49:18.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/91/822891/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/822891/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/822891/6"}}},"commit":{"parents":[{"commit":"aa798fc006e0d4bce38751af7b0c14d97b0fecfb","subject":"Merge \"Add hacking check D710 for LOG.warn()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aa798fc006e0d4bce38751af7b0c14d97b0fecfb"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-24 07:14:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-24 07:48:35.000000000","tz":-480},"subject":"Fix designate-manage not handling NS records properly","message":"Fix designate-manage not handling NS records properly\n\nThis patch addresses a few problems with the manage\ncommand and NS records. First of all it\u0027s possible\nto add NS records to a zone with a different name and\nthat would break the command.\n\nIn addition this bug fixes a scenario where the\nNS record at some point was manually edited and\nno longer being a managed record, cauing the command\nto fail because the recordset already exists.\n\nChange-Id: I4e6ea0b6b717d2a1b5cc420874d1bb8fb290e04b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5804d76d7b9b211e19fbbb9d437b627f6c10ec77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5804d76d7b9b211e19fbbb9d437b627f6c10ec77"}]},"branch":"refs/heads/master"},"c09af7df4252997be07b3bb7223d6ac75c811ee2":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2021-12-27 00:13:28.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/91/822891/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/822891/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/822891/7"}}},"commit":{"parents":[{"commit":"aa798fc006e0d4bce38751af7b0c14d97b0fecfb","subject":"Merge \"Add hacking check D710 for LOG.warn()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aa798fc006e0d4bce38751af7b0c14d97b0fecfb"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-24 07:14:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-27 00:12:31.000000000","tz":-480},"subject":"Fix designate-manage pool update bugs","message":"Fix designate-manage pool update bugs\n\nThis patch addresses a few problems with the manage\ncommand for pools and NS records.\n\n- Fixed an issue where having multiple NS records\n  would break the pool command.\n\n- Fixed a scenario where manually (non-managed) NS\n  records could break the pool command.\n\n- Fixed a potential edge case that could break the\n  pool command.\n\nThe biggest change is that we now only manage the NS\nrecord for the zone itself. This was always the case,\nbut because we didn\u0027t check for this specifically,\nother NS records would conflict with the command.\n\nChange-Id: I4e6ea0b6b717d2a1b5cc420874d1bb8fb290e04b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c09af7df4252997be07b3bb7223d6ac75c811ee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c09af7df4252997be07b3bb7223d6ac75c811ee2"}]},"branch":"refs/heads/master"},"fa12671c671b325dd48e32e092d2f2007d211ee6":{"kind":"REWORK","_number":8,"created":"2021-12-27 19:16:28.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/91/822891/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/822891/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/822891/8"}}},"commit":{"parents":[{"commit":"aa798fc006e0d4bce38751af7b0c14d97b0fecfb","subject":"Merge \"Add hacking check D710 for LOG.warn()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aa798fc006e0d4bce38751af7b0c14d97b0fecfb"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-24 07:14:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-27 19:16:08.000000000","tz":-480},"subject":"Fix designate-manage pool update bugs","message":"Fix designate-manage pool update bugs\n\nThis patch addresses a few problems with the manage\ncommand for pools and NS records.\n\n- Fixed an issue where having multiple NS records\n  would break the pool command.\n\n- Fixed a scenario where manually (non-managed) NS\n  records could break the pool command.\n\n- Fixed a potential edge case that could break the\n  pool command.\n\nThe biggest change is that we now only manage the NS\nrecord for the zone itself. This was always the case,\nbut because we didn\u0027t check for this specifically,\nother NS records would conflict with the command.\n\nChange-Id: I4e6ea0b6b717d2a1b5cc420874d1bb8fb290e04b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa12671c671b325dd48e32e092d2f2007d211ee6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa12671c671b325dd48e32e092d2f2007d211ee6"}]},"branch":"refs/heads/master"},"bff3d5f6e31fe595a77143ec4ac779c187bf72a8":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2022-01-06 03:51:17.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/91/822891/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/822891/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/822891/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/822891/9"}}},"commit":{"parents":[{"commit":"97339d78e245e053fa8c792ed3d3649f369623ca","subject":"Merge \"Updating python testing classifier as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/97339d78e245e053fa8c792ed3d3649f369623ca"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-24 07:14:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 03:51:17.000000000","tz":0},"subject":"Fix designate-manage pool update bugs","message":"Fix designate-manage pool update bugs\n\nThis patch addresses a few problems with the manage\ncommand for pools and NS records.\n\n- Fixed an issue where having multiple NS records\n  would break the pool command.\n\n- Fixed a scenario where manually (non-managed) NS\n  records could break the pool command.\n\n- Fixed a potential edge case that could break the\n  pool command.\n\nThe biggest change is that we now only manage the NS\nrecord for the zone itself. This was always the case,\nbut because we didn\u0027t check for this specifically,\nother NS records would conflict with the command.\n\nChange-Id: I4e6ea0b6b717d2a1b5cc420874d1bb8fb290e04b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bff3d5f6e31fe595a77143ec4ac779c187bf72a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bff3d5f6e31fe595a77143ec4ac779c187bf72a8"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
