)]}'
{"id":"openstack%2Fdesignate~823108","triplet_id":"openstack%2Fdesignate~master~I897d7da185c2dd246b80d1a598e9e8a5c667304e","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"28609":{"account":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"last_update":"2022-01-05 18:27:08.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-01-01 13:35:39.000000000","reason":"Dr. Jens Harbott (frickler) replied on the change"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-01-05 17:19:33.000000000","reason":"Michael Johnson replied on the change"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-01-05 18:27:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I897d7da185c2dd246b80d1a598e9e8a5c667304e","subject":"Modernize PTR implementation in Central","status":"MERGED","created":"2021-12-28 20:43:24.000000000","updated":"2022-01-05 18:28:52.000000000","submitted":"2022-01-05 18:27:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":27,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823108","meta_rev_id":"8cb85458e90c6f03b90d5a36dae6a6248f1ca373","_number":823108,"virtual_id_number":823108,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-05 18:27:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-01-01 13:35:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-05 17:19:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-05 17:19:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-01-01 13:35:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2022-01-05 17:19:51.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-29 00:31:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-29 01:36:20.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-12-29 01:36:20.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-12-29 01:36:20.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"state":"REVIEWER"}],"messages":[{"id":"479e6368be215c2a72b6a8e903bd95c7aa93a21d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-28 20:43:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86dc939809a9b36fcae5516b71e795246c16ec2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-28 21:25:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"48ea981074060f1a2baef20df72f6b82e1cfb738","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-28 22:15:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"011cba8577250eb857cee4ffe92f1c63e183890e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-28 22:21:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1fb5179bb4305e6296e6f5dc25e332ed07792fca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-28 22:22:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c44575a3d3032c338569fdacebea40b465da49a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-28 23:19:56.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e1f6347308696aaeca297dc5e565dba72e15999b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-28 23:25:11.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"29b7fb98efdbdbf62fb67c233eea905dea06eba4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 00:31:32.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/00b5f2ab579445dc88c78fdc463dbf3d : SUCCESS in 44m 45s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/0c158c5146864825856a601d1b98dfa6 : SUCCESS in 59m 56s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3dee08d088af4475b38651b5f4b338ea : SUCCESS in 41m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/362c1109660541f19fde41c3703c2285 : SUCCESS in 41m 07s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/02e4aa661c844068b14ed86f0da75464 : SUCCESS in 1h 01m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1bd6dc4300a944d09ef27b6dae93ccb2 : SUCCESS in 42m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a73190e26b9d47719b03b1091c09c915 : SUCCESS in 42m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea87b57e1cf5422d94a3fc10145b4521 : SUCCESS in 5m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3de2e2f57f594872bb4d199dd9a00d57 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c427392027640d0be36cdb7444436ea : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9601e8cdab9c4b1fbac34d9f4c5dc1d1 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9eebbc6b189431489577286332c7dd2 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81475deb70ff4618b614873ef42f9ea1 : SUCCESS in 8m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c7fbf01512224997bd2a806cd0c6e101 : SUCCESS in 25m 42s","accounts_in_message":[],"_revision_number":7},{"id":"0ad618f5d4271ead952fa540255fee70cbbd305b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-29 00:34:44.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"d28ded8ab8fe6894551ea703ef402bb2c279372c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 01:34:39.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/028c5dd344da4fe9ae3f2c47f86b682d : SUCCESS in 45m 32s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/da79b529d0be414f8295984d7e366c5a : SUCCESS in 37m 04s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/28c95d9d69c248b58717caa98c1525d9 : SUCCESS in 34m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/07985cb2595d4abeaec9f6de245bb84c : SUCCESS in 30m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7fe0176cf54e4d01b45b3d5e6377141f : SUCCESS in 44m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2bbe65189065466091ce15334b2de1ad : SUCCESS in 30m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8615ef9ae90e4017ad189bd9748f1da3 : SUCCESS in 29m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8dcfa2be4ae54a7db231350bf43e65ac : SUCCESS in 4m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fcae4fd122947b5b60897f45d287a47 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ac93ca7edd34c229e88de2a585e02b5 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff6098834bf84b5187400f1937fe8403 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6deaead56c3f47d89dcdabb24adf9c5a : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e2189cca3b04ca587c49435f1487f9c : SUCCESS in 7m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/98ac8f2f38704566b257c852dc4708c6 : SUCCESS in 32m 32s","accounts_in_message":[],"_revision_number":8},{"id":"a3a24bdb40680b2a4a4b0220fd9de6c7dbf5fd84","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-29 01:35:37.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"dd5aa62627f7724c91f48d2e4c128d4013605290","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-29 01:36:20.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"94b7fc2549c7d9f8bbde1b8603b94f5970910017","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 02:20:11.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/89341ea7b73a481c9e33c4877cc1d956 : SUCCESS in 28m 37s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/3158181113924990bf25efeedb283229 : SUCCESS in 37m 17s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/d680c58c2511434fbc8c072bf23a7aa2 : SUCCESS in 31m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4da4225cd0664b0b9267a242e61148d2 : SUCCESS in 32m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f2cb3ad81bab42469d20de3ec765f2fd : SUCCESS in 42m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8585225b1442419eb0dc97e705181107 : SUCCESS in 29m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c03c57cdb6574f3fa0fea92d32cb46dd : SUCCESS in 32m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13add6c7718f4ab3a4f2fda4f33e1248 : SUCCESS in 5m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0b8dae34f7f4252a36af15a8564d29d : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e93b8b5d76049b9a14246b5226200aa : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9932085ab2b4fe089ebe021df2ccd29 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c78484a4503243b4a0df36d6ab741f74 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f70d5dc8c22342078f556062296fe478 : SUCCESS in 9m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4ed87310a3b14ab7b074c4f187941aa2 : FAILURE in 35m 21s","accounts_in_message":[],"_revision_number":8},{"id":"e2f62fb96e8fc9b4fa19b6b2375c47dc11092f15","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-29 02:47:01.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3815469239d777ee60ab59dec4131e34333608ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 04:00:13.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/244b85ad5f284568bdf2dcbd0ab304f6 : SUCCESS in 42m 48s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/426cb232b04e4ac79dbd444b3929597f : SUCCESS in 51m 22s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/5aed054fb2504fd282e3f844e3238a45 : SUCCESS in 34m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/087f588c07734af2966d56e6954a1f2b : SUCCESS in 39m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a8ad1cc339a84562b987c8311de2ee5c : SUCCESS in 59m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e7e30e0d7fc14acb92f5491b1eae6fb1 : SUCCESS in 38m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/17f8ccd5a03349c382e1cfab8f95857f : SUCCESS in 40m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88dd1dde0af34abb84b1fc996ad4424c : SUCCESS in 7m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70d9d4f6dc784254bf958db3083c5684 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/989dfc01c19d4c6a85df59e8ba77dc4a : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/969a59faef814deebe9d752c4aa574da : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d22cc9d8eb74b8fb149a0f8afc61286 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dd3c4c703af42a88e93effe65b7d32e : SUCCESS in 9m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/251d29a167144cb2824d22eeba9a3cfb : SUCCESS in 30m 06s","accounts_in_message":[],"_revision_number":8},{"id":"e9c9c6943d8fdcda3df2f0b5b29f476ff4b3c2f2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-29 04:15:37.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"deb555dff3d4ae1fa9f853180767187969050c0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 05:28:25.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1e4de20484664cf1bccde371823a8890 : SUCCESS in 50m 10s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/3e59ed166d0b42b0990447f6a82f3137 : SUCCESS in 1h 07m 42s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/783f6a04e85548be808c4098e36b445f : SUCCESS in 43m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/030d0e55aceb46cf9e1275c0f7e385b7 : SUCCESS in 41m 12s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/62006f29141949c098c8cce0089cd9a9 : SUCCESS in 57m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a3f0069e55954769894c4962d885e797 : SUCCESS in 41m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b9b08f1f48174457af79ce422637edfd : SUCCESS in 29m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd917b28dc2247a3afbb2b71109eeb39 : SUCCESS in 5m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ea8e6a0eed04fdda108e94fe56a3e51 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa2342a19ffd41678931a214a9a71c83 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d303a9c6949e45a3ae7f0996db090596 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3d7de1e98d4492c83e15326b2b01d79 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e3170f455a34d04a1b9e1f671a64141 : SUCCESS in 7m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3c4604b1e7af49b59a9a0c688063055f : SUCCESS in 28m 51s","accounts_in_message":[],"_revision_number":8},{"id":"9c8f1c232d030c69ca2cb957111f0e081a030573","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-29 05:33:41.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2fd326d331e64af25cd1da88d218463e4c71dbf3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-29 06:35:25.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b76f2f83a85274a3cb2a8439d8ce2db90c024820","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 06:52:46.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0d03e464d9f44cb68479fdb0f22f0855 : SUCCESS in 36m 08s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/8d3a72ba5f554e38abd723a30671c734 : SUCCESS in 53m 30s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/d53e66180e3f4b2982acff6dffc6b675 : SUCCESS in 35m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b9c5e522ff8648988a868d7ccf60cbfe : SUCCESS in 37m 52s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4079dcd3a65f4c8bb6cbb546c59512c0 : SUCCESS in 1h 00m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/41ebb1de5bb64110bc95610325ca099b : SUCCESS in 56m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/48d3cfad0c714811814383afd77d90c9 : SUCCESS in 45m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61f0fd1970294ea0b291ad21220ac2d5 : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe6c7e346cb4432692b8315369f1658d : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4ce16181d35443294b0a18fb94ef598 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38ef0004f8f5426c91409641aec7b54e : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fa8afdc9ec8441295c6deb6159934de : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/267c823a023c4aecbb698adffbec66ad : SUCCESS in 7m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ce826fef83a54829aa9f3b2ed54d5db4 : SUCCESS in 30m 04s","accounts_in_message":[],"_revision_number":8},{"id":"7ad029ade8f679afeeea2aef41ec0ccf5c8e73c7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-29 07:40:07.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"721cae6c969bb76b6f58f1424dce89c6244b7862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 08:29:05.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/99aedc247a034b79b40e9bf282933486 : SUCCESS in 44m 11s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/d941a90d185b46b9995af560f0deccd6 : SUCCESS in 41m 23s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3d0b41addd2249e88afa49892a26b6be : SUCCESS in 42m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f62562722ea94f3db50acf47dd621460 : SUCCESS in 42m 18s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/811be2eee391472bb46f0ca1639ceadf : SUCCESS in 46m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/27a3490efe0d4773af6d2356feb9757b : SUCCESS in 42m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8b17c78a28a84c1a899d6a6671f20a1e : SUCCESS in 43m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc2fe593c0a3456dbed9f61ae961163b : SUCCESS in 5m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31e2aad2aa834937bcce6c8658942c7f : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b48ad86ae274d6c8070154d5f41b745 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/175703d96ebc462aa1949d89a8dd5a35 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa217fdebc6f4976ba04c90894e36678 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56f418f52019418e9119ab284104979a : SUCCESS in 7m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bee0cf22869b46e1b87449060573604c : SUCCESS in 27m 24s","accounts_in_message":[],"_revision_number":8},{"id":"1b553205f9e7976e8ebd907d96fe2653005bee4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-29 08:44:21.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"342c9d5d82051a1f725642285c5ce56db2f252a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-29 08:49:07.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"ab65004656ae9ff6fa198847db978c89b7991bf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-29 08:50:58.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"000781aa42a684b92f58da2e1e7469b9ab396f07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 09:42:34.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/083e926ba23a4ce0a6ab1183ae4581e6 : SUCCESS in 31m 01s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/8364c9e850874a53a8ab830e04dca3c6 : SUCCESS in 41m 03s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8f8bbc9ee0744533ad466ce0a2e67dd2 : SUCCESS in 45m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ebf66af71ca947929c7dfb2defff6370 : SUCCESS in 35m 43s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d55bc7d5761d4dacab5c31d86409ea32 : SUCCESS in 46m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/929678a17fb84e8daf256f9d58d0f465 : SUCCESS in 29m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e9a1f24f729e46b3ae51b28e96b705eb : SUCCESS in 30m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d88045f7c42b4580a4b973911380ca1f : SUCCESS in 5m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a682dc721b4344c39eb16690e0a8a910 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5440803f5d10417597baa0ccbb36720a : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/360c088672c24df788569b84d917ce4e : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7e453a9de0642299e83bf8952e9cb19 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05240e3cc29c4568b72b4fcf77a15ae1 : SUCCESS in 10m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/872c02d48aca4f5bab6152ff7ab12805 : SUCCESS in 32m 24s","accounts_in_message":[],"_revision_number":11},{"id":"52435733721c1554228d3d01f65acee96d5880d3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-29 18:21:48.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"ff68b30af5dea05b659c9948b91956fe613abfa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-29 18:58:14.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"6a52739c2ab9e0897238c29db4f7fcbcecb3564e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-29 19:00:21.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"66fd6ed32fefd5b8e9e8020f8e24063613efaad5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 19:49:34.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4baa44f9214445149a87b1a5337a8d3f : FAILURE in 21m 16s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/417f4a54643e44d180108ca28a082e56 : SUCCESS in 36m 34s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/dc8b291fe2434c0a8bb97f6eb046e5dc : FAILURE in 22m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c50fd676a6994201a28adb3719e8e2c9 : FAILURE in 19m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8370b4d6c2c64906aeeea412d0df0379 : FAILURE in 47m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/698b9ba4477d48cc8a04fe3a1f321dfd : FAILURE in 19m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e83780534ea548708fe29f68fd5a1267 : FAILURE in 24m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bfb7bddfd6b49d7891afc0cd8d4c096 : SUCCESS in 5m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b427209955e34f859964ad6692dd77a2 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/135b6cb358f94a6689b8266dd6f274bd : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b096c2cc7544308a706082c8ec32797 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edd116415c064390a57b39de082c4e74 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/470502355b724995ae236bb2ba60ea1e : SUCCESS in 8m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/151c1ee852c0475096f77b6f2287af7b : FAILURE in 27m 32s","accounts_in_message":[],"_revision_number":13},{"id":"e4ef46760f54a8c47f5c6e52f7171a0df1a14614","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-29 20:30:36.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"d1f714ba88056b7f799c7c1c2cd9649429cc8517","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-29 21:09:34.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"9346f3948948c50584562be74c0c6e4ca370dfd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 22:11:58.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8d0ec2630553471bb7514dba00eaeab4 : SUCCESS in 44m 34s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/730710812dac40be93feed330cbbb7d5 : SUCCESS in 52m 32s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/dfe87da4050b4db995286523205597c6 : SUCCESS in 46m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ebd4b30a0e4d4ea383afaf5e7c8431f5 : SUCCESS in 42m 57s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3698a937ec914f619e8283474c853423 : SUCCESS in 57m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/19e8a97a49bd4936979bac8245198383 : SUCCESS in 40m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/28ed44fca08642a5a6c99d35b629ca85 : SUCCESS in 39m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/159ed30ad6714e71b738676332f0d565 : SUCCESS in 6m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93f2540f0818408d83ad1172808d42ce : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0530590378f0403e95ba386156e1917a : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0be7cfed6cd4ca296c17e1908dd41ee : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b901b8f3881c4e56bff75c8d68fd5c50 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c16df392b302421fb330e929124e7a0b : SUCCESS in 7m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b120b114ef674c7c9fb35dbe7d13caa2 : SUCCESS in 29m 53s","accounts_in_message":[],"_revision_number":14},{"id":"ed61b02eaf0498d9c468901ff50db4c43ca11d44","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-29 23:52:48.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"77774b9a487ecdd965c7d2d2bead7984da6085a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-29 23:59:13.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"6f7cde8a5a839f7b6e26d99a2e61c461b6ee18e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-30 00:01:27.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"e2d5676cfebde15c0569ffa40d334f0d63f96efb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 00:47:52.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f633d0c0f35c4e2fa3c8e627d8dfdd7b : SUCCESS in 41m 37s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/89edb755c1474c40924a16b1736cf034 : SUCCESS in 35m 15s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/11564b019dc043b985c6757ecdef5471 : SUCCESS in 29m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f3f2285a208e4a7ca0a30b4c1a32d625 : SUCCESS in 27m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/753c54d34367471b96d8ed6d27719249 : SUCCESS in 44m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cd5a2423c37345b6a9d86dcb2f3d20fe : SUCCESS in 28m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9ce728354ba64837bf99996f0fc39749 : SUCCESS in 30m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a06564e81d994a03ab1d6fe9bd4244d6 : SUCCESS in 4m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f62b7532e274b0ca2d71f4b06cb0824 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cd6661c1946467a916d4c01c9fe2671 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf524ad59efd4e588242134ca620d738 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d06f807b8114325840742a1eb44556d : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9220a1e1de048dc8e8f3488e99b28a8 : SUCCESS in 6m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a6f9527aecfe46aaae9b2f3dacb8ed86 : SUCCESS in 29m 20s","accounts_in_message":[],"_revision_number":16},{"id":"56c2636fb830de8415281ad157a86d96b85dab04","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-31 12:31:31.000000000","message":"Patch Set 16:\n\n(4 comments)","accounts_in_message":[],"_revision_number":16},{"id":"f8e89f5e59054c0b1be06cb8437305b89b3d194f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-31 19:18:45.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"95e4250ee90d10808cc1ec3fcf7adc62ec6fca47","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-31 19:20:40.000000000","message":"Patch Set 17:\n\n(3 comments)","accounts_in_message":[],"_revision_number":17},{"id":"136a8a21c27db0abdbf4bb77254c4e0640d6f9f8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-31 19:23:02.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"92abbb3cfb8cea803f7913a08d374b32c4c04bf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-31 20:28:28.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/16f454cd5acf43c49f4943ee70f5ccca : SUCCESS in 57m 35s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/69e1749aeba24877a2c76534c48af7ca : SUCCESS in 46m 58s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/c9714f55889f4be1b648fbdab3916cf0 : SUCCESS in 39m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c527da36c77a47bcb51ad7134b4650f4 : SUCCESS in 29m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0b411fbeeba5443c95910de0cb8c8d2d : SUCCESS in 43m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9bad9c5bcc544088b123287a3a7aa9d3 : SUCCESS in 31m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7391076de727402c91a8dc005b41a965 : SUCCESS in 28m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/710cc9c701a040abbb8415ed32ed3ef7 : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e22e588e6b264520ba9d84253c141a19 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1af4acb15dd546eea48d9118de1cccab : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9923b7da9fe040ccb8dbfad2d36293df : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca95e871159a4a559be3c2e224fdc385 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/401eeb1f082a487595a5c47aaca94c9a : SUCCESS in 14m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/091675b91d0046c9b496f51f5498f7a2 : SUCCESS in 26m 13s","accounts_in_message":[],"_revision_number":17},{"id":"54e7eb2e3d9bdb9a852daa7dc3f1b2dd63eddbeb","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-01 00:37:56.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"b9a66e7d8436e36eb35c91ac4ed8be95045484a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-01 01:42:48.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"7cba2c0f2f8ea4f75522f5a3a02b70dc15f8d5f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-01 01:52:52.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"640d6789f30188ed1764ba693ffa615207b5ea90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-01 02:48:48.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1f5345f4e8f64407951188dfe86b37a8 : SUCCESS in 32m 46s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/19b6e89f9c00473699aaab792c2b989b : SUCCESS in 51m 03s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/17e4168e19bf433a9c41011dc2fbb93b : SUCCESS in 46m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ef16937280fa47aba27b708aefeeb0b9 : SUCCESS in 41m 23s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/38a8f115f1b74e5eaf963bd4a916cae0 : SUCCESS in 55m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/106ea1251a0d4e95bcd38e4edd8ce5e6 : SUCCESS in 34m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6406a34e93774fcabd6b54814861aff8 : SUCCESS in 35m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a2997ff23654ba79212fd0d82435585 : SUCCESS in 4m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90fa1b421c9b4323a20556225f54f212 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b2a80e18c554892bd9214cb4cb6b019 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/815e527d539947bbb99c5ab12c4048ed : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/043200353e3f411f9d35f3d96755b993 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e07df42e969e4192ad589fdcb9247b49 : SUCCESS in 7m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7ef4b4017e1a40ccafcd700fbc4db055 : SUCCESS in 32m 22s","accounts_in_message":[],"_revision_number":19},{"id":"7504c729cd2e8fd957346ffb5f5b568832154adc","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-01-01 13:35:39.000000000","message":"Patch Set 19: Code-Review+2 Review-Priority+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"0e2842bd20568cc091e7774832bb935aa97838b5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-02 04:18:20.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"a2293bb99cb0f278c9f930f779fc4121e15cbbef","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-04 00:46:46.000000000","message":"Topic ptr2 removed","accounts_in_message":[],"_revision_number":19},{"id":"b551976b1105206b0820eb7a1c7ed2e62828c724","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-04 19:05:24.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"24ee1036f809657925d331b65452e4f4f03e225f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 20:11:15.000000000","message":"Patch Set 19:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2dfdbb5198934c2a997068a01c9e40c5 : SUCCESS in 50m 38s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/09c27c6058bb4bf9899ddf1affddcfe6 : SUCCESS in 54m 33s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/0ddec400b011410eab64327b101fbe41 : SUCCESS in 45m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d2ff43aa77264b369361286f4a175cf1 : SUCCESS in 39m 03s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/60d737901c934a37bfc3eeeb88b8ed1c : SUCCESS in 1h 00m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/887739159c184c4596a4f94a676c92e5 : SUCCESS in 41m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0e16f3b926f244b69574687790eecfaf : SUCCESS in 40m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/798ed7d53fc44c66928eacbdc7c2920f : SUCCESS in 6m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b76c3e937674957a09c58ce73270177 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9467c119ffc4318a89e37abc4b2fba4 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86407d1e642644c7a0e4e07a64f053f3 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16659d47844a4e62912f66ee0a34399c : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e572b461cf74b38b7fb74995d14e1d8 : SUCCESS in 7m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ea978e52da4540ac92a5b6613af45cdd : SUCCESS in 34m 00s","accounts_in_message":[],"_revision_number":19},{"id":"81ddd437b0de2833d04c351b49d89eb64a1e5802","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-04 20:12:41.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"d28041e0bd7484cd386ec8706df139ea495391af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 21:14:42.000000000","message":"Patch Set 19:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/866e746af1f448249a9a8310b4c19545 : SUCCESS in 38m 15s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/5b687c95a900486d9c610ac4ebf504fc : SUCCESS in 49m 32s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/195c92b47d9549ccad9d923ead1dc9f1 : SUCCESS in 41m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/89d1aa533c6342729c3a4c72ffd97507 : SUCCESS in 39m 39s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6f87e9b480dd4336a50aa8736155a896 : SUCCESS in 59m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/39a6e91c46534da3b9d56257f10fec37 : SUCCESS in 36m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/de0dd33949064311beefa631fc7490ae : SUCCESS in 40m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac53859359a24698aef0763d33072492 : SUCCESS in 6m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ea446c7e9e34d549a71868e0e08d591 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4a8b644307e4c62bb849fd585725668 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2f465f8c2f2400ba68bb10cf3154fb5 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d33cb6b4f0446c5a607399a4bfd9b38 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ad2f0957ea448e6b5b98c671d8b2b62 : SUCCESS in 7m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/88a52a36d3184691b9be6afa4c3f7322 : SUCCESS in 28m 51s","accounts_in_message":[],"_revision_number":19},{"id":"059f39ffba715193790f44ed3c940d8b60ec5ec9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-04 21:15:37.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"e426b4543d15299fe44f921eab45b4638d9296f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 22:20:49.000000000","message":"Patch Set 19:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cfd76360f94246df9bbf5a2ba888cc63 : SUCCESS in 38m 27s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/4b256b71376946599558d4455401013a : SUCCESS in 1h 03m 59s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/751a8ae32821451e9d4b72b52fdcab5d : SUCCESS in 33m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3d4ea2b106674a9e80aac769ac6c871c : SUCCESS in 36m 22s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/653fc0afb6c346d7995fc1dbdfd80e7f : SUCCESS in 51m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cacb2aeedd7048ec8f121467ac3fd938 : SUCCESS in 32m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/31af8bcc79ac42448ec74d5c41518335 : SUCCESS in 37m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c1ffe10b4cf4e1aaa8ce061be5d4b30 : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe8c0593d95a43cbb6d26c4117077fc1 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c83790f4cd94b0b92f0fc4b03af88fa : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/303f8bbe20564d61b7a244c05d1e5ff2 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa79627a0a0741c3adeb82bd48be1e8a : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c96925421b1c4da7bb2f1df56c9c590a : SUCCESS in 7m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/17cc646c259345e38e49202582817e2c : SUCCESS in 32m 55s","accounts_in_message":[],"_revision_number":19},{"id":"4c8006bce8f3fc9fb045959a9247845ccb619cc1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-05 17:19:33.000000000","message":"Patch Set 19: Workflow+1 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"74f2b44800bf20d20b6afee8353e9d3171613183","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-05 17:19:51.000000000","message":"Patch Set 19: Code-Review+2 -Review-Priority","accounts_in_message":[],"_revision_number":19},{"id":"68f7c0c1a0ae550254cbf6db87db19ed1251059f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 17:20:27.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"2b5dcaf810056185ab87fa4e60fde060597faf34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 18:27:08.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bdee5e8a2419412386e37e9d12a76d80 : SUCCESS in 37m 29s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/d109b0204b4148c994e50e2233d4a471 : SUCCESS in 39m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0ce78653b51745b29668b0c56e8d06b9 : SUCCESS in 42m 27s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a9af05c945ed4dbc84caf268d962f6a4 : SUCCESS in 1h 02m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1b33abd104b54432a0751f5bae8e36cd : SUCCESS in 42m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8957053aaef34cdd8580998e43b31826 : SUCCESS in 40m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be8bf5a967334ee78fd070014c17cc87 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5941d1ceb194868b1bebd4f0cb86fb4 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac058e90385a4acdb246f89d2a26857b : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d863ebf1ae3491ab0d44e33c43531d5 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62b7c8caeffb4d3db9bac8c07fcf24df : SUCCESS in 8m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4d6f1325664446d0886f5ab1a46715df : SUCCESS in 28m 36s","accounts_in_message":[],"_revision_number":19},{"id":"5a9d26fafbd55340daf7198f79eccc7612966e34","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 18:27:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":19},{"id":"8cb85458e90c6f03b90d5a36dae6a6248f1ca373","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 18:28:52.000000000","message":"Patch Set 19:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b69b0bfd9e44ad28b1b5c62fec41df8 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"7ec7adc44a9914ad9492014ca0edec2341fb5de0","revisions":{"5e438f05410612aa2e93fb8428a157f6e6de3cdb":{"kind":"REWORK","_number":1,"created":"2021-12-28 20:43:24.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/823108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/823108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/823108/1"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 20:41:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 20:41:46.000000000","tz":-480},"subject":"[WIP] PTR","message":"[WIP] PTR\n\nChange-Id: I897d7da185c2dd246b80d1a598e9e8a5c667304e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5e438f05410612aa2e93fb8428a157f6e6de3cdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5e438f05410612aa2e93fb8428a157f6e6de3cdb"}]},"branch":"refs/heads/master"},"0a39cc24f28aaf54d20be7f30bb0bd6722968e41":{"kind":"REWORK","_number":2,"created":"2021-12-28 21:25:16.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/823108/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/823108/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/823108/2"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 20:41:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 21:25:08.000000000","tz":-480},"subject":"[WIP] PTR","message":"[WIP] PTR\n\nChange-Id: I897d7da185c2dd246b80d1a598e9e8a5c667304e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a39cc24f28aaf54d20be7f30bb0bd6722968e41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a39cc24f28aaf54d20be7f30bb0bd6722968e41"}]},"branch":"refs/heads/master"},"751b0036ed26a29917fd9896357af79bfcd6068c":{"kind":"REWORK","_number":3,"created":"2021-12-28 22:15:44.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/823108/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/823108/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/823108/3"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 20:41:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 22:15:28.000000000","tz":-480},"subject":"[WIP] PTR","message":"[WIP] PTR\n\nChange-Id: I897d7da185c2dd246b80d1a598e9e8a5c667304e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/751b0036ed26a29917fd9896357af79bfcd6068c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/751b0036ed26a29917fd9896357af79bfcd6068c"}]},"branch":"refs/heads/master"},"d3a42a144dc71ff2ff97d383a0e2cea60f6e1c87":{"kind":"REWORK","_number":4,"created":"2021-12-28 22:21:49.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/823108/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/823108/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/823108/4"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 20:41:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 22:21:40.000000000","tz":-480},"subject":"[WIP] PTR","message":"[WIP] PTR\n\nChange-Id: I897d7da185c2dd246b80d1a598e9e8a5c667304e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d3a42a144dc71ff2ff97d383a0e2cea60f6e1c87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d3a42a144dc71ff2ff97d383a0e2cea60f6e1c87"}]},"branch":"refs/heads/master"},"4dfe82ae77837423dfde2f783ca1a9a496c458fc":{"kind":"REWORK","_number":5,"created":"2021-12-28 22:22:56.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/823108/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/823108/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/823108/5"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 20:41:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 22:22:43.000000000","tz":-480},"subject":"[WIP] PTR","message":"[WIP] PTR\n\nChange-Id: I897d7da185c2dd246b80d1a598e9e8a5c667304e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4dfe82ae77837423dfde2f783ca1a9a496c458fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4dfe82ae77837423dfde2f783ca1a9a496c458fc"}]},"branch":"refs/heads/master"},"36989ea7454f49ad7c05f491a91d8a18d9fc2ee2":{"kind":"REWORK","_number":6,"created":"2021-12-28 23:19:56.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/823108/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/823108/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/823108/6"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 20:41:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 23:19:48.000000000","tz":-480},"subject":"[WIP] PTR","message":"[WIP] PTR\n\nChange-Id: I897d7da185c2dd246b80d1a598e9e8a5c667304e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/36989ea7454f49ad7c05f491a91d8a18d9fc2ee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/36989ea7454f49ad7c05f491a91d8a18d9fc2ee2"}]},"branch":"refs/heads/master"},"fcf7b1267e5a743819b4f32406deeaa6f4b10323":{"kind":"REWORK","_number":7,"created":"2021-12-28 23:25:11.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/823108/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/823108/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/823108/7"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 20:41:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 23:25:04.000000000","tz":-480},"subject":"[WIP] PTR","message":"[WIP] PTR\n\nChange-Id: I897d7da185c2dd246b80d1a598e9e8a5c667304e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fcf7b1267e5a743819b4f32406deeaa6f4b10323"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fcf7b1267e5a743819b4f32406deeaa6f4b10323"}]},"branch":"refs/heads/master"},"0ee89a7144525a144136ef7d3108b06ff2be8723":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2021-12-29 00:34:44.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/823108/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/823108/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/823108/8"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 20:41:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-29 00:27:26.000000000","tz":-480},"subject":"Modernize PTR implementation in Central","message":"Modernize PTR implementation in Central\n\nThe main objective of this change it to move\naway from the create/delete/update record and\nmake use of recordsets for all implementations.\n\nThe PTR code is the last code path making use of\n(create|delete|update)_record.\n\n- Fixed typos.\n- Use project_id instead of tenant_id.\n\nThe end goal is to stabilize the functional testing.\n\nChange-Id: I897d7da185c2dd246b80d1a598e9e8a5c667304e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0ee89a7144525a144136ef7d3108b06ff2be8723"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0ee89a7144525a144136ef7d3108b06ff2be8723"}]},"branch":"refs/heads/master"},"99da3930a90e0e083cdf1bdd9cc22bbce5991d3a":{"kind":"REWORK","_number":9,"created":"2021-12-29 08:44:21.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/823108/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/823108/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/823108/9"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 20:41:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-29 08:43:44.000000000","tz":-480},"subject":"Modernize PTR implementation in Central","message":"Modernize PTR implementation in Central\n\nUpdated the PTR (floating ips) code to behave closer to how we handle\nthis type of operations in designate.api.v2. This should resolve\nsome issues (e.g. race conditions) with the older floating ips code.\n\nAdditional changes.\n- Fixed minor typos in documentation.\n- Updated wording where it makes sense to use project instead of tenant.\n\nChange-Id: I897d7da185c2dd246b80d1a598e9e8a5c667304e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/99da3930a90e0e083cdf1bdd9cc22bbce5991d3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/99da3930a90e0e083cdf1bdd9cc22bbce5991d3a"}]},"branch":"refs/heads/master"},"379981aae35583b7dd5e146533091dc39d1c6035":{"kind":"REWORK","_number":10,"created":"2021-12-29 08:49:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/823108/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/823108/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/823108/10"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 20:41:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-29 08:48:59.000000000","tz":-480},"subject":"Modernize PTR implementation in Central","message":"Modernize PTR implementation in Central\n\nUpdated the PTR (floating ips) code to behave closer to how we handle\nthis type of operations in designate.api.v2. This should resolve\nsome issues (e.g. race conditions) with the older floating ips code.\n\nAdditional changes.\n- Fixed minor typos in documentation.\n- Updated wording where it makes sense to use project instead of tenant.\n\nChange-Id: I897d7da185c2dd246b80d1a598e9e8a5c667304e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/379981aae35583b7dd5e146533091dc39d1c6035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/379981aae35583b7dd5e146533091dc39d1c6035"}]},"branch":"refs/heads/master"},"4b5cb69a3a3b88875e3d8fa357107fbfa1888f8d":{"kind":"REWORK","_number":11,"created":"2021-12-29 08:50:58.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/823108/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/823108/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/823108/11"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 20:41:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-29 08:50:50.000000000","tz":-480},"subject":"Modernize PTR implementation in Central","message":"Modernize PTR implementation in Central\n\nUpdated the PTR (floating ips) code to behave closer to how we handle\nthis type of operations in designate.api.v2. This should resolve\nsome issues (e.g. race conditions) with the older floating ips code.\n\nAdditional changes.\n- Fixed minor typos in documentation.\n- Updated wording where it makes sense to use project instead of tenant.\n\nChange-Id: I897d7da185c2dd246b80d1a598e9e8a5c667304e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b5cb69a3a3b88875e3d8fa357107fbfa1888f8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b5cb69a3a3b88875e3d8fa357107fbfa1888f8d"}]},"branch":"refs/heads/master"},"75252219470d413747aa178915fa392efbb076e7":{"kind":"REWORK","_number":12,"created":"2021-12-29 18:58:14.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/823108/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/823108/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/823108/12"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 20:41:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-29 18:58:07.000000000","tz":-480},"subject":"Modernize PTR implementation in Central","message":"Modernize PTR implementation in Central\n\nUpdated the PTR (floating ips) code to behave closer to how we handle\nthis type of operations in designate.api.v2. This should resolve\nsome issues (e.g. race conditions) with the older floating ips code.\n\nAdditional changes.\n- Fixed minor typos in documentation.\n- Updated wording where it makes sense to use project instead of tenant.\n\nChange-Id: I897d7da185c2dd246b80d1a598e9e8a5c667304e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75252219470d413747aa178915fa392efbb076e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75252219470d413747aa178915fa392efbb076e7"}]},"branch":"refs/heads/master"},"c5aed6486e1881e2bf6e9c2f9d696e2914b4b58d":{"kind":"REWORK","_number":13,"created":"2021-12-29 19:00:21.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/823108/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/823108/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/823108/13"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 20:41:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-29 18:59:53.000000000","tz":-480},"subject":"Modernize PTR implementation in Central","message":"Modernize PTR implementation in Central\n\nUpdated the PTR (floating ips) code to behave closer to how we handle\nthis type of operations in designate.api.v2. This should resolve\nsome issues (e.g. race conditions) with the older floating ips code.\n\nAdditional changes.\n- Fixed minor typos in documentation.\n- Updated wording where it makes sense to use project instead of tenant.\n\nChange-Id: I897d7da185c2dd246b80d1a598e9e8a5c667304e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c5aed6486e1881e2bf6e9c2f9d696e2914b4b58d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c5aed6486e1881e2bf6e9c2f9d696e2914b4b58d"}]},"branch":"refs/heads/master"},"b9079fcc5c37729f1b213d107c42aa55ff27b6d1":{"kind":"REWORK","_number":14,"created":"2021-12-29 21:09:34.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/823108/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/823108/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/823108/14"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 20:41:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-29 21:09:21.000000000","tz":-480},"subject":"Modernize PTR implementation in Central","message":"Modernize PTR implementation in Central\n\nUpdated the PTR (floating ips) code to behave closer to how we handle\nthis type of operations in designate.api.v2. This should resolve\nsome issues (e.g. race conditions) with the older floating ips code.\n\nAdditional changes.\n- Fixed minor typos in documentation.\n- Updated wording where it makes sense to use project instead of tenant.\n\nChange-Id: I897d7da185c2dd246b80d1a598e9e8a5c667304e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9079fcc5c37729f1b213d107c42aa55ff27b6d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9079fcc5c37729f1b213d107c42aa55ff27b6d1"}]},"branch":"refs/heads/master"},"9ef93618d043781fbd6236848296790f875e4ca3":{"kind":"REWORK","_number":15,"created":"2021-12-29 23:59:13.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/823108/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/823108/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/823108/15"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 20:41:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-29 23:59:06.000000000","tz":-480},"subject":"Modernize PTR implementation in Central","message":"Modernize PTR implementation in Central\n\nUpdated the PTR (floating ips) code to behave closer to how we handle\nthis type of operations in designate.api.v2. This should resolve\nsome issues (e.g. race conditions) with the older floating ips code.\n\nAdditional changes.\n- Fixed minor typos in documentation.\n- Updated wording where it makes sense to use project instead of tenant.\n\nChange-Id: I897d7da185c2dd246b80d1a598e9e8a5c667304e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ef93618d043781fbd6236848296790f875e4ca3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ef93618d043781fbd6236848296790f875e4ca3"}]},"branch":"refs/heads/master"},"ce376c6cf219908344504778aed126238f7a11df":{"kind":"REWORK","_number":16,"created":"2021-12-30 00:01:27.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/823108/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/823108/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/823108/16"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 20:41:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-30 00:01:20.000000000","tz":-480},"subject":"Modernize PTR implementation in Central","message":"Modernize PTR implementation in Central\n\nUpdated the PTR (floating ips) code to behave closer to how we handle\nthis type of operations in designate.api.v2. This should resolve\nsome issues (e.g. race conditions) with the older floating ips code.\n\nAdditional changes.\n- Fixed minor typos in documentation.\n- Updated wording where it makes sense to use project instead of tenant.\n\nChange-Id: I897d7da185c2dd246b80d1a598e9e8a5c667304e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ce376c6cf219908344504778aed126238f7a11df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ce376c6cf219908344504778aed126238f7a11df"}]},"branch":"refs/heads/master"},"a32f8ccefb44e8d23dec8bd03d7d5ac60b6e1727":{"kind":"REWORK","_number":17,"created":"2021-12-31 19:18:45.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/823108/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/823108/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/823108/17"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 20:41:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-31 19:18:35.000000000","tz":-480},"subject":"Modernize PTR implementation in Central","message":"Modernize PTR implementation in Central\n\nUpdated the PTR (floating ips) code to behave closer to how we handle\nthis type of operations in designate.api.v2. This should resolve\nsome issues (e.g. race conditions) with the older floating ips code.\n\nAdditional changes.\n- Fixed minor typos in documentation.\n- Updated wording where it makes sense to use project instead of tenant.\n\nChange-Id: I897d7da185c2dd246b80d1a598e9e8a5c667304e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a32f8ccefb44e8d23dec8bd03d7d5ac60b6e1727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a32f8ccefb44e8d23dec8bd03d7d5ac60b6e1727"}]},"branch":"refs/heads/master"},"1a0e4dd84f95d5c144353abea9beed05cfd1da5e":{"kind":"REWORK","_number":18,"created":"2022-01-01 01:42:48.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/823108/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/823108/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/823108/18"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 20:41:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-01 01:42:41.000000000","tz":-480},"subject":"Modernize PTR implementation in Central","message":"Modernize PTR implementation in Central\n\nUpdated the PTR (floating ips) code to behave closer to how we handle\nthis type of operations in designate.api.v2. This should resolve\nsome issues (e.g. race conditions) with the older floating ips code.\n\nAdditional changes.\n- Fixed minor typos in documentation.\n- Updated wording where it makes sense to use project instead of tenant.\n\nChange-Id: I897d7da185c2dd246b80d1a598e9e8a5c667304e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1a0e4dd84f95d5c144353abea9beed05cfd1da5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1a0e4dd84f95d5c144353abea9beed05cfd1da5e"}]},"branch":"refs/heads/master"},"7ec7adc44a9914ad9492014ca0edec2341fb5de0":{"kind":"REWORK","_number":19,"created":"2022-01-01 01:52:52.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/823108/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/823108/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/823108/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/823108/19"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-28 20:41:46.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-01 01:52:44.000000000","tz":-480},"subject":"Modernize PTR implementation in Central","message":"Modernize PTR implementation in Central\n\nUpdated the PTR (floating ips) code to behave closer to how we handle\nthis type of operations in designate.api.v2. This should resolve\nsome issues (e.g. race conditions) with the older floating ips code.\n\nAdditional changes.\n- Fixed minor typos in documentation.\n- Updated wording where it makes sense to use project instead of tenant.\n\nChange-Id: I897d7da185c2dd246b80d1a598e9e8a5c667304e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7ec7adc44a9914ad9492014ca0edec2341fb5de0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7ec7adc44a9914ad9492014ca0edec2341fb5de0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
