)]}'
{"id":"openstack%2Fdesignate~823230","triplet_id":"openstack%2Fdesignate~master~I5fe06cf99bb9787ec3afc67ab60a7330fe281513","project":"openstack/designate","branch":"master","topic":"rootwrap","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-11 07:08:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5fe06cf99bb9787ec3afc67ab60a7330fe281513","subject":"Sync rootwrap.conf from oslo.rootwrap","status":"MERGED","created":"2022-01-02 12:24:34.000000000","updated":"2022-08-11 07:09:34.000000000","submitted":"2022-08-11 07:08:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823230-rootwrap","meta_rev_id":"ebae26792093c67faac1fcc3254b748d3d92b957","_number":823230,"virtual_id_number":823230,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-11 07:08:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-11 05:04:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-08-11 05:04:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-08-11 05:04:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-02 12:32:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-03 01:34:40.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-08-09 20:44:55.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"ae2485f1f7469e46891be669fc78b7db8176a5c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-02 12:24:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80c78fc46acdb6abac1b18cb7103d603002bb11a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-02 12:32:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/267931567f2d4cca9a3ca8859d180374 : SUCCESS in 5m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36301dfd6bb74750aea0e1e539f07040 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d20825cc701a488180437737e225a75a : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddb19053d25f45ee93991f52bed733ee : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71581d7efcd145a3b6697fe427ec1fe9 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62aef62e0dbc46819ca7c43aa8ded10f : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":1},{"id":"055ad00f3188246591f27205e93af15cba66b9ab","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-02 12:33:00.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"f18a2c445cfb10c95b9e379b181d0ad82d767d85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-03 00:25:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c0e3906ba298d53cf0531e8467f366ae18e9865a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-03 00:25:52.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0b88427aa00a80bd4a9ce64af51d25a725bfdd19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 01:24:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/12ac65db01914e5090312bd32748c1da : SUCCESS in 28m 21s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/ff20437a2be541a2ac45610b810ffdb8 : SUCCESS in 54m 31s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/7b9ba46fd1344adab36771096b52d235 : SUCCESS in 44m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6811a1075c9e45a7bebff63a0ff57dc5 : SUCCESS in 35m 04s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/176372eca6fc4b338018d1222b7b5ba1 : SUCCESS in 42m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5201829b53a5464a958f41eb4389ab5a : SUCCESS in 42m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/304b211ec9b44f00b620146aba39c424 : SUCCESS in 49m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a0e91daf6384c878512f8ce891aac4a : SUCCESS in 5m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da8953f4f34c4203a6b4571d812a6b62 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/691bc47994dd40bf9e3d0116386a788b : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed3a0e2afe1047458628cee989bfa738 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8d66b52019042cbb37b69f68ab73e88 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1fed9285765495186cf08202ea8fc13 : SUCCESS in 8m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/238e11d19c2948dda9511f88e3dc7c8e : SUCCESS in 2m 17s","accounts_in_message":[],"_revision_number":3},{"id":"af75b05ff6664e9adbcbfd7c314d1f34d5bd0015","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-03 01:34:40.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6f9bbff0e85661806bed033f720f2b25c99a8225","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-16 07:39:49.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"6b4f84ce17e9fe23ada859c81553a1496dd041aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-02 22:10:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3bb48d71e2a41473e62ebe7ebf7d62d8d2ca3e8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 23:33:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78af7b350a434a6e84121041610d099a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f52e1261c9334142bafb147e409d4f41 : SUCCESS in 1h 00m 14s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/aeee183e9b964352aa03d096e06fad02 : SUCCESS in 1h 08m 47s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/cba1ff7961044e8fb3db56d7a346af31 : SUCCESS in 1h 09m 29s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/c7f557fd18a24b5db7489170ab202aa8 : SUCCESS in 54m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c9728067543548688e4cb8cb623d5747 : SUCCESS in 50m 44s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6c29d65419b4457baec71b3f2983f1d1 : SUCCESS in 1h 19m 54s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/721c784668134382834428c3eb0e49f6 : SUCCESS in 1h 14m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e6db794cb714471c84a50ac009024f3a : SUCCESS in 54m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/db9810aca6b94de89b3a0e8f6e3cd068 : SUCCESS in 52m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bb3c51693a148e5a9d5289a5ca3aa30 : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4c308d0df7c49348d3dcefa60290f0b : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/185527d4c703438f937f7c53e6cf696a : SUCCESS in 7m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bc098f612f541bca09637d77d95d566 : SUCCESS in 7m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ebcaa815ac854731902e37546f933baa : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f14d5faa9ade4d18b112f089383231de : SUCCESS in 8m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/251c113f84a543a2bbcee191c3a73584 : SUCCESS in 2m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/817e28605de3446abf653cf2ef6db9f1 : SUCCESS in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"291c7bc286e1d6ea1cf4336e85aa41fc89e3305b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-09 20:44:55.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4bee0d04b81695f15df3a6c155b5858c0bb72172","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-11 05:04:16.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"fb1c6a3cdbc1040090e9b160318e666fb7d7b7de","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-11 05:04:22.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"065cc960ce3f00fbe1ac108315b9965163c17df3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 06:14:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ca0b10a2c4e40d690df06e402d8f826\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fd5a8ad93183434eade16b72bdae3b04 : SUCCESS in 43m 43s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/cb3e2dafed434b7fa2345bc067d1702c : FAILURE in 19m 05s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b5b556476e04d8cb470931274444044 : FAILURE in 15m 15s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/c308c9a8995a40a09a677645b2bece85 : SUCCESS in 47m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/17c93a3b4a59492cb59cba8631e4a437 : SUCCESS in 45m 17s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e0e55c572d6f4dddb0b3e43c35ec7e68 : SUCCESS in 1h 08m 23s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/04972b9ec38b40f69fb8d7a6295c5c6b : SUCCESS in 1h 01m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0824544898324af8a0e7c33e2d2b35f3 : SUCCESS in 49m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a8a12299e46943519cb00fb4c3c030cf : SUCCESS in 44m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/176070b1972c4a18a1644e2693577bcc : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/165b19a47a204a5482896188f0d9d4dc : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/506742465f22466697319f2cdbd31035 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbde03f9db9d4274ae4de1a874c06d86 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f7837fbc6274597b8f0c2b0a37b5266 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e57ed9f3bb64224b61adffa18f4955d : SUCCESS in 8m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dcc938c280994f87bad68b7460167b6e : SUCCESS in 2m 43s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c6d931e1104242cfa7aca5a605460654 : SUCCESS in 5m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a4431574992160633042c5329fc2186e2199bab0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 06:14:45.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"cf2815c64b69c06a80148ca0930a72e99e10f0a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 07:08:26.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9fdef00b4454e61869e3ad1ae3fef8c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e8fe63619f844d2a977aa131620999c3 : SUCCESS in 36m 13s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4eca80cd1f9747eca23d3595e25c10ef : SUCCESS in 41m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bbe59262b7924acdbe3b338735fbba15 : SUCCESS in 34m 25s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3385aaef267940b085bfdba218a80f19 : SUCCESS in 52m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/11703b1b425d458b9464d5f6c17a2246 : SUCCESS in 40m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6cbe6e5ae26747849b9163ca1c74e93c : SUCCESS in 42m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2326ec4e66e4986940ae511c6ef4ae7 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2c265e531764c7c86578c61d51612a7 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/999094b27eda4f8ab5f4d9f8deebb353 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d824b89d67f4ee799818851c676115a : SUCCESS in 6m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8471ce0206e049db8bf2ca09285f4531 : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":5},{"id":"83fed6ceff703505f6dfb976fc76d503c6a835df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 07:08:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"ebae26792093c67faac1fcc3254b748d3d92b957","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 07:09:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9e0df60c9014a4d95edffd52e957024\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8d2b7d92974432e9bb0256930612967 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"16eba621b743e0465e94cb3723f827d98dee9a6c","revisions":{"8bdd9f770cfc1fa678d63fee4558e300a29b47d8":{"kind":"REWORK","_number":1,"created":"2022-01-02 12:24:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/823230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/30/823230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/823230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/823230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/823230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/30/823230/1"}}},"commit":{"parents":[{"commit":"77f239634242315e1533d8c969fd1bcc59cc8a12","subject":"Merge \"Allow TXT record over 255 characters if split\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/77f239634242315e1533d8c969fd1bcc59cc8a12"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-02 12:24:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-02 12:24:17.000000000","tz":540},"subject":"Sync rootwrap.conf from oslo.rootwrap","message":"Sync rootwrap.conf from oslo.rootwrap\n\nThe current rootwrap.conf file is outdated and doesn\u0027t include some\nparameters. This change updates the content to make it consistent with\nthe latest example file in oslo.rootwrap.\n\nChange-Id: I5fe06cf99bb9787ec3afc67ab60a7330fe281513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8bdd9f770cfc1fa678d63fee4558e300a29b47d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8bdd9f770cfc1fa678d63fee4558e300a29b47d8"}]},"branch":"refs/heads/master"},"de7e5ddb1019275dccef20a771f47b0c741f0014":{"kind":"REWORK","_number":2,"created":"2022-01-03 00:25:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/823230/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/30/823230/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/823230/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/823230/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/823230/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/30/823230/2"}}},"commit":{"parents":[{"commit":"4807c23228fbb23af3087ee072bf73d7fc43aff5","subject":"Fix race condition in the sink when deleting records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4807c23228fbb23af3087ee072bf73d7fc43aff5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-02 12:24:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-03 00:25:32.000000000","tz":540},"subject":"Sync rootwrap.conf from oslo.rootwrap","message":"Sync rootwrap.conf from oslo.rootwrap\n\nThe current rootwrap.conf file is outdated and doesn\u0027t include some\nparameters. This change updates the content to make it consistent with\nthe latest example file in oslo.rootwrap.\n\nChange-Id: I5fe06cf99bb9787ec3afc67ab60a7330fe281513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/de7e5ddb1019275dccef20a771f47b0c741f0014"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/de7e5ddb1019275dccef20a771f47b0c741f0014"}]},"branch":"refs/heads/master"},"6c69c9be42c71a2352ff9a7082ec250974f437cc":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-01-03 00:25:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/823230/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/30/823230/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/823230/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/823230/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/823230/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/30/823230/3"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-02 12:24:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-03 00:25:52.000000000","tz":0},"subject":"Sync rootwrap.conf from oslo.rootwrap","message":"Sync rootwrap.conf from oslo.rootwrap\n\nThe current rootwrap.conf file is outdated and doesn\u0027t include some\nparameters. This change updates the content to make it consistent with\nthe latest example file in oslo.rootwrap.\n\nChange-Id: I5fe06cf99bb9787ec3afc67ab60a7330fe281513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6c69c9be42c71a2352ff9a7082ec250974f437cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6c69c9be42c71a2352ff9a7082ec250974f437cc"}]},"branch":"refs/heads/master","description":"Rebase"},"74174984b2cbd3c98c27a93d9bcaf6a73b1d7308":{"kind":"REWORK","_number":4,"created":"2022-08-02 22:10:53.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/30/823230/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/30/823230/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/823230/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/823230/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/823230/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/30/823230/4"}}},"commit":{"parents":[{"commit":"9876b7b9357765e430833b149fe7480ddb005550","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9876b7b9357765e430833b149fe7480ddb005550"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-02 12:24:17.000000000","tz":540},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-08-02 22:10:52.000000000","tz":0},"subject":"Sync rootwrap.conf from oslo.rootwrap","message":"Sync rootwrap.conf from oslo.rootwrap\n\nThe current rootwrap.conf file is outdated and doesn\u0027t include some\nparameters. This change updates the content to make it consistent with\nthe latest example file in oslo.rootwrap.\n\nChange-Id: I5fe06cf99bb9787ec3afc67ab60a7330fe281513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/74174984b2cbd3c98c27a93d9bcaf6a73b1d7308"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/74174984b2cbd3c98c27a93d9bcaf6a73b1d7308"}]},"branch":"refs/heads/master"},"16eba621b743e0465e94cb3723f827d98dee9a6c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-08-11 05:04:16.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/30/823230/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/30/823230/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/823230/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/823230/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/823230/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/30/823230/5"}}},"commit":{"parents":[{"commit":"f3c29e8bceea832e1cb787740fa676b10a410848","subject":"Merge \"Fixed issues with __repr__ and __str__ on objects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f3c29e8bceea832e1cb787740fa676b10a410848"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-02 12:24:17.000000000","tz":540},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-08-11 05:04:16.000000000","tz":0},"subject":"Sync rootwrap.conf from oslo.rootwrap","message":"Sync rootwrap.conf from oslo.rootwrap\n\nThe current rootwrap.conf file is outdated and doesn\u0027t include some\nparameters. This change updates the content to make it consistent with\nthe latest example file in oslo.rootwrap.\n\nChange-Id: I5fe06cf99bb9787ec3afc67ab60a7330fe281513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/16eba621b743e0465e94cb3723f827d98dee9a6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/16eba621b743e0465e94cb3723f827d98dee9a6c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
