)]}'
{"id":"openstack%2Fdesignate~823235","triplet_id":"openstack%2Fdesignate~master~I7b2f9ef125c8fa1c3538067759e1192dc0c032a1","project":"openstack/designate","branch":"master","topic":"rootwrap","hashtags":[],"change_id":"I7b2f9ef125c8fa1c3538067759e1192dc0c032a1","subject":"WIP: Bump oslo.rootwrap","status":"ABANDONED","created":"2022-01-02 12:32:54.000000000","updated":"2022-01-03 00:25:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b73e1eee35a01d7a9c47976cd8885299fd250cfb","_number":823235,"virtual_id_number":823235,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-02 13:41:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-02 13:41:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e4e7f6d4858ea0b1ff2755fee5360a44540d75d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-02 12:32:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c923e513394429fd385617e6857d77098310e9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-02 13:41:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/73cd98a3accd40b19b39b9b73c0b8695 : SUCCESS in 41m 31s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/962829f76fc641d58a740ee6521d0d28 : SUCCESS in 1h 02m 51s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e83a5c9ca82e4d07bf1964a8b39834cf : SUCCESS in 41m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/41ca36754e8c4569850d701be7d6d6e0 : SUCCESS in 39m 49s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0a5b15a2ec3c42b68144d7800f185678 : SUCCESS in 58m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/72e9a6d450ed4bffb8df01bbd6684331 : SUCCESS in 40m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/59ece6b65f224118b5d06122e28df1b3 : SUCCESS in 41m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7ef8f0f339a40d6adade4325b788345 : SUCCESS in 5m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9bcf1f1650274d94a61d97a872323169 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/944ff68e2d764cd49282d0214969f5de : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b586432155e3477084c1f882e32f2445 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b349985a4844c60bac0902732f8f42b : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63d971e304a94800b5c9d9b35955f3c6 : SUCCESS in 7m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bc6cda9f877a41de8a262b74bb68fe4c : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b73e1eee35a01d7a9c47976cd8885299fd250cfb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-03 00:25:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12bc3ae3f5caf5db589cbf091e897f02d0695617","revisions":{"12bc3ae3f5caf5db589cbf091e897f02d0695617":{"kind":"REWORK","_number":1,"created":"2022-01-02 12:32:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/823235/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/35/823235/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/823235/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/823235/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/823235/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/35/823235/1"}}},"commit":{"parents":[{"commit":"77f239634242315e1533d8c969fd1bcc59cc8a12","subject":"Merge \"Allow TXT record over 255 characters if split\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/77f239634242315e1533d8c969fd1bcc59cc8a12"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-02 12:32:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-02 12:32:38.000000000","tz":540},"subject":"WIP: Bump oslo.rootwrap","message":"WIP: Bump oslo.rootwrap\n\nChange-Id: I7b2f9ef125c8fa1c3538067759e1192dc0c032a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/12bc3ae3f5caf5db589cbf091e897f02d0695617"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/12bc3ae3f5caf5db589cbf091e897f02d0695617"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
