)]}'
{"id":"openstack%2Fdesignate~823469","triplet_id":"openstack%2Fdesignate~stable%2Fxena~I06cf5b5cbe5544327e30971115f3bf98214293f6","project":"openstack/designate","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-06-03 15:59:04.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"I06cf5b5cbe5544327e30971115f3bf98214293f6","subject":"Fix a typo in the tsigkey policy file","status":"MERGED","created":"2022-01-05 22:42:50.000000000","updated":"2022-06-03 17:14:15.000000000","submitted":"2022-06-03 17:13:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823469","cherry_pick_of_change":823585,"cherry_pick_of_patch_set":1,"meta_rev_id":"a10a2407b060433946b7fb5754275a3f454a10ca","_number":823469,"virtual_id_number":823469,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-03 17:12:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-06-03 17:14:15.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-03 15:59:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-03 15:59:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-05 23:45:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-25 16:06:17.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"60ab542edbb2ebf0865600a6c4396898466e2d2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-05 22:42:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4db659a31edfebfb4527b77379fa269e09e48b87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 23:45:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2721fa569d034c43a4bb40062621b2a8 : SUCCESS in 28m 56s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8149136b1dbe48de9335d4f114f43dae : SUCCESS in 38m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f35b2c0a88fb48308c3999520c34761a : SUCCESS in 41m 35s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/97e537861b5b455d9542e417e6b7f47c : SUCCESS in 1h 00m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9ac17b1dd5df4a0388732080f617920a : SUCCESS in 37m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1498f62057fe4720a6105d56d6bf1798 : SUCCESS in 36m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ab9a54ca4d6413fa186ecc04a89eb5a : SUCCESS in 5m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d2d0b16f73642b9bc9817f58f3bb11c : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7889351fcb5947e1ba7bdf1654fd5f0e : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f393d6ba4fa42369f0446c17c8e9668 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d16dd3eb21742ed8e8aeaa8332fd562 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3f381c4a7884e829f31bdf25838c6be : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83c4e209bb59400ebfd128e99437fb67 : SUCCESS in 8m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f9887883fd944949a01437c27589b815 : SUCCESS in 29m 11s","accounts_in_message":[],"_revision_number":1},{"id":"e251d89e42203114938aaa5d9fa868a1b84f94fd","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-01-25 16:06:17.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"023971350606a2383180c4d87ce925f40c3c687a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-05 00:22:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1c714b5d446184b65fec9d414be09738f624713f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 01:46:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f8da731eda8e426eae57bcb51920df29 : SUCCESS in 38m 39s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/859df2106a994600ade0dbe970c1810c : SUCCESS in 55m 34s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/509a098d290a4aaf8b51338d10d6f277 : SUCCESS in 37m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7385aeefad2d4aeaafc82e024ad83d45 : SUCCESS in 30m 56s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c3bb4b6c61ca45d18ae1a75aee908725 : SUCCESS in 1h 19m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/07a11fe906794db9ab36e2c0e25c3efe : SUCCESS in 34m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/69c0bc8dd57e4c3188e82fa618dbc0ae : SUCCESS in 35m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51949674c7f545c591bfe3aedda4139f : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/206e19b393834210b1a2f7940d8a3cd4 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ca2a4a587f240b2a5a5e111e8f69504 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/036fceb19e3448dcb2324d267a1b9039 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00467431be884b13bd0b13959a094f03 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e009935420b841ac8901f0bbe5356171 : SUCCESS in 7m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/58a3117862b64372a2a7e2322c79c5a1 : SUCCESS in 32m 24s","accounts_in_message":[],"_revision_number":2},{"id":"3fc4689079f5f229c83048cdebeee0e83ce0585e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-31 23:02:27.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"80bf24c4f4ff81cdb1cd610d9285192d48e7d13a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 00:14:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f532e69513f849b380b1ada5b4e09112 : SUCCESS in 39m 07s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/9d6bc08f78f74f3bb7664a41140b99b1 : FAILURE in 31m 08s (non-voting)\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/86ad2b1df91649f4960f98232a84ed98 : SUCCESS in 45m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c1ebc6941d1748669d963a0532ad9eb5 : SUCCESS in 1h 06m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ef4504684d6d4fe398bf98a57ea5d3d0 : SUCCESS in 37m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2419fce1de214f369d10ba6aa576dd83 : SUCCESS in 44m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d92b54d4d3a94279b87a119409615112 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58dc1882bc554f69904e203e7050c608 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec2f70c944324e179f0fffb6d879e629 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbbde80d9b28494e8dad6a0545c559c5 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27840720c91c4eec887ec8139d120c20 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c408845bbe04c9dbd6c9ce2074c565d : SUCCESS in 8m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ff262a65e6af49869116de2fa6387200 : SUCCESS in 31m 45s","accounts_in_message":[],"_revision_number":3},{"id":"c1e175807e4b084ac376be5fb0956269f86e64d9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-03 15:59:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"38f865750639a73d8733e4624ec61891622aa4a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 15:59:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1966c20aadc826a3f0a81836f13b1fe6184e8101","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 17:12:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c051512cfd2d4b238763ae9c2c075a0b : SUCCESS in 49m 15s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/d3f9aa3707e84ffa82cb59ba8422cd78 : SUCCESS in 42m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/de5b9055db7c42449db4c9d7a6975143 : SUCCESS in 40m 18s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7823ae1122594a86813880c250e03779 : SUCCESS in 1h 08m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/860bed06e43e438eb629e4ff635049fd : SUCCESS in 44m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/03f28251b9ad432380437cdcefe9efc4 : SUCCESS in 34m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/415e3385bdfa4fd1b76c11cbb169b135 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0d4ede2ca534c15a78f23824d8a0354 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f1e249fd03a49d697d0939704f64f5a : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebc2166178ec41b98a4e2d1a1b81b88a : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":3},{"id":"083987e2b63e179885c9d7b6523c3a47c6173d2b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 17:13:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"df7b15df5128912beca7fceb9826d4e8d141765e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 17:14:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfd904999bf1459caff69d6229fadcc4 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f4396df723a701a5b38027ceb2375fde0837b89b","revisions":{"ba6a7e552325e4545e26cd8bbf172df59b42b074":{"kind":"REWORK","_number":1,"created":"2022-01-05 22:42:50.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/69/823469/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/69/823469/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/823469/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/823469/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/823469/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/69/823469/1"}}},"commit":{"parents":[{"commit":"ad87c95e2c9a92149f62cdcbee4326dca457e2aa","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ad87c95e2c9a92149f62cdcbee4326dca457e2aa"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-01-05 22:40:40.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-01-05 22:42:50.000000000","tz":0},"subject":"Fix a typo in the tsigkey policy file","message":"Fix a typo in the tsigkey policy file\n\nThis patch fixes a typo in the deprecated reason string variable\nname.\n\nChange-Id: I06cf5b5cbe5544327e30971115f3bf98214293f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ba6a7e552325e4545e26cd8bbf172df59b42b074"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ba6a7e552325e4545e26cd8bbf172df59b42b074"}]},"branch":"refs/heads/stable/xena"},"a20db46396500d38d6222a135638e3e66f5e3551":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-05 00:22:10.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/69/823469/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/69/823469/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/823469/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/823469/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/823469/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/69/823469/2"}}},"commit":{"parents":[{"commit":"fa1bce20e6bbab43c07e69901283811de5b3adf2","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa1bce20e6bbab43c07e69901283811de5b3adf2"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-01-05 22:40:40.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-05 00:22:10.000000000","tz":0},"subject":"Fix a typo in the tsigkey policy file","message":"Fix a typo in the tsigkey policy file\n\nThis patch fixes a typo in the deprecated reason string variable\nname.\n\nChange-Id: I06cf5b5cbe5544327e30971115f3bf98214293f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a20db46396500d38d6222a135638e3e66f5e3551"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a20db46396500d38d6222a135638e3e66f5e3551"}]},"branch":"refs/heads/stable/xena","description":"Rebase"},"f4396df723a701a5b38027ceb2375fde0837b89b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-05-31 23:02:27.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/69/823469/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/69/823469/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/823469/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/823469/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/823469/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/69/823469/3"}}},"commit":{"parents":[{"commit":"c78db47c746cab2fc505bd60f11f05a08eb94120","subject":"Update zuul queue configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c78db47c746cab2fc505bd60f11f05a08eb94120"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-01-05 22:40:40.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-31 23:02:26.000000000","tz":0},"subject":"Fix a typo in the tsigkey policy file","message":"Fix a typo in the tsigkey policy file\n\nThis patch fixes a typo in the deprecated reason string variable\nname.\n\nChange-Id: I06cf5b5cbe5544327e30971115f3bf98214293f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f4396df723a701a5b38027ceb2375fde0837b89b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f4396df723a701a5b38027ceb2375fde0837b89b"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
