)]}'
{"id":"openstack%2Fdesignate~823585","triplet_id":"openstack%2Fdesignate~master~I06cf5b5cbe5544327e30971115f3bf98214293f6","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-02-18 02:49:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I06cf5b5cbe5544327e30971115f3bf98214293f6","subject":"Fix a typo in the tsigkey policy file","status":"MERGED","created":"2022-01-05 22:41:50.000000000","updated":"2022-02-18 02:53:02.000000000","submitted":"2022-02-18 02:49:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823585","meta_rev_id":"82d45e5b328578ed6569946a7e7007e200582b83","_number":823585,"virtual_id_number":823585,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-18 02:49:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-01-25 16:03:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-06 01:18:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-17 21:30:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-05 23:38:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-06 01:18:19.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-01-25 16:03:50.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"5a2fb66c028ced6eacc24f4f3d6c8b8964aa11fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-05 22:41:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0be2786229c2b5fc4b7523545b6b992607e14485","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 23:38:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c010fd6391dd4c208a7c77a34d95d220 : SUCCESS in 38m 24s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/76132d7147f449f7a794a80e21be6256 : SUCCESS in 37m 48s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/879bd3773ec5462d8ba243df6ebfb4f8 : SUCCESS in 48m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f99cd02da51b4f1fb7141cf942aa6aa6 : SUCCESS in 36m 30s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/48e9840c7dab488abdc382e41b575a1f : SUCCESS in 54m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/59a7afc06a0d48c594da539bbafd8a6c : SUCCESS in 40m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/24a708bcd263434998dea84554e87f8a : SUCCESS in 44m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85d38dad01974e53877ba13d41545699 : SUCCESS in 5m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c2a54a66f474d8bb1deb3d175d7bab1 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1666870de134036b58f1b703cb384a9 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0dab295a01b4f3fb600f18198ee7d9c : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0944967456c448d813a5cab2804167f : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd5f791e46574fb18a0e0c980be22fc8 : SUCCESS in 7m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/25247a8da1ee45438a4139789c6fb0cc : SUCCESS in 29m 47s","accounts_in_message":[],"_revision_number":1},{"id":"de3d6202d8211f316c770a9a86f68425a9765710","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-06 01:18:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1a81364f0d3ada41217f45c3ab1b694dfa3aacb1","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-01-25 16:03:50.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea18768954097bab866c11e12b7227282238544f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-02-17 21:30:29.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6462ef4fe2bd45043c0f321d9cf3fa680daf2c00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 21:32:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ae2dbaf195925228bf294d4f25a6208677c1b1ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 02:49:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a975b0b2bdfc46189ae3d77c04b110f5 : SUCCESS in 27m 47s\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/32006cc6e1634134abaf6b81ba685e2c : SUCCESS in 46m 58s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/d567f81833e04dbaa7e013193a4d87c4 : SUCCESS in 48m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4e48c1010b034f4780833e2ddd9cea90 : SUCCESS in 47m 59s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/468ef9dc4920473e954fe01fd9c69ca1 : SUCCESS in 1h 10m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8126be5d1d2f4bb8863d3b0d2bd75ec0 : SUCCESS in 49m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/48c01b019f9743f8be04975fb822c6bf : SUCCESS in 47m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cba899c9d294481b01aa7eff0c06b96 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66daceb0a27f4776aa0d4d5776f81eef : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44652d1c440a4c2abc4511876475205a : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/406c93798fde4eaba0449d882c7ef449 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fe571891d4948d99e5311f66ce3379e : SUCCESS in 8m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8916d2b8df274b9ead1a9f7664e151a5 : SUCCESS in 40m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f1461599f3df07bc0cc2078e7b1b4dd1091cd7a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 02:49:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"82d45e5b328578ed6569946a7e7007e200582b83","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 02:53:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f09a9e689e21410696bb043dea540692 : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"395a4a2911b2a66bfce71699f0aa46b802ae3dbc","revisions":{"395a4a2911b2a66bfce71699f0aa46b802ae3dbc":{"kind":"REWORK","_number":1,"created":"2022-01-05 22:41:50.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/85/823585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/85/823585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/85/823585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/85/823585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/85/823585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/85/823585/1"}}},"commit":{"parents":[{"commit":"97339d78e245e053fa8c792ed3d3649f369623ca","subject":"Merge \"Updating python testing classifier as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/97339d78e245e053fa8c792ed3d3649f369623ca"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-01-05 22:40:40.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-01-05 22:41:48.000000000","tz":0},"subject":"Fix a typo in the tsigkey policy file","message":"Fix a typo in the tsigkey policy file\n\nThis patch fixes a typo in the deprecated reason string variable\nname.\n\nChange-Id: I06cf5b5cbe5544327e30971115f3bf98214293f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/395a4a2911b2a66bfce71699f0aa46b802ae3dbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/395a4a2911b2a66bfce71699f0aa46b802ae3dbc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
