)]}'
{"id":"openstack%2Fdesignate~823623","triplet_id":"openstack%2Fdesignate~master~Ia1194ab838c52d5d91cb1d26c4556c73b4f3a745","project":"openstack/designate","branch":"master","topic":"clean_rpc","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-01-29 00:08:54.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-01-28 23:19:59.000000000","reason":"Michael Johnson replied on the change"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-01-29 00:08:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia1194ab838c52d5d91cb1d26c4556c73b4f3a745","subject":"Fix duplicate zone when creating ptr records","status":"MERGED","created":"2022-01-06 05:37:34.000000000","updated":"2022-01-29 00:10:22.000000000","submitted":"2022-01-29 00:08:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"823623-clean_rpc","meta_rev_id":"e2a6a90fa6c1553436c709585d662ee5d0390e2a","_number":823623,"virtual_id_number":823623,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-29 00:08:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-28 23:19:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-28 23:19:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-06 06:29:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-21 03:17:17.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-01-21 03:17:17.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"06416a95b8b9cc4d37d5438169b09079c3c80ebe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-06 05:37:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ef5283a7e039a6fe9139aba7b4fae7fc84ea27a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 06:29:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e4dd4fa1757d4124a7d1078db1ab40cd : SUCCESS in 32m 41s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/47e184aff94d4f77a47ed32f2b13f1a9 : SUCCESS in 50m 50s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8e0b4e6b87d84b32a5200b9dd0a919d0 : SUCCESS in 33m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0ef12e839244465fb08850ed2a3f83b0 : SUCCESS in 33m 55s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/eb2fdf37b718444b908c71fe460e78ee : SUCCESS in 48m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ad97a8069d094e73aabd000bbcc718e7 : SUCCESS in 39m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9e23ec2094334f2eafa36e8feb97008e : SUCCESS in 33m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e69a5b382974f789d2601bfca7843b7 : SUCCESS in 6m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4295b565d3b5492eaae0302183f3ec55 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df9dd58fea9d439582473128ba028ca5 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a55848c2015b4fa5bd4989cbf19fd41c : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4598097799a4e199f098db66cf88bab : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c70b85d22e174f558fe4dcd8a12e41db : SUCCESS in 7m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/12fddf5e7fe44ef4bd79d49a1459fd7c : SUCCESS in 28m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5a418ac321b936a7d3705f4ca225073d07a484ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-08 02:11:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"eca3fef3ac6da2ecbbea4258b1d1123eb7b443b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-08 03:31:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7988ad92587040368a0fe8de2faada1a : SUCCESS in 40m 40s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/6e2b048757e94b6ba3e31736e2e3fce0 : SUCCESS in 49m 10s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f8ffcb2c070544b48d70a23a88f05008 : SUCCESS in 44m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7af443a69dcb476e823e44c0f710dddb : SUCCESS in 38m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ae5502d18f8740889df98a589cf61a69 : SUCCESS in 57m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dc5d862a811d4e1794b2f3ebf5b4e57c : SUCCESS in 39m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8a32b984272348e1b49a1b883d59bd32 : SUCCESS in 44m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/280c2d2fd5a346d68227263209774d68 : SUCCESS in 6m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc3d598dbc5f473fb1bb55eeb6394e88 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddda253f058a4b84a78b39eb60e963a8 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8010559c3ec348dca8731fc666b10e5b : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffcfcd372b4849fcb484601df01f2cdf : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cb556eed4314519b97c802ff17641f1 : SUCCESS in 8m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6b7fabe5d6a54984952b16b4740335d4 : SUCCESS in 32m 05s","accounts_in_message":[],"_revision_number":2},{"id":"22612761a78b52ffab77d4b6c3023a8c6329e39c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-09 06:10:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cab1f0319f44858245ed86dd4949af5f3b799f49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-09 07:16:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d56daab6eb014a88be5efce1bfea4c50 : SUCCESS in 38m 43s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/fdd2f346affb4ae1898016568194dc80 : SUCCESS in 34m 40s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/6260d5a7902942a0a6b28ceaff7fd6d0 : SUCCESS in 36m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e82fed7c41504bf189fd988691ca7600 : SUCCESS in 40m 11s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/85517fd95837438dada4c401dd6310ad : SUCCESS in 1h 01m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b9e72562a2014d47a841304ce4b80c6b : SUCCESS in 32m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1ef00d5abfc54cd1a3b067e3e4a94ccf : SUCCESS in 39m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6900c837626a4656b32222610fecae42 : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc6dd085ccee4ed481c4e638520bf1f0 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dec7005c5ab9488fb302839c44f2fb9a : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9872fe8196bf46bfb8cef1970e0b274d : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd636bbf28084b738a65b4b99d2be417 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd6b2355fa1546788176fda9f5ec1646 : SUCCESS in 7m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f8c928e6c5bb49f1ab585d41a0886109 : SUCCESS in 34m 16s","accounts_in_message":[],"_revision_number":3},{"id":"16a361c302dd15b645c064a0b463d0bf51f9ac34","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-09 08:21:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"16656d2b94678e03ef9439cd8e7518c82320f5a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-09 09:16:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/87865ef4ff654ec3824621ee460eea18 : SUCCESS in 34m 29s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/a0a47c145d9748a89a72036dd35a9cee : SUCCESS in 53m 38s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/420775fb1d37432a90e0751cc8e4bd20 : SUCCESS in 35m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2b866955a5a24cedb066a7cfb1624b99 : SUCCESS in 39m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6b902756a9fb40f9a76ecdd6abc1e920 : SUCCESS in 53m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/69ea958b3eaa4a96a3f7afdf8f85f0de : SUCCESS in 37m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b84a4df78625487185d485d7ea9567c1 : SUCCESS in 39m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3ee435be1ee481db0356d3add83bf33 : SUCCESS in 5m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48dea7d57b5b473a805fd3278b0cf624 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46cebe2153474bce8371d21013fc655e : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85e579c4430b42c7b679c13dfd6eb6a1 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23ea126ff03d4d68a24ff540e9f9b7d0 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b0ac869448c478da1f4f3b43548fd18 : SUCCESS in 8m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/32792ef55cf740a0bcae718e6a2e79f8 : SUCCESS in 30m 54s","accounts_in_message":[],"_revision_number":3},{"id":"92613a34221f0445a0fd32e7a41dd7a05968ccfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-10 08:44:02.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"475338adf81f2990bc9ec937d06585ef348f67fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 09:52:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9fd53ddc487943dfa0bd416ef649e829 : SUCCESS in 44m 52s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/a23dee5d352d40989939fc3e711c2bee : SUCCESS in 55m 33s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/9db8d290d6184132a14f88b743193fb8 : SUCCESS in 37m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6aaf58d93db34ffeaa267a3aa873188c : SUCCESS in 37m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/21078b78a1c143db8fc057a2f95c4ef4 : SUCCESS in 1h 02m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c4f1e38b30fd4985b35efbcf046c5eaa : SUCCESS in 41m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b68d9bd8c7e64e6ba43f07213245c2b6 : SUCCESS in 38m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56cf7e7d7b864c5584b1587ff08f6a78 : SUCCESS in 5m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db5487dfce5442ecab95ad2ed9432569 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1e3b681b26b4326907efa6d4f8d1382 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2093632cfa3d484690cd405f3d9c93e0 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1e04cc6747b47fa867a2d6d2155fb19 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09df3bd64a35425699415013e8d8d96d : SUCCESS in 7m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c9d89a956cac4be9a2ba343a4e575a2e : SUCCESS in 30m 14s","accounts_in_message":[],"_revision_number":4},{"id":"0f8e845d19333799812df5002808f2168a64da50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-10 21:00:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f732b85e69b84ebf84132069b0fe5c4f854a546e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 21:57:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/adc4199166d44ba4850f0bf4b898f45a : SUCCESS in 37m 56s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/92bc04ad0c7f478ca1e4a71424d9dc5f : SUCCESS in 44m 03s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1692fc6b14a6433994d2b8c17e4944cc : SUCCESS in 38m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/56b7da0ff88e44819c6b8ecb42ffb8d8 : SUCCESS in 43m 04s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/00e0fe273ede4330b1964f23718fd1c8 : SUCCESS in 55m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cc74a46a592641d39bce805327820c26 : SUCCESS in 37m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b8a0f6d580e8407f84ab85fc53d9ce78 : SUCCESS in 38m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/770376b1625f4169816d9e2790fe0fe8 : SUCCESS in 6m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11cb99bcc5d7415f884f4c326ebe1f71 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5e4fdc5d557464fabdd332484ff6628 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/283b7c8ff9954d578374fd87893d18a3 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d27154fc737440696c6cd4c9dbb9b2c : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8879341f0d03448290ee35f4249fdad2 : SUCCESS in 8m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2e4cc435c4204e6caead89187d5d97cf : SUCCESS in 31m 03s","accounts_in_message":[],"_revision_number":5},{"id":"88e6486396c790c7fb5e2a6f99854fe84b9b587b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-21 03:17:17.000000000","message":"Patch Set 5: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"94e1a3a9db268f5d61e5b915406f89cfdef58308","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-01-25 07:51:06.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"c758b27f77e0f18c2f14152d79042152b787e4fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-25 21:03:06.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"11955e03dd67d6756f89185a74a830ec32183c6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 22:08:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/58bb52c4a77e4fd49ea3ab51aeb23cb4 : SUCCESS in 39m 31s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/baf1614aafe149dc8b30c155fda0e8ac : SUCCESS in 1h 00m 15s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f5ef660103041ca9aa36d25d8a08457 : SUCCESS in 45m 50s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1244dea16f4144d9a474415bfe2d8751 : SUCCESS in 41m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d18a22a2768f40cfb32a97203d63a6dc : SUCCESS in 43m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d05b70d16413472098c4bf5cf4381d47 : SUCCESS in 59m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/85a54376834b434495efcd7a333cb038 : SUCCESS in 36m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/989a92c8247348e2b9479c0a654386e7 : SUCCESS in 42m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01ef12b5ed764373a0d48f1c233902f3 : SUCCESS in 8m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/717897710a7741ccb86662f8cc3e1fee : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b2af10f20d2443690552c590a7202ce : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19d72a57014e4a4f873ea32739f39205 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19111160cf7b4ef9bea2b0e6f424c5dc : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77e299753cd24a42b6be11610d194b18 : SUCCESS in 8m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/017ad9bd2b794437b1a09256e5189190 : SUCCESS in 39m 17s","accounts_in_message":[],"_revision_number":6},{"id":"7a1571caf854b68d5c478ed3937ae71e5f55451f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-26 00:34:41.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"d8a9c8bb4af903f1d568f35ce216e893f129afaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-26 04:53:31.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"03d4773c46601ae59210a93afc97de99c3acbb2e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-26 04:56:20.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"44b63b347e65a377c00f6b0df908b1c6a8da2cd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-26 04:58:30.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5cba737ed3762b60bdd7aaff65b3a2232c8889ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-26 05:00:27.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2e23d1c27317b0f6712f5a3827d0c18ef73edb21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-26 05:36:05.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"8ce199198393553fa862893fe8a7a11c868344b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-26 05:43:00.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"f56d519f7daea02e206808c543ddc18a6960a877","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-26 05:58:39.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"0941dd4f53b0cb6ca61e002325ca78482ad76620","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 07:01:13.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f3ac4e041bd34de6b2a9847da6d383a1 : SUCCESS in 46m 30s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/db58fd28ad88435b8f6f7bf35e5c529c : SUCCESS in 53m 10s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b9f0d9a224a447e9540de81f73a34d3 : SUCCESS in 42m 00s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8115d4a4e60141069077fd9c2b8503db : SUCCESS in 38m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/462afb8d1d69494b8f673ab4f2c90308 : SUCCESS in 38m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/089771aa6c334c5cbb8a20bb70129916 : SUCCESS in 58m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/da56a523fa8940b7b409493d0490f2ea : SUCCESS in 41m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/84d89d777fa6472c9b608d27dc68c474 : SUCCESS in 45m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9e65b36e5804879a4f734d00b2e03a4 : SUCCESS in 7m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/644773cb677447bb8ce75eeb44adf466 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29fa863b83224de28214c98bcd95c932 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae51be17025d4fc1a08611c15c02200e : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/922334b09d844bd5b8602610cf5d199f : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdbd0820e52e41dcbab77865026446ac : SUCCESS in 8m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b6f8d843e49f48ceacf8212f8a816398 : SUCCESS in 38m 37s","accounts_in_message":[],"_revision_number":12},{"id":"7867c1c2c8cf784a42bd64b7cc5a741aa70a9c29","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-01-27 09:20:02.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"823b7a6e59b25c9f26d014a7a95cb24bb98a17b9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-27 17:48:26.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"465fc4c160b17643ae450725a4df649b8c7c1f48","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-28 23:19:59.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"ef736224ff96414c0d6ff417f5424e3b412d3177","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 23:20:46.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"c4482d6722403db6e3de6fa0646abaac496959b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-29 00:08:53.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bc0d609ed0ae4760baaa15d805dd42c3 : SUCCESS in 30m 45s\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8afd24bb45004cd3b0de4d0df9f0ebb1 : SUCCESS in 35m 58s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e36f92d6fe4044369271cf853e63d13b : SUCCESS in 32m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8974d4da97dc4030af3c1ff1284249a1 : SUCCESS in 37m 54s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/74ce22c9d2cc4bdc920340b5d9a39ab1 : SUCCESS in 44m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fea4a9481f8c4ddcb77b98e571c8eca6 : SUCCESS in 31m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2c2290cca25c431e9fdaaa9cf6e7cb62 : SUCCESS in 32m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cef0f486c9184a49aeb293b3c9f5ebc9 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/354822f3a1064391ba1aade9dbddc413 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f44a4799abe340fb892602b78026520f : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90d004bd24b24e7f876c58d9028845f5 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5954cbbe0aa45ec95a1e462b588d7af : SUCCESS in 7m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fb2026bd901b482f8f888bcf2814e3bd : SUCCESS in 26m 50s","accounts_in_message":[],"_revision_number":12},{"id":"220de21dfcb35633ccbc4044f788936dc266ccd1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-29 00:08:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"e2a6a90fa6c1553436c709585d662ee5d0390e2a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-29 00:10:22.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cea3de870974b1bb024f72a9972ab7a : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"0c7d218ba103e8260322e40f76a49a8c92556bfe","revisions":{"e08c1dae19d9aab6b582be95879fa60233751b58":{"kind":"REWORK","_number":1,"created":"2022-01-06 05:37:34.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/23/823623/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/823623/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/823623/1"}}},"commit":{"parents":[{"commit":"97339d78e245e053fa8c792ed3d3649f369623ca","subject":"Merge \"Updating python testing classifier as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/97339d78e245e053fa8c792ed3d3649f369623ca"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:35:14.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:35:14.000000000","tz":-480},"subject":"Fix duplicate zone when creating ptr records","message":"Fix duplicate zone when creating ptr records\n\nThis fixes a race-condition when creating multiple\nPTR records under the same zone. There is a brief\nwindow when creating two identical zones can cause\nan error. This adds a fallback that should be enough\nto reduce the chances of a race-condition to almost zero.\n\nChange-Id: Ia1194ab838c52d5d91cb1d26c4556c73b4f3a745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e08c1dae19d9aab6b582be95879fa60233751b58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e08c1dae19d9aab6b582be95879fa60233751b58"}]},"branch":"refs/heads/master"},"1fd0e33569346503b83351fd34ee6ddb4f7a6e9b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-01-08 02:11:52.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/23/823623/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/823623/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/823623/2"}}},"commit":{"parents":[{"commit":"bff3d5f6e31fe595a77143ec4ac779c187bf72a8","subject":"Fix designate-manage pool update bugs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bff3d5f6e31fe595a77143ec4ac779c187bf72a8"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:35:14.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-08 02:11:52.000000000","tz":0},"subject":"Fix duplicate zone when creating ptr records","message":"Fix duplicate zone when creating ptr records\n\nThis fixes a race-condition when creating multiple\nPTR records under the same zone. There is a brief\nwindow when creating two identical zones can cause\nan error. This adds a fallback that should be enough\nto reduce the chances of a race-condition to almost zero.\n\nChange-Id: Ia1194ab838c52d5d91cb1d26c4556c73b4f3a745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fd0e33569346503b83351fd34ee6ddb4f7a6e9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fd0e33569346503b83351fd34ee6ddb4f7a6e9b"}]},"branch":"refs/heads/master","description":"Rebase"},"6a17d05b82f3e4c3120166310e2c32027da67a12":{"kind":"REWORK","_number":3,"created":"2022-01-09 06:10:02.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/23/823623/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/823623/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/823623/3"}}},"commit":{"parents":[{"commit":"bff3d5f6e31fe595a77143ec4ac779c187bf72a8","subject":"Fix designate-manage pool update bugs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bff3d5f6e31fe595a77143ec4ac779c187bf72a8"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:35:14.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-09 06:08:34.000000000","tz":-480},"subject":"Fix duplicate zone when creating ptr records","message":"Fix duplicate zone when creating ptr records\n\nThis fixes a race-condition when creating multiple\nPTR records under the same zone. There is a brief\nwindow when creating two identical zones can cause\nan error. This adds a fallback that should be enough\nto reduce the chances of a race-condition to almost zero.\n\nI also added a threaded test that caught multiple other\nbugs in this code.\n- find_recordset would throw an unwanted exception.\n- A transaction would interrupt most error handling.\n\nChange-Id: Ia1194ab838c52d5d91cb1d26c4556c73b4f3a745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6a17d05b82f3e4c3120166310e2c32027da67a12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6a17d05b82f3e4c3120166310e2c32027da67a12"}]},"branch":"refs/heads/master"},"62e3cd5383354177b52fab6f8213fe9aadaaedb3":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-01-10 08:44:02.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/23/823623/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/823623/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/823623/4"}}},"commit":{"parents":[{"commit":"bff3d5f6e31fe595a77143ec4ac779c187bf72a8","subject":"Fix designate-manage pool update bugs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bff3d5f6e31fe595a77143ec4ac779c187bf72a8"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:35:14.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-10 08:44:02.000000000","tz":0},"subject":"Fix duplicate zone when creating ptr records","message":"Fix duplicate zone when creating ptr records\n\nThis fixes a race-condition when creating multiple PTR records\nunder the same zone. There is a brief window when creating two\nidentical zones can cause an error. This adds a fallback that\nshould prevent the error.\n\nI also added a threaded test that caught multiple additional\nbugs in this code.\n- Wrong find_recordset used caused the wrong exception to be thrown.\n- The transaction workflow would break error handling.\n\nChange-Id: Ia1194ab838c52d5d91cb1d26c4556c73b4f3a745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/62e3cd5383354177b52fab6f8213fe9aadaaedb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/62e3cd5383354177b52fab6f8213fe9aadaaedb3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"75dfe6707aaf0360736a89dd566d3104e81a4a7d":{"kind":"REWORK","_number":5,"created":"2022-01-10 21:00:08.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/23/823623/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/823623/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/823623/5"}}},"commit":{"parents":[{"commit":"bff3d5f6e31fe595a77143ec4ac779c187bf72a8","subject":"Fix designate-manage pool update bugs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bff3d5f6e31fe595a77143ec4ac779c187bf72a8"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:35:14.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-10 20:59:58.000000000","tz":-480},"subject":"Fix duplicate zone when creating ptr records","message":"Fix duplicate zone when creating ptr records\n\nThis fixes a race-condition when creating multiple PTR records\nunder the same zone. There is a brief window when creating two\nidentical zones can cause an error. This adds a fallback that\nshould prevent the error.\n\nI also added a threaded test that caught multiple additional\nbugs in this code.\n- Wrong find_recordset used caused the wrong exception to be thrown.\n- The transaction workflow would break error handling.\n\nChange-Id: Ia1194ab838c52d5d91cb1d26c4556c73b4f3a745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75dfe6707aaf0360736a89dd566d3104e81a4a7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75dfe6707aaf0360736a89dd566d3104e81a4a7d"}]},"branch":"refs/heads/master"},"c9a7f3392c87dedf5603e59b6821e74d297efe13":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-01-25 21:03:06.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/23/823623/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/823623/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/823623/6"}}},"commit":{"parents":[{"commit":"93bab684f41660195591fb3652d56be88e91975e","subject":"Merge \"Add a BIND9 job on CentOS 9 Stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/93bab684f41660195591fb3652d56be88e91975e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:35:14.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-25 21:03:06.000000000","tz":0},"subject":"Fix duplicate zone when creating ptr records","message":"Fix duplicate zone when creating ptr records\n\nThis fixes a race-condition when creating multiple PTR records\nunder the same zone. There is a brief window when creating two\nidentical zones can cause an error. This adds a fallback that\nshould prevent the error.\n\nI also added a threaded test that caught multiple additional\nbugs in this code.\n- Wrong find_recordset used caused the wrong exception to be thrown.\n- The transaction workflow would break error handling.\n\nChange-Id: Ia1194ab838c52d5d91cb1d26c4556c73b4f3a745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c9a7f3392c87dedf5603e59b6821e74d297efe13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c9a7f3392c87dedf5603e59b6821e74d297efe13"}]},"branch":"refs/heads/master","description":"Rebase"},"793aa3d1c7ecaabe4e6e80deb15c4e0711f26d15":{"kind":"REWORK","_number":7,"created":"2022-01-26 04:53:31.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/23/823623/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/823623/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/823623/7"}}},"commit":{"parents":[{"commit":"93bab684f41660195591fb3652d56be88e91975e","subject":"Merge \"Add a BIND9 job on CentOS 9 Stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/93bab684f41660195591fb3652d56be88e91975e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:35:14.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-26 04:53:13.000000000","tz":-480},"subject":"Fix duplicate zone when creating ptr records","message":"Fix duplicate zone when creating ptr records\n\nThis fixes a race-condition when creating multiple PTR records\nunder the same zone. There is a brief window when creating two\nidentical zones can cause an error. This adds a fallback that\nshould prevent the error.\n\nI also added a threaded test that caught multiple additional\nbugs in this code.\n- Wrong find_recordset used caused the wrong exception to be thrown.\n- Fixed transaction workflow breaking error handling.\n\nChange-Id: Ia1194ab838c52d5d91cb1d26c4556c73b4f3a745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/793aa3d1c7ecaabe4e6e80deb15c4e0711f26d15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/793aa3d1c7ecaabe4e6e80deb15c4e0711f26d15"}]},"branch":"refs/heads/master"},"15ffa11173ec40a6d7bcfff11c8544c239ef7b0e":{"kind":"REWORK","_number":8,"created":"2022-01-26 04:58:30.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/23/823623/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/823623/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/823623/8"}}},"commit":{"parents":[{"commit":"93bab684f41660195591fb3652d56be88e91975e","subject":"Merge \"Add a BIND9 job on CentOS 9 Stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/93bab684f41660195591fb3652d56be88e91975e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:35:14.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-26 04:58:23.000000000","tz":-480},"subject":"Fix duplicate zone when creating ptr records","message":"Fix duplicate zone when creating ptr records\n\nThis fixes a race-condition when creating multiple PTR records\nunder the same zone. There is a brief window when creating two\nidentical zones can cause an error. This adds a fallback that\nshould prevent the error.\n\nI also added a threaded test that caught multiple additional\nbugs in this code.\n- Wrong find_recordset used caused the wrong exception to be thrown.\n- Fixed transaction workflow breaking error handling.\n\nChange-Id: Ia1194ab838c52d5d91cb1d26c4556c73b4f3a745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15ffa11173ec40a6d7bcfff11c8544c239ef7b0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15ffa11173ec40a6d7bcfff11c8544c239ef7b0e"}]},"branch":"refs/heads/master"},"ad1d923fbcbb02a7c10018bbe17269868553a568":{"kind":"REWORK","_number":9,"created":"2022-01-26 05:00:27.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/23/823623/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/823623/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/823623/9"}}},"commit":{"parents":[{"commit":"93bab684f41660195591fb3652d56be88e91975e","subject":"Merge \"Add a BIND9 job on CentOS 9 Stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/93bab684f41660195591fb3652d56be88e91975e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:35:14.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-26 05:00:19.000000000","tz":-480},"subject":"Fix duplicate zone when creating ptr records","message":"Fix duplicate zone when creating ptr records\n\nThis fixes a race-condition when creating multiple PTR records\nunder the same zone. There is a brief window when creating two\nidentical zones can cause an error. This adds a fallback that\nshould prevent the error.\n\nI also added a threaded test that caught multiple additional\nbugs in this code.\n- Wrong find_recordset used caused the wrong exception to be thrown.\n- Fixed transaction workflow breaking error handling.\n\nChange-Id: Ia1194ab838c52d5d91cb1d26c4556c73b4f3a745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ad1d923fbcbb02a7c10018bbe17269868553a568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ad1d923fbcbb02a7c10018bbe17269868553a568"}]},"branch":"refs/heads/master"},"8199f76bab8751f37dcc9f8bfaa86e219ed4d476":{"kind":"REWORK","_number":10,"created":"2022-01-26 05:36:05.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/23/823623/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/823623/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/823623/10"}}},"commit":{"parents":[{"commit":"4d29c539756d071b2398680ce1a061e48940ec1b","subject":"Updated status logic to always NOTIFY on change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4d29c539756d071b2398680ce1a061e48940ec1b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:35:14.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-26 05:26:36.000000000","tz":-480},"subject":"Fix duplicate zone when creating ptr records","message":"Fix duplicate zone when creating ptr records\n\nThis fixes a race-condition when creating multiple PTR records\nunder the same zone. There is a brief window when creating two\nidentical zones can cause an error. This adds a fallback that\nshould prevent the error.\n\nI also added a threaded test that caught multiple additional\nbugs in this code.\n- Wrong find_recordset used caused the wrong exception to be thrown.\n- The transaction workflow would break error handling.\n\nChange-Id: Ia1194ab838c52d5d91cb1d26c4556c73b4f3a745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8199f76bab8751f37dcc9f8bfaa86e219ed4d476"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8199f76bab8751f37dcc9f8bfaa86e219ed4d476"}]},"branch":"refs/heads/master"},"d1138165fb43441cc1bddfe7adfcadbdd6d5bcbf":{"kind":"REWORK","_number":11,"created":"2022-01-26 05:43:00.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/23/823623/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/823623/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/823623/11"}}},"commit":{"parents":[{"commit":"4d29c539756d071b2398680ce1a061e48940ec1b","subject":"Updated status logic to always NOTIFY on change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4d29c539756d071b2398680ce1a061e48940ec1b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:35:14.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-26 05:41:49.000000000","tz":-480},"subject":"Fix duplicate zone when creating ptr records","message":"Fix duplicate zone when creating ptr records\n\nThis fixes a race-condition when creating multiple PTR records\nunder the same zone. There is a brief window when creating two\nidentical zones can cause an error. This adds a fallback that\nshould prevent the error.\n\nI also added a threaded test that caught multiple additional\nbugs in this code.\n- Wrong find_recordset used caused the wrong exception to be thrown.\n- The transaction workflow would break error handling.\n\nChange-Id: Ia1194ab838c52d5d91cb1d26c4556c73b4f3a745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d1138165fb43441cc1bddfe7adfcadbdd6d5bcbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d1138165fb43441cc1bddfe7adfcadbdd6d5bcbf"}]},"branch":"refs/heads/master"},"0c7d218ba103e8260322e40f76a49a8c92556bfe":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2022-01-26 05:58:39.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/23/823623/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/823623/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/823623/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/823623/12"}}},"commit":{"parents":[{"commit":"93bab684f41660195591fb3652d56be88e91975e","subject":"Merge \"Add a BIND9 job on CentOS 9 Stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/93bab684f41660195591fb3652d56be88e91975e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:35:14.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-26 05:58:39.000000000","tz":0},"subject":"Fix duplicate zone when creating ptr records","message":"Fix duplicate zone when creating ptr records\n\nThis fixes a race-condition when creating multiple PTR records\nunder the same zone. There is a brief window when creating two\nidentical zones can cause an error. This adds a fallback that\nshould prevent the error.\n\nI also added a threaded test that caught multiple additional\nbugs in this code.\n- Wrong find_recordset used caused the wrong exception to be thrown.\n- The transaction workflow would break error handling.\n\nChange-Id: Ia1194ab838c52d5d91cb1d26c4556c73b4f3a745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0c7d218ba103e8260322e40f76a49a8c92556bfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0c7d218ba103e8260322e40f76a49a8c92556bfe"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
