)]}'
{"id":"openstack%2Fdesignate~823624","triplet_id":"openstack%2Fdesignate~master~I5d566588be66e9ed0df9484e36504a69b4f4b5a9","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-02-06 08:02:57.000000000","reason":"Dr. Jens Harbott (frickler) replied on the change"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-02-06 10:07:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5d566588be66e9ed0df9484e36504a69b4f4b5a9","subject":"Updated status logic to always NOTIFY on change","status":"MERGED","created":"2022-01-06 05:50:23.000000000","updated":"2022-02-06 10:08:43.000000000","submitted":"2022-02-06 10:07:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823624","meta_rev_id":"fea47f9c64632a59f85c0a5b067e28fc6f945ab7","_number":823624,"virtual_id_number":823624,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-06 10:07:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-02-06 08:02:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-02-06 08:02:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-06 06:42:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-25 07:47:37.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2022-02-04 23:52:25.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-02-06 08:02:57.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"f4c5d39711997dff044130a9c4a83cb90b66e7b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-06 05:50:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"893332095995e923763a2e687f0a9dc97ac5c091","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 06:42:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/eb4d947042f54598bc133c072e58cbff : SUCCESS in 46m 36s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/1a757de399be4aaa9d337e91826761a6 : SUCCESS in 44m 52s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/7f83115ac27b4976963c288b8fe63681 : SUCCESS in 34m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ced7666059ef44848bd52a2e72eaf9df : SUCCESS in 30m 23s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fbc93e6c2799465984be7db6f3cb8e34 : SUCCESS in 50m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ba65555c1d4846c5b4bdacbe8fdd5497 : SUCCESS in 32m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/447bdb767853439aa1517ae275f9744f : SUCCESS in 31m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/447c192451374564b6507db9416500ee : SUCCESS in 5m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c4d975b045b4eab9a81613c792d2fc5 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2440c9824e8c4b1893b03830d316118a : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1832ee80e1704bdab036fdb0f390a683 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24e276f721fe4213842258ba6827e52c : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b33746fc43db4e8da6c8737e0828df39 : SUCCESS in 7m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/df6c41cf55e14625987ce0b5c8f0ca49 : SUCCESS in 28m 42s","accounts_in_message":[],"_revision_number":1},{"id":"7bbe44653b500f277b1ead826fdd7f92a89c3326","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-08 02:11:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"000af07229f24978d3a6482cc7a52a1f98aae9dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-08 03:38:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/402538a547b54e778b5d27b4d06f9620 : SUCCESS in 38m 41s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/d6cb7149247444ff81b93ce40c5ec830 : SUCCESS in 45m 56s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/55d6d556c19445e7899d83a2038b2bac : SUCCESS in 38m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cf5718a0a903458e85de486b1d106f1f : SUCCESS in 37m 12s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2200122c917148faa382c63db54d47ff : SUCCESS in 1h 00m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5162faa84dd248c687aa44823fc0d48d : SUCCESS in 41m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/044090d16a0a4c75b13271784d550b7a : SUCCESS in 43m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef86d991c7274bf486bd7582a7c7c3e7 : SUCCESS in 6m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69db4fad291f4597be3c821dd5829e8a : SUCCESS in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1903242e2b0b480d8cceb0fe3e4cb328 : SUCCESS in 8m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37b7988f5470429b9b30889d4d507253 : SUCCESS in 9m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0acf611cb7d44262928150623ef34e19 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20e5fb3f647a4789a9a24be80e17d10b : SUCCESS in 8m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ee45d1f0c88a45d794655fcc4ff647dd : SUCCESS in 28m 10s","accounts_in_message":[],"_revision_number":2},{"id":"4db3d960da1b1c29797e283aec564c8825db291d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-09 01:29:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"72435e834d0073a699d0dab0addbc290033349d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-09 01:59:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9257dd77b397447d295fb6af6c099176f7a048f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-09 02:00:28.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"beef3de30d537146b45e5411393190a73239724b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-09 02:21:28.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"37c945d679977da495509f48289d7428b6d0cab7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-09 03:38:38.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e8fd3262a78f4448ae88c870f4dde0cc : SUCCESS in 39m 57s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/05d04db998cb4849b822e7b45e3fe3b0 : SUCCESS in 1h 02m 11s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/a923e7032aff4fcd8a9979180095bb52 : SUCCESS in 36m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2d72cfb029bf4c59a33db088ede69403 : SUCCESS in 37m 29s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/96f0085a4c024d378fb9bb533ae405dc : SUCCESS in 57m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7043c4817e454180b14667b6889922f8 : SUCCESS in 42m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2e827f21319a4ba18f266d608892a959 : SUCCESS in 44m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c6cfbc341474cf7977598becb50bf7c : SUCCESS in 8m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/784bad206d2744538b44455278a02834 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f49607f755d84f1ebaa9eeed6db40cbf : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6e53d60b2bf482da19f9d56e73f1bdf : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/471ce910f8f14f5e86c1974229568e2b : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65dca36ff73045c9862ded1cba9affc1 : SUCCESS in 8m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/905f88300bf84143902543aa7d746ee6 : SUCCESS in 30m 44s","accounts_in_message":[],"_revision_number":6},{"id":"90e352590c0098ac4e3a06946f38eaffb0ee2a13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-09 03:45:17.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f0fce43b9e8737312d5fb75fde762f385d9675fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-09 04:53:55.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9370d1ac6d79497d9cbe0bc0e668f121 : SUCCESS in 44m 53s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/f2262a28e0b144e99ce90ad39014a387 : SUCCESS in 45m 36s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/b7af83a029644d9da9c6adcd591609b0 : SUCCESS in 45m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/68f6c85cffaa4a6cad0a9a24bf90efe4 : SUCCESS in 36m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6b4fb0cf35b54c84a89342a62886e535 : SUCCESS in 1h 03m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/870e95452496470db6b8114ab4d51e2c : FAILURE in 43m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e5aea9eda8994967b9e596cff2b9d6cd : SUCCESS in 35m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd5443c4f31f4f18af1b29692f64cdf0 : SUCCESS in 6m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b624ea9f6a4e41c7ba4e2fc68c366c06 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f74c5906d20475ba90a1ccc556c81b6 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9661d748ab2345d49511011158e26a26 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3dc62dcd74534ee2b8db03706d0616a6 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b4df6fa29af4b3f8b0767d3a90bf8aa : SUCCESS in 7m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4412ab60e53844acb5d67703942f608a : SUCCESS in 30m 05s","accounts_in_message":[],"_revision_number":7},{"id":"d9c7cec0492350cfaea22238004e4c320e933aa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-09 06:10:35.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"85092b1ee4efc081504410626b3b89afa4cab96a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-09 07:07:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8fc3408d44df4e298fa64239f23ab784 : SUCCESS in 37m 14s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/430cd49eb48645898905f0eb03730870 : SUCCESS in 42m 58s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2ea6e91e63cb4744acb1c723d156cafa : SUCCESS in 36m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4333e130d761487b9731927bb8ef067e : SUCCESS in 36m 35s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0c25ab05b62f474196ae3b8e5f09fe07 : SUCCESS in 54m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4e2daf74a997453386027f02ec1687de : SUCCESS in 33m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/38e628266a0c433ea174f2e0564e6398 : SUCCESS in 29m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cedc70c344a94767b449001efe37d72a : SUCCESS in 6m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fd8ff882e6f48b4ac440d04d506c117 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8523c0fcc8d847a087af951f825459ba : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d9dd01d9c574e8aa2214c228c3617d2 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac4b17d67ab54cb4967f11738a7e2cf8 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cca9c6a384d649a0bf8d2345d2017fc8 : SUCCESS in 7m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cf4c79587f344ee1b73266b5f6609857 : SUCCESS in 31m 19s","accounts_in_message":[],"_revision_number":8},{"id":"621051025726658d9257aad4558cd61bd8405457","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-09 08:21:54.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"561a19d7c1bc84a2ae2250252886ff2b603016f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-09 09:29:57.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d924119d94714389bb505281b6a1c07f : SUCCESS in 49m 26s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/385226e1aeea45fb91e8e0f600dbfbd3 : SUCCESS in 54m 24s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/145aa1de380b488ba36b20ed52983f4f : SUCCESS in 40m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/81cbff2d9b26436dab2af290d99a6df0 : SUCCESS in 38m 38s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c2db19cd194b43a898bc93956d47fa8e : SUCCESS in 1h 03m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/81e84c247eb04ce8b1f0f5e49a923886 : SUCCESS in 46m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/75194bcb58a84e30a823db8ccaa2c984 : SUCCESS in 48m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/294ee09f6fe240fa9bf18dadb713f717 : SUCCESS in 6m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/701e2bf988e946bc9bb882aadf69f93f : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e01228fea624417a2accdbab8544527 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/028439436c674ab99b2b23b481dc17ed : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56cc72c784df461d9fa25526e8b585a2 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d590c4bb4ea140f088a445213045084c : SUCCESS in 7m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1612adf0323d49e29dc55a38634c01cd : SUCCESS in 30m 45s","accounts_in_message":[],"_revision_number":8},{"id":"3b6e92d441027207088fc0ccaa680e955bf5fc17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-09 10:27:19.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"828117d7dbedf419f3a8779d7a7681821cbed5d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-09 11:39:14.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4c6e24fe027a4c678b1aa2e942b82559 : SUCCESS in 35m 46s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/5cfadd278b6c407c9e1e8ef13023bb04 : FAILURE in 1h 10m 27s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1aae02e4f266483c84ea81844e887109 : SUCCESS in 36m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d6ed3e82098c47e4af7c1422ece75928 : SUCCESS in 35m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/875e11249b2f4c60ad52b60d12c5c9b2 : SUCCESS in 59m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/50fc971e12684a7eb8a52326c473c1fd : SUCCESS in 36m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a76ec5c4e8574fa389cb2bf8204e5ca2 : SUCCESS in 36m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00b50769e03343cfae826db5aed29f92 : SUCCESS in 5m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a05ae4a198054cfa80a7fc25f4fe9ac8 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29102e19fd3f4aab8c532540f70e681d : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ed82a02e5a6473891cdb5df451e8619 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d0894207c7e4a0aa9fc28f79a985e5a : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/747f03d3690d41aeb331de56e23e1e99 : SUCCESS in 9m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ae468b59090b4ef09d967f79339898c5 : SUCCESS in 31m 48s","accounts_in_message":[],"_revision_number":9},{"id":"d8b3a65e8bda6e99d266e8dc52afb8dbd18ea40b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-10 08:50:49.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"08aa9937c865c43f79e06854e4b62147c000c4ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 09:51:35.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f6508ae9a0e141a9a07fdadb85910f0a : SUCCESS in 38m 41s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/100941970da74194b323e21eb2d2bfd9 : SUCCESS in 43m 48s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/746c45ab1b8e4f219f2f86c3b19c501a : SUCCESS in 42m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4722b471d14547b4a935f0ea69c777aa : SUCCESS in 38m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5f8686718fe84b90b1c6d5f0d2ea10a3 : SUCCESS in 57m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0db6cec3b86a4b7d84fe123e24093d9e : SUCCESS in 42m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8ff1c89b15b54e1fa7622d02ffa135e6 : SUCCESS in 35m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10f865248b994262a4152c6f5824d989 : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8eadb54548640c7b5f950b9afe99b4b : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7a940e60253498db78a33fa2b0af111 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ada55ccde0c848e792c924d8ad6d35f7 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad05945b20f44726a65bfa0be8f1a804 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b960ab6020b94a419d8e919b8ab4a884 : SUCCESS in 7m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/72777e512c3e475f8344978d58f517b7 : SUCCESS in 37m 45s","accounts_in_message":[],"_revision_number":10},{"id":"e1fa5252ce103037851bcc039be38784849e0b2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-10 21:00:21.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"a9ecd41fc5b601dc5d5050471a25d28feb23f5ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 21:56:41.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1a1b644566374f73b05a9d969ded60f6 : SUCCESS in 40m 24s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/fa4a24730a744b39881b6c4bde1dce66 : SUCCESS in 50m 04s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/42f3c8b1454844bb97ef50a8297b9bb5 : SUCCESS in 35m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/65ada940f9344968a277cc750428f0d0 : SUCCESS in 37m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cce0e4d510874149bc72dfd1baad7111 : SUCCESS in 54m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4b11916e7127413fb73c82d83aa549a2 : SUCCESS in 39m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/454df4f73838403eb71ecc4e1c475c32 : SUCCESS in 36m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb61c6eec6cf457ca3edb8b872f72384 : SUCCESS in 5m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/005d2ea0482d4328aaaea5640d4a0b41 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/481faaf6a4814d64a6b7140f50bf410c : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d7f58cb020846599e81fc33436d562f : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28c59d009b3b4b8daf99d858b13043f5 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cabfcfa9b1864b84b88c07bd3341d800 : SUCCESS in 9m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/48b18ac2736546c5a9a2e45e74f57f53 : SUCCESS in 29m 53s","accounts_in_message":[],"_revision_number":11},{"id":"da7a66c12569cdca044e913e89fbcf4cdfa63df4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-01-25 07:47:37.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"ca4b5d295ecea49ab2089cae604d3e2bef3039da","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-25 18:00:15.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"6148dc9dfff13ef9391e34a8ff2ca091594b9df8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-25 18:05:36.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"bbaf16a259b71c0c8713dbbd462806bdbd22e35c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-25 18:06:31.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"2caedcb97b5df882beeb5627d9495958f404fb74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 20:30:09.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e3a1dcbe287d404eaadfcd2df8e01460 : TIMED_OUT in 1h 17m 39s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/14f40f0abc9f4e90958b126043258fcc : TIMED_OUT in 1h 18m 19s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f0a1dac70ac84a90b1a771661d881364 : TIMED_OUT in 1h 15m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c26a2e69d36b4cdb8a1c8cfee77f81cd : TIMED_OUT in 1h 14m 25s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9908b1f2d0544b85be882c9404f03745 : FAILURE in 1h 07m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/aacba5c376ce49a4a2107e3c8e45df22 : TIMED_OUT in 2h 04m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/32c71ffe1d0b4832a361728a18848b64 : TIMED_OUT in 2h 07m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a33bfc7ffe94cdbbfd119627926dc27 : FAILURE in 7m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65577c8649d448149435c90966a857f5 : FAILURE in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/debd34e3426043f1b0114911f5aa1669 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7022b3000c97431bababf2fa2cd52367 : FAILURE in 8m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/087ce40242ed48569fb2e2a531aacc53 : FAILURE in 8m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4f96b7613b24a9da28e0bc515887cc4 : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f82e0f5aecfe4a3d86fdaa96767eae4b : SUCCESS in 5m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9873ba2d38634f9a9a6718294c3e4ff1 : FAILURE in 40m 18s","accounts_in_message":[],"_revision_number":12},{"id":"0040d2e9db07f7ebe3f4173a57f0b87d62011f94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-25 20:59:24.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"9a3d9b282712c61ce2eb5e4e92b6a62d8416ea75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-25 21:01:45.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"180548d632560c4a01e8b6347f3ab8104a2420c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-25 21:03:12.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"95456e584e57f11f45b61fb9cf9077e97102df3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 22:10:48.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/af1b3b959b1046fea12d1bcd7cc3af62 : SUCCESS in 28m 48s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/a2e2c23e843842b38cbb9e7fd3143968 : SUCCESS in 40m 01s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/03d60d69531d442abec6dbd587a68524 : SUCCESS in 44m 24s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/566d648540b2480d96206b15ed368392 : SUCCESS in 42m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/88f516e7280d4514bea94b3da9717891 : SUCCESS in 30m 53s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0e088740228e416589233ffa700f8445 : SUCCESS in 1h 02m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/272889f0707444b4a9756915fe938553 : SUCCESS in 33m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4e99c6a1b79744969dab4771f5a338c2 : SUCCESS in 33m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ce05ad161c04988a9214c57c973dc06 : FAILURE in 6m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/335062ae6d954937af0a8daf09cc2e46 : FAILURE in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebaee406ad604578826baef6074dbaba : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1198caad4c094929aeab98240f1dfa6e : FAILURE in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df68b7d49d724f07ad9f2e02b82efede : FAILURE in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ece9d7fd41bb4a26a4ccc9fc584e5a51 : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78486339f40c4984b80127e889e34503 : SUCCESS in 5m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6abb49cfa8ba4ce287ff14093f808f92 : SUCCESS in 31m 34s","accounts_in_message":[],"_revision_number":15},{"id":"495943c2c3a44b88f9ce06bfdbc9707d45aa6892","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-26 05:08:42.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"4b827fd3770a8674d2089e85129815b88da49c5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-26 05:23:17.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"7ccb98c3c46658e69add4de2dce9d3d23b74d43f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-26 05:59:03.000000000","message":"Patch Set 18: Patch Set 17 was rebased","accounts_in_message":[],"_revision_number":18},{"id":"8b674e4c941086beb71cf04c3c17d46be3f375f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 07:01:14.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/76797209ec3e4ab398a1a4522500ad84 : SUCCESS in 33m 20s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/e6053ab8fd2243ba887f66bc5a397378 : SUCCESS in 40m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/aba82e96dcef43a19a7dd7b504250519 : SUCCESS in 34m 58s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8f298a6979da412abfaf935a3ac54ed6 : SUCCESS in 31m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/44c68d4f54a64e77a7a67e6145af754f : SUCCESS in 30m 48s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9e94b8430e3845c2ad49233aececd864 : SUCCESS in 49m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ec7f05cec7ab404eb5d11a4943ee1a5c : SUCCESS in 54m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/df0724831af44ceb8f2b9389c2be6969 : SUCCESS in 29m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92f45c8c49974e08bd60e3bd3b673e64 : SUCCESS in 6m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83fb4919aa7f48798e3768e02e800e5c : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9181d97e9504351a03d4387280985bc : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e72bdf7e1f649038591d8fbc7b5b929 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d58dbbd1f2904756ab8b358e94c10086 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4493e5bc88544abd8e63b4ceb8aab2e3 : SUCCESS in 8m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74ea8fcb4b9a4b5fb1fe63614f9ae4fb : SUCCESS in 4m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c6a72cf5a3a042d89379f55ac01ecd7d : SUCCESS in 32m 15s","accounts_in_message":[],"_revision_number":18},{"id":"7728a3c667badfe41ce380e302540603464ffeed","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-29 01:54:42.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"52cd099eae9e01bd168160223281fa36d1328c75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-29 03:06:53.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/38d51dae7c924f4d9e01780d64a254fe : SUCCESS in 51m 04s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/34bb7ce33a8b4f3fb9170b48e43caa16 : SUCCESS in 1h 01m 28s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f013b7bf628e49a8ad46973dabd02ed2 : SUCCESS in 54m 00s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/72e56492f5a642e5bb88327714be9cd0 : SUCCESS in 43m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/58d779456d8e4aa28d6771b0344d198c : SUCCESS in 46m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6be65ff0a4d948a1b1a434d1bb0e096d : SUCCESS in 1h 03m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/11bde156ec3545f7a948bfc852b308de : SUCCESS in 48m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/72ebc4460953431f84d970db3cfcd18e : SUCCESS in 43m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47d0023f975c4991be3d66423386c099 : SUCCESS in 6m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3af8572c243d463490f1838a8d5f5031 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59b14ad512894bb08c3a9e2ed53fd24c : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffab8df6cb2b44eea8e2906885a4805a : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/962b7f007fe14e37915046dbf3693467 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ffc3c402de04053a2feb8653a210c67 : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9c1bfa858e6421591589facfc17c26a : SUCCESS in 5m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ace395cc986f40fbb423e364123be493 : SUCCESS in 32m 30s","accounts_in_message":[],"_revision_number":18},{"id":"a0cf46232787f437b95d58db1f1598232420a5c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-31 06:13:26.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"e8b9bbe8fe734b0983186be8099396c34d4a6d3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 06:58:17.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c00211dd91654efc920b5544b0395ce6 : SUCCESS in 31m 34s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/d6082e965f794b18a35cf5bc14d8d4fc : FAILURE in 8m 27s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3a1ed88a24646be84a1af777e868580 : SUCCESS in 34m 55s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/6edd1773afcc465da9cce62fadca1299 : SUCCESS in 30m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/737c2f73826d4b62b4848e49c097691c : SUCCESS in 31m 19s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6fd551f4f98c412dacb14104763a0af1 : SUCCESS in 42m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e1021a64ae064008a307822f235f6d9e : SUCCESS in 32m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5f85fb3b332b473a8b726d698496e498 : SUCCESS in 31m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c84e315997e42c2adaec34320d42f4d : SUCCESS in 4m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe97080bf62d4ec88e9bf76e67d5a3a8 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19ccd60e382c45039b1be84c8d51fbae : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b4fa6328c334efaa6f87f959cf3527d : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8aff9d8b17584f9cbda12245fafeb128 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c4d1c3972ea42c3bea22bbdf2851c79 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e51cf9421f9c48189da5b98e5f5b3361 : SUCCESS in 4m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b0d28f462304489bb0f1b5a53d740ea4 : SUCCESS in 29m 14s","accounts_in_message":[],"_revision_number":19},{"id":"8118990f6da7ef5292d538bf890fa3fe667b436d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-02-01 03:59:13.000000000","message":"Topic none removed","accounts_in_message":[],"_revision_number":19},{"id":"3d1191852109a5a6563c118c205e0ed81a7a85c7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-02-02 06:59:39.000000000","message":"Patch Set 19:\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"334c1af8aebc7e1006806477a430bcc4fc1c9f8b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-02-02 18:15:35.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"1e14626df65088861c4819b1fd6e6756797db7f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-02-02 20:18:49.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"d7c087c32ccf393d3974ecac19db08bfc686b623","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-02-02 20:19:35.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"e0b12b8651c65a94539c2819e11f5d4b1e1ce2c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 21:22:52.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/04119f5c23754967a3100c6fc1429ff8 : SUCCESS in 42m 45s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/7f25320dce7e40b99c9eb7480432e320 : SUCCESS in 49m 21s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9af379c6e538455f86a70d6df9b23976 : SUCCESS in 41m 49s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/09a2573721794953a77dc80885f74302 : SUCCESS in 42m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d4da28f849ae4955a5b6c1c1271561bf : SUCCESS in 30m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7041789a4a844cfaa8c60a4cd75fd6cb : SUCCESS in 1h 00m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3f73fa1106b54175ba76844cfe95b8e5 : SUCCESS in 28m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/38732b2729464752b2af5af3df91f178 : SUCCESS in 43m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b7c9eddae914462ae3164809841277f : SUCCESS in 5m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3344c192bcec4692ac7212210f001ef9 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a28e136b4ff746218f1674a20c55e038 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cbde1ccd9484a1bb77197ebed99f5de : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05133c0f32a640a784eef4cccb78825e : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb907fd2ecd84e3f80dd553da9395c7f : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ae895c7fb134b83ab66013c2423c87a : SUCCESS in 4m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/04c980bd2fc849548b89816fab289a68 : SUCCESS in 28m 52s","accounts_in_message":[],"_revision_number":20},{"id":"d4eadd6481e0af0fd0d9a538c99713b0d369872d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-04 23:52:25.000000000","message":"Patch Set 20: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"dc774607966fb3845e966abe01dcaf99491dcd69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-02-06 08:01:20.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"4d4b4e1d4c5d3e5421d6054112ee2ee2b423550e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-02-06 08:02:57.000000000","message":"Patch Set 21: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"d9f14ef0f67a62a4c39c3a4f573aae0e35272ad7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-06 09:12:17.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0ba08955b67345cb93b7608363425ed2 : SUCCESS in 45m 16s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/e1cb7c17f83a4e1ba2ec25b8b95072d7 : SUCCESS in 1h 05m 41s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/07b4491670d542858bc76976b76384b4 : SUCCESS in 48m 38s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2d55899ea016491c9a3c49a9522c50d2 : SUCCESS in 45m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/69193fe6a9014cb996c9577bd4fa4cc4 : SUCCESS in 41m 54s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3902482a73a345a39d1a356b8b7fa097 : SUCCESS in 1h 03m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/338d55ddaeae493b8f0278424f6d2879 : SUCCESS in 42m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/df9ab5cc7895471bbb2a46ee41d86948 : SUCCESS in 45m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9386d30f0be74856a0599a0feb0ec3a9 : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2f716d5e84b464199b9e1bf3f4116ad : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/484d006c19e544f9b3da6b2088b70528 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9dd7d71906ef4fedbc31b7c4f0e4951f : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c59fa60b46e44720a55ad4163f8f4874 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8df54961b8e647dfa0be92c5825b6452 : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f2bd7640c734eec8da49b62438d754c : SUCCESS in 4m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/990c702fdfae4b9ea7799475899c9437 : SUCCESS in 31m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/62db6895633c456880cf1bc9c9e9c07a : FAILURE in 4m 54s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"f674d5a2b690b2051f1f9d1c8efda477212382d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-06 09:12:57.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":21},{"id":"24e289dd9e65899d2b6869100d47b2ca8c154f36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-06 10:07:18.000000000","message":"Patch Set 21: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/50561c1b00944c139554a52d6ba4501f : SUCCESS in 43m 47s\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f0578e6c8fb43859b1da966d796e927 : SUCCESS in 50m 09s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/96374a7b3d4a49babad5e0760facb459 : SUCCESS in 29m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/53b53f129bb4488fb6e2a0d5ae23d9ec : SUCCESS in 46m 35s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ba42c1064e4c487085a7937c1c9dfb23 : SUCCESS in 44m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9852ef6ce9934a4a8ad1b7a4bff860b8 : SUCCESS in 27m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/197e8f629f1b4c30ac802521783d723c : SUCCESS in 28m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa6e5c306d5c4aebb117044fb2b1a265 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a79394fbafb41a4b48345d74babc050 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b0c231afb7740e0ae17d1d5d692efac : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b4db13066c648ecbce16589bd19960a : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b096d4ae810c4d44a91e2772b215c292 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7ea750a4235488e9a876d6c8b4509b7 : SUCCESS in 3m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f6cea92a4a1944e88d61c3d8a59ab3b7 : SUCCESS in 27m 07s","accounts_in_message":[],"_revision_number":21},{"id":"371c6c847b713d70e39b9027795330d6636c10f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-06 10:07:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":21},{"id":"fea47f9c64632a59f85c0a5b067e28fc6f945ab7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-06 10:08:43.000000000","message":"Patch Set 21:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71f59fe7620346b6a967b7e2eb9c4a2d : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a20598f9b6744ec4b261267765681595 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"3c495ed76dae17a0d74d4afe5b666f6068cafa47","revisions":{"3e9169d1a1ef0a778aa6d6255de0d218cec80b5a":{"kind":"REWORK","_number":1,"created":"2022-01-06 05:50:23.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/24/823624/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/823624/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/823624/1"}}},"commit":{"parents":[{"commit":"e08c1dae19d9aab6b582be95879fa60233751b58","subject":"Fix duplicate zone when creating ptr records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e08c1dae19d9aab6b582be95879fa60233751b58"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"subject":"Updated status logic to always notify on change","message":"Updated status logic to always notify on change\n\nWith the current logic there are scenarios where\nactions can be overwritten by other actions. This\nchange helps protect against that, and it also\nensures that a NOTIFY is always sent on change.\n\nChange-Id: I5d566588be66e9ed0df9484e36504a69b4f4b5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3e9169d1a1ef0a778aa6d6255de0d218cec80b5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3e9169d1a1ef0a778aa6d6255de0d218cec80b5a"}]},"branch":"refs/heads/master"},"274a410f60c71559dd48789920f11e9ee3f5195e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-01-08 02:11:55.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/24/823624/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/823624/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/823624/2"}}},"commit":{"parents":[{"commit":"bff3d5f6e31fe595a77143ec4ac779c187bf72a8","subject":"Fix designate-manage pool update bugs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bff3d5f6e31fe595a77143ec4ac779c187bf72a8"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-08 02:11:55.000000000","tz":0},"subject":"Updated status logic to always notify on change","message":"Updated status logic to always notify on change\n\nWith the current logic there are scenarios where\nactions can be overwritten by other actions. This\nchange helps protect against that, and it also\nensures that a NOTIFY is always sent on change.\n\nChange-Id: I5d566588be66e9ed0df9484e36504a69b4f4b5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/274a410f60c71559dd48789920f11e9ee3f5195e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/274a410f60c71559dd48789920f11e9ee3f5195e"}]},"branch":"refs/heads/master","description":"Rebase"},"08d3b5271cdf0c77f1ac05b8c18d020dddef9dad":{"kind":"REWORK","_number":3,"created":"2022-01-09 01:29:44.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/24/823624/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/823624/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/823624/3"}}},"commit":{"parents":[{"commit":"e08c1dae19d9aab6b582be95879fa60233751b58","subject":"Fix duplicate zone when creating ptr records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e08c1dae19d9aab6b582be95879fa60233751b58"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-09 01:29:21.000000000","tz":-480},"subject":"Updated status logic to always NOTIFY on change","message":"Updated status logic to always NOTIFY on change\n\nWith the current logic there are scenarios where\nactions can be overwritten by other actions. This\nchange helps protect against that, and it also\nensures that a NOTIFY is always sent on change.\n\nChange-Id: I5d566588be66e9ed0df9484e36504a69b4f4b5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08d3b5271cdf0c77f1ac05b8c18d020dddef9dad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08d3b5271cdf0c77f1ac05b8c18d020dddef9dad"}]},"branch":"refs/heads/master"},"ec3082ccdce23fe0bd92700e7a3deeb80de5d67c":{"kind":"REWORK","_number":4,"created":"2022-01-09 01:59:34.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/24/823624/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/823624/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/823624/4"}}},"commit":{"parents":[{"commit":"e08c1dae19d9aab6b582be95879fa60233751b58","subject":"Fix duplicate zone when creating ptr records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e08c1dae19d9aab6b582be95879fa60233751b58"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-09 01:59:21.000000000","tz":-480},"subject":"Updated status logic to always NOTIFY on change","message":"Updated status logic to always NOTIFY on change\n\nWith the current logic there are scenarios where\nactions can be overwritten by other actions. This\nchange helps protect against that, and it also\nensures that a NOTIFY is always sent on change.\n\nChange-Id: I5d566588be66e9ed0df9484e36504a69b4f4b5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ec3082ccdce23fe0bd92700e7a3deeb80de5d67c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ec3082ccdce23fe0bd92700e7a3deeb80de5d67c"}]},"branch":"refs/heads/master"},"69a4d59d22f248fff65266ceef73c43269ac1c72":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-01-09 02:00:28.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/24/823624/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/823624/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/823624/5"}}},"commit":{"parents":[{"commit":"1fd0e33569346503b83351fd34ee6ddb4f7a6e9b","subject":"Fix duplicate zone when creating ptr records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fd0e33569346503b83351fd34ee6ddb4f7a6e9b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-09 02:00:28.000000000","tz":0},"subject":"Updated status logic to always NOTIFY on change","message":"Updated status logic to always NOTIFY on change\n\nWith the current logic there are scenarios where\nactions can be overwritten by other actions. This\nchange helps protect against that, and it also\nensures that a NOTIFY is always sent on change.\n\nChange-Id: I5d566588be66e9ed0df9484e36504a69b4f4b5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/69a4d59d22f248fff65266ceef73c43269ac1c72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/69a4d59d22f248fff65266ceef73c43269ac1c72"}]},"branch":"refs/heads/master","description":"Rebase"},"69cd02e0d0c8440bad74864938f75f9847cfbc47":{"kind":"REWORK","_number":6,"created":"2022-01-09 02:21:28.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/24/823624/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/823624/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/823624/6"}}},"commit":{"parents":[{"commit":"1fd0e33569346503b83351fd34ee6ddb4f7a6e9b","subject":"Fix duplicate zone when creating ptr records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fd0e33569346503b83351fd34ee6ddb4f7a6e9b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-09 02:21:15.000000000","tz":-480},"subject":"Updated status logic to always NOTIFY on change","message":"Updated status logic to always NOTIFY on change\n\nWith the current logic there are scenarios where\nactions can be overwritten by other actions. This\nchange helps protect against that, and it also\nensures that a NOTIFY is always sent on change.\n\nChange-Id: I5d566588be66e9ed0df9484e36504a69b4f4b5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/69cd02e0d0c8440bad74864938f75f9847cfbc47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/69cd02e0d0c8440bad74864938f75f9847cfbc47"}]},"branch":"refs/heads/master"},"ee1ab56742f49a8fb104398da41fdb5c9a38c22e":{"kind":"REWORK","_number":7,"created":"2022-01-09 03:45:17.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/24/823624/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/823624/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/823624/7"}}},"commit":{"parents":[{"commit":"1fd0e33569346503b83351fd34ee6ddb4f7a6e9b","subject":"Fix duplicate zone when creating ptr records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fd0e33569346503b83351fd34ee6ddb4f7a6e9b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-09 03:30:55.000000000","tz":-480},"subject":"Updated status logic to always NOTIFY on change","message":"Updated status logic to always NOTIFY on change\n\nWith the current logic there are scenarios where\nactions can be overwritten by other actions. This\nchange helps protect against that, and it also\nensures that a NOTIFY is always sent on change.\n\nChange-Id: I5d566588be66e9ed0df9484e36504a69b4f4b5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ee1ab56742f49a8fb104398da41fdb5c9a38c22e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ee1ab56742f49a8fb104398da41fdb5c9a38c22e"}]},"branch":"refs/heads/master"},"1a758d7e02d48a013c9bc56353a11aa5ad4744d4":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-01-09 06:10:35.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/24/823624/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/823624/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/823624/8"}}},"commit":{"parents":[{"commit":"6a17d05b82f3e4c3120166310e2c32027da67a12","subject":"Fix duplicate zone when creating ptr records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6a17d05b82f3e4c3120166310e2c32027da67a12"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-09 06:10:35.000000000","tz":0},"subject":"Updated status logic to always NOTIFY on change","message":"Updated status logic to always NOTIFY on change\n\nWith the current logic there are scenarios where\nactions can be overwritten by other actions. This\nchange helps protect against that, and it also\nensures that a NOTIFY is always sent on change.\n\nChange-Id: I5d566588be66e9ed0df9484e36504a69b4f4b5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1a758d7e02d48a013c9bc56353a11aa5ad4744d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1a758d7e02d48a013c9bc56353a11aa5ad4744d4"}]},"branch":"refs/heads/master","description":"Rebase"},"ff817d712190b5d442f762ae87f4388dd0cb4b3f":{"kind":"REWORK","_number":9,"created":"2022-01-09 10:27:19.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/24/823624/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/823624/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/823624/9"}}},"commit":{"parents":[{"commit":"6a17d05b82f3e4c3120166310e2c32027da67a12","subject":"Fix duplicate zone when creating ptr records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6a17d05b82f3e4c3120166310e2c32027da67a12"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-09 10:23:19.000000000","tz":-480},"subject":"Updated status logic to always NOTIFY on change","message":"Updated status logic to always NOTIFY on change\n\nWith the current logic there are scenarios where\nactions can be overwritten by other actions. This\nchange helps protect against that, and it also\nensures that a NOTIFY is always sent on change.\n\nChange-Id: I5d566588be66e9ed0df9484e36504a69b4f4b5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ff817d712190b5d442f762ae87f4388dd0cb4b3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ff817d712190b5d442f762ae87f4388dd0cb4b3f"}]},"branch":"refs/heads/master"},"55e256a23597aa40fbc1fcf12f2f93515f114b4c":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2022-01-10 08:50:49.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/24/823624/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/823624/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/823624/10"}}},"commit":{"parents":[{"commit":"6a17d05b82f3e4c3120166310e2c32027da67a12","subject":"Fix duplicate zone when creating ptr records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6a17d05b82f3e4c3120166310e2c32027da67a12"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-10 08:50:49.000000000","tz":0},"subject":"Updated status logic to always NOTIFY on change","message":"Updated status logic to always NOTIFY on change\n\nIn this patch we revise the workflow after the zone api has\nfinished updating the upstream dns servers after a change. The goal\nis to fix issues with actions that would overwrite other actions,\npotentially causing zone change notifactions to not be sent in a\ntimely manner.\n\nAdditional changes.\n- Changed update_status method args\n- Improved unit test coverage of multiple code paths.\n\nChange-Id: I5d566588be66e9ed0df9484e36504a69b4f4b5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/55e256a23597aa40fbc1fcf12f2f93515f114b4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/55e256a23597aa40fbc1fcf12f2f93515f114b4c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"47ca776ac4e81d90851e7470b9cc7a9b666116c9":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-01-10 21:00:21.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/24/823624/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/823624/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/823624/11"}}},"commit":{"parents":[{"commit":"75dfe6707aaf0360736a89dd566d3104e81a4a7d","subject":"Fix duplicate zone when creating ptr records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75dfe6707aaf0360736a89dd566d3104e81a4a7d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-10 21:00:21.000000000","tz":0},"subject":"Updated status logic to always NOTIFY on change","message":"Updated status logic to always NOTIFY on change\n\nIn this patch we revise the workflow after the zone api has\nfinished updating the upstream dns servers after a change. The goal\nis to fix issues with actions that would overwrite other actions,\npotentially causing zone change notifactions to not be sent in a\ntimely manner.\n\nAdditional changes.\n- Changed update_status method args\n- Improved unit test coverage of multiple code paths.\n\nChange-Id: I5d566588be66e9ed0df9484e36504a69b4f4b5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/47ca776ac4e81d90851e7470b9cc7a9b666116c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/47ca776ac4e81d90851e7470b9cc7a9b666116c9"}]},"branch":"refs/heads/master","description":"Rebase"},"0f09f729cebcf49ca204dc7568e07d191b99d70f":{"kind":"REWORK","_number":12,"created":"2022-01-25 18:05:36.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/24/823624/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/823624/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/823624/12"}}},"commit":{"parents":[{"commit":"75dfe6707aaf0360736a89dd566d3104e81a4a7d","subject":"Fix duplicate zone when creating ptr records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75dfe6707aaf0360736a89dd566d3104e81a4a7d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-25 18:05:27.000000000","tz":-480},"subject":"Updated status logic to always NOTIFY on change","message":"Updated status logic to always NOTIFY on change\n\nIn this patch we revise the workflow after the zone api has\nfinished updating the upstream dns servers after a change. The goal\nis to fix issues with actions that would overwrite other actions,\npotentially causing zone change notifactions to not be sent in a\ntimely manner.\n\nAdditional changes.\n- Changed update_status method args\n- Improved unit test coverage of multiple code paths.\n\nChange-Id: I5d566588be66e9ed0df9484e36504a69b4f4b5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0f09f729cebcf49ca204dc7568e07d191b99d70f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0f09f729cebcf49ca204dc7568e07d191b99d70f"}]},"branch":"refs/heads/master"},"dda5acbb86789d2a2e3c5ffdceb3cd8bf674baf8":{"kind":"REWORK","_number":13,"created":"2022-01-25 20:59:24.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/24/823624/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/823624/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/823624/13"}}},"commit":{"parents":[{"commit":"75dfe6707aaf0360736a89dd566d3104e81a4a7d","subject":"Fix duplicate zone when creating ptr records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75dfe6707aaf0360736a89dd566d3104e81a4a7d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-25 20:59:16.000000000","tz":-480},"subject":"Updated status logic to always NOTIFY on change","message":"Updated status logic to always NOTIFY on change\n\nIn this patch we revise the workflow after the zone api has\nfinished updating the upstream dns servers after a change. The goal\nis to fix issues with actions that would overwrite other actions,\npotentially causing zone change notifactions to not be sent in a\ntimely manner.\n\nAdditional changes.\n- Changed update_status method args\n- Improved unit test coverage of multiple code paths.\n\nChange-Id: I5d566588be66e9ed0df9484e36504a69b4f4b5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dda5acbb86789d2a2e3c5ffdceb3cd8bf674baf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dda5acbb86789d2a2e3c5ffdceb3cd8bf674baf8"}]},"branch":"refs/heads/master"},"4fd2bb7da75e5aa91a06ffe95b22d93b05dbb94c":{"kind":"REWORK","_number":14,"created":"2022-01-25 21:01:45.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/24/823624/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/823624/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/823624/14"}}},"commit":{"parents":[{"commit":"75dfe6707aaf0360736a89dd566d3104e81a4a7d","subject":"Fix duplicate zone when creating ptr records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75dfe6707aaf0360736a89dd566d3104e81a4a7d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-25 21:01:37.000000000","tz":-480},"subject":"Updated status logic to always NOTIFY on change","message":"Updated status logic to always NOTIFY on change\n\nIn this patch we revise the workflow after the zone api has\nfinished updating the upstream dns servers after a change. The goal\nis to fix issues with actions that would overwrite other actions,\npotentially causing zone change notifactions to not be sent in a\ntimely manner.\n\nAdditional changes.\n- Changed update_status method args\n- Improved unit test coverage of multiple code paths.\n\nChange-Id: I5d566588be66e9ed0df9484e36504a69b4f4b5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4fd2bb7da75e5aa91a06ffe95b22d93b05dbb94c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4fd2bb7da75e5aa91a06ffe95b22d93b05dbb94c"}]},"branch":"refs/heads/master"},"ffd47afadae6e10f4eca251af204df235c5e08e0":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2022-01-25 21:03:12.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/24/823624/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/823624/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/823624/15"}}},"commit":{"parents":[{"commit":"c9a7f3392c87dedf5603e59b6821e74d297efe13","subject":"Fix duplicate zone when creating ptr records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c9a7f3392c87dedf5603e59b6821e74d297efe13"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-25 21:03:12.000000000","tz":0},"subject":"Updated status logic to always NOTIFY on change","message":"Updated status logic to always NOTIFY on change\n\nIn this patch we revise the workflow after the zone api has\nfinished updating the upstream dns servers after a change. The goal\nis to fix issues with actions that would overwrite other actions,\npotentially causing zone change notifactions to not be sent in a\ntimely manner.\n\nAdditional changes.\n- Changed update_status method args\n- Improved unit test coverage of multiple code paths.\n\nChange-Id: I5d566588be66e9ed0df9484e36504a69b4f4b5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffd47afadae6e10f4eca251af204df235c5e08e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffd47afadae6e10f4eca251af204df235c5e08e0"}]},"branch":"refs/heads/master","description":"Rebase"},"55f3e2d11c7706a4d7212f9f7c128915e6ea5271":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2022-01-26 05:08:42.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/24/823624/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/823624/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/823624/16"}}},"commit":{"parents":[{"commit":"93bab684f41660195591fb3652d56be88e91975e","subject":"Merge \"Add a BIND9 job on CentOS 9 Stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/93bab684f41660195591fb3652d56be88e91975e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-26 05:08:42.000000000","tz":0},"subject":"Updated status logic to always NOTIFY on change","message":"Updated status logic to always NOTIFY on change\n\nIn this patch we revise the workflow after the zone api has\nfinished updating the upstream dns servers after a change. The goal\nis to fix issues with actions that would overwrite other actions,\npotentially causing zone change notifactions to not be sent in a\ntimely manner.\n\nAdditional changes.\n- Changed update_status method args\n- Improved unit test coverage of multiple code paths.\n\nChange-Id: I5d566588be66e9ed0df9484e36504a69b4f4b5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/55f3e2d11c7706a4d7212f9f7c128915e6ea5271"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/55f3e2d11c7706a4d7212f9f7c128915e6ea5271"}]},"branch":"refs/heads/master","description":"Rebase"},"4d29c539756d071b2398680ce1a061e48940ec1b":{"kind":"REWORK","_number":17,"created":"2022-01-26 05:23:17.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/24/823624/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/823624/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/823624/17"}}},"commit":{"parents":[{"commit":"93bab684f41660195591fb3652d56be88e91975e","subject":"Merge \"Add a BIND9 job on CentOS 9 Stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/93bab684f41660195591fb3652d56be88e91975e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-26 05:23:10.000000000","tz":-480},"subject":"Updated status logic to always NOTIFY on change","message":"Updated status logic to always NOTIFY on change\n\nIn this patch we revise the workflow after the zone api has\nfinished updating the upstream dns servers after a change. The goal\nis to fix issues with actions that would overwrite other actions,\npotentially causing zone change notifactions to not be sent in a\ntimely manner.\n\nAdditional changes.\n- Changed update_status method args\n- Improved unit test coverage of multiple code paths.\n\nChange-Id: I5d566588be66e9ed0df9484e36504a69b4f4b5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4d29c539756d071b2398680ce1a061e48940ec1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4d29c539756d071b2398680ce1a061e48940ec1b"}]},"branch":"refs/heads/master"},"6187551f51fd20928ae86de1d548e1701e5d24d4":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2022-01-26 05:59:03.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/24/823624/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/823624/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/823624/18"}}},"commit":{"parents":[{"commit":"0c7d218ba103e8260322e40f76a49a8c92556bfe","subject":"Fix duplicate zone when creating ptr records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0c7d218ba103e8260322e40f76a49a8c92556bfe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-26 05:59:03.000000000","tz":0},"subject":"Updated status logic to always NOTIFY on change","message":"Updated status logic to always NOTIFY on change\n\nIn this patch we revise the workflow after the zone api has\nfinished updating the upstream dns servers after a change. The goal\nis to fix issues with actions that would overwrite other actions,\npotentially causing zone change notifactions to not be sent in a\ntimely manner.\n\nAdditional changes.\n- Changed update_status method args\n- Improved unit test coverage of multiple code paths.\n\nChange-Id: I5d566588be66e9ed0df9484e36504a69b4f4b5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6187551f51fd20928ae86de1d548e1701e5d24d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6187551f51fd20928ae86de1d548e1701e5d24d4"}]},"branch":"refs/heads/master","description":"Rebase"},"d22fddfdf0c61a9ff892a82e08982a8c3a2edce5":{"kind":"REWORK","_number":19,"created":"2022-01-31 06:13:26.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/24/823624/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/823624/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/823624/19"}}},"commit":{"parents":[{"commit":"0c7d218ba103e8260322e40f76a49a8c92556bfe","subject":"Fix duplicate zone when creating ptr records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0c7d218ba103e8260322e40f76a49a8c92556bfe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-31 06:13:15.000000000","tz":-480},"subject":"Updated status logic to always NOTIFY on change","message":"Updated status logic to always NOTIFY on change\n\nIn this patch we revise the workflow after the zone api has\nfinished updating the upstream dns servers after a change. The goal\nis to fix issues with actions that would overwrite other actions,\npotentially causing zone change notifications to not be sent in a\ntimely manner.\n\nAdditional changes.\n- Changed update_status method args\n- Improved unit test coverage of multiple code paths.\n\nChange-Id: I5d566588be66e9ed0df9484e36504a69b4f4b5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d22fddfdf0c61a9ff892a82e08982a8c3a2edce5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d22fddfdf0c61a9ff892a82e08982a8c3a2edce5"}]},"branch":"refs/heads/master"},"f4c5f7717fd398d21796fbfa4a7d720b8bf8df1a":{"kind":"REWORK","_number":20,"created":"2022-02-02 20:18:49.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/24/823624/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/823624/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/823624/20"}}},"commit":{"parents":[{"commit":"0c7d218ba103e8260322e40f76a49a8c92556bfe","subject":"Fix duplicate zone when creating ptr records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0c7d218ba103e8260322e40f76a49a8c92556bfe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-02-02 20:18:38.000000000","tz":-480},"subject":"Updated status logic to always NOTIFY on change","message":"Updated status logic to always NOTIFY on change\n\nIn this patch we revise the workflow after the zone api has\nfinished updating the upstream dns servers after a change. The goal\nis to fix issues with actions that would overwrite other actions,\npotentially causing zone change notifications to not be sent in a\ntimely manner.\n\nAdditional changes.\n- Changed update_status method args\n- Improved unit test coverage of multiple code paths.\n\nChange-Id: I5d566588be66e9ed0df9484e36504a69b4f4b5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f4c5f7717fd398d21796fbfa4a7d720b8bf8df1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f4c5f7717fd398d21796fbfa4a7d720b8bf8df1a"}]},"branch":"refs/heads/master"},"3c495ed76dae17a0d74d4afe5b666f6068cafa47":{"kind":"REWORK","_number":21,"created":"2022-02-06 08:01:20.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/24/823624/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/823624/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/823624/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/823624/21"}}},"commit":{"parents":[{"commit":"0c7d218ba103e8260322e40f76a49a8c92556bfe","subject":"Fix duplicate zone when creating ptr records","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0c7d218ba103e8260322e40f76a49a8c92556bfe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-06 05:48:15.000000000","tz":-480},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2022-02-06 08:01:20.000000000","tz":0},"subject":"Updated status logic to always NOTIFY on change","message":"Updated status logic to always NOTIFY on change\n\nIn this patch we revise the workflow after the zone api has\nfinished updating the upstream dns servers after a change. The goal\nis to fix issues with actions that would overwrite other actions,\npotentially causing zone change notifications to not be sent in a\ntimely manner.\n\nAdditional changes.\n- Changed update_status method args\n- Improved unit test coverage of multiple code paths.\n\nChange-Id: I5d566588be66e9ed0df9484e36504a69b4f4b5a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3c495ed76dae17a0d74d4afe5b666f6068cafa47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3c495ed76dae17a0d74d4afe5b666f6068cafa47"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
