)]}'
{"id":"openstack%2Fdesignate~825512","triplet_id":"openstack%2Fdesignate~master~Id265fa1dfa5c0703d8e95b5de1334d4ca716fbc0","project":"openstack/designate","branch":"master","topic":"caaemailsubaddr","attention_set":{},"removed_from_attention_set":{"19298":{"account":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"last_update":"2022-01-25 09:47:09.000000000","reason":"Change was submitted"},"32755":{"account":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"last_update":"2022-01-25 09:47:09.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-01-25 09:47:09.000000000","reason":"Change was submitted"},"5572":{"account":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"last_update":"2022-01-24 14:37:29.000000000","reason":"DEKehn replied on the change"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-01-22 21:10:18.000000000","reason":"Michael Johnson replied on the change"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-01-21 03:15:22.000000000","reason":"Erik Olof Gunnar Andersson replied on the change"}},"hashtags":[],"change_id":"Id265fa1dfa5c0703d8e95b5de1334d4ca716fbc0","subject":"Allow email subadresses to be used within mail url of CAA records","status":"MERGED","created":"2022-01-20 14:38:28.000000000","updated":"2022-01-25 09:48:40.000000000","submitted":"2022-01-25 09:47:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"825512-caaemailsubaddr","meta_rev_id":"23e7d76840ec9c5fca91baac516536a927ed2800","_number":825512,"virtual_id_number":825512,"owner":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-25 09:47:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-01-24 14:37:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2022-01-22 21:10:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-25 06:40:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-25 06:40:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"CC":[{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},{"_account_id":32823,"name":"Hannes von Haugwitz","display_name":"hvhaugwitz","email":"hannes@vonhaugwitz.com","username":"hvhaugwitz"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-20 14:51:28.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},"state":"CC"},{"updated":"2022-01-20 14:51:28.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32823,"name":"Hannes von Haugwitz","display_name":"hvhaugwitz","email":"hannes@vonhaugwitz.com","username":"hvhaugwitz"},"state":"CC"},{"updated":"2022-01-20 14:51:28.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2022-01-20 14:51:28.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-01-20 14:51:28.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-01-20 15:55:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-21 19:09:44.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2022-01-21 19:42:05.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"7cde1fcf7da1f2876d2c4de48e32cb196692b419","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2022-01-20 14:38:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16758649f9c74fe7fbf6973e907a0f410818e206","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2022-01-20 14:43:02.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ed6d4b291888b8ebec7d818917dc780c4e478ffc","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2022-01-20 14:51:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03b949debad5cb558fed50a58723441be4d89094","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 15:55:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e4d7228e13324c04b32356fe9a3056f8 : SUCCESS in 48m 20s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/130dff052c8e42ecb38c7b417f8d0aa1 : SUCCESS in 55m 09s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1bdb6508f55f4e82b7e3075898e23061 : SUCCESS in 45m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d20c45874b904bcf8115301832320f56 : SUCCESS in 45m 58s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/49d38dd857164087a3b12c2da3cb1bd7 : SUCCESS in 1h 05m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2943cac106d04eeea66a339c82d6b2b1 : SUCCESS in 45m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/63143cb39c124212a510db0cfb761b0e : SUCCESS in 56m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed33388470cf44e3bc5b7183572e7645 : SUCCESS in 6m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/983666cb4f014556b2fb5efcd9fb6b6a : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/602c4d5db57c4a799a2e47b32477f63c : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41290813f2054e749965feecdf17ecc8 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2351ec1121e548fa96faef3b040c22ec : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd7c12b9d7194c7d8bb7c9b5731e7798 : SUCCESS in 8m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8c70b81e660f4a478b8934bf5a9af918 : SUCCESS in 42m 30s","accounts_in_message":[],"_revision_number":2},{"id":"0f1df74aac24994637001181ed999af91bcb9191","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-21 03:15:22.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e3717507a2ad6cf2730277baf10cc101957df378","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-01-21 19:09:44.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"155fb5db9cc71b3ce80e60ce8e1d6b95acd7e408","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-21 19:42:05.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5d43447f16f7f9904daf0b01d7cb1ccd90b9d4f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2022-01-22 19:24:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0312d9c7edb9eadf9c934b7a038f93a83691c203","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2022-01-22 19:26:33.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a6e3907b59229ee35fc68a14d67ea35c2e7a8700","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-22 20:47:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bdb1993fec7c478e827b4d2b738d0246 : FAILURE in 59m 32s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/2382e5ccf23c4413b661e41586816671 : FAILURE in 1h 14m 11s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/bf66f582a19d46b7bc8434f291c3e204 : TIMED_OUT in 1h 13m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fbeb2501176b40ad803ff3940b29c2c5 : FAILURE in 1h 05m 39s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5c7536f0fcdf4f15b4c1519f42e507b0 : FAILURE in 1h 17m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c30124fcb8e8472aba3e7a69fcf139de : FAILURE in 1h 04m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/006a201e0b844d248b3002ce2a3a1004 : FAILURE in 1h 12m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f728455d46a544b48bdf1b84e6475ace : FAILURE in 7m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5de73ef6c9a4264bd0df271a7ba60e9 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76deb9ec7fa641cd8769c7f368bcce63 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7115f85187274220a4dadd622f4d5b20 : FAILURE in 7m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f0b0107227b4eb48b9bdde769b2f23e : FAILURE in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50797d93e1444b109ff85abd80e88688 : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abf207ca908442d9b1a620257c5c8988 : SUCCESS in 3m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/07e5e45be4b445ae987a3d3e02ed2a10 : SUCCESS in 33m 41s","accounts_in_message":[],"_revision_number":3},{"id":"5955556ff866770ed2f709afbf874ba6dc139854","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-22 21:10:18.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ca67b932cd9c34178613fc65d24f6bb406f05c74","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-01-24 14:37:29.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a2b68098247463d7d018d5ae4bd99ce0861627c2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-24 18:46:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0d9612826cb274ee5761bbbb4315aa614c221c18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 20:50:59.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8142dff80cdc408db2ad87324b233335 : TIMED_OUT in 1h 13m 45s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/0c469bdfa8534d55871b846c4e3d0167 : TIMED_OUT in 1h 15m 43s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8c6829ecbfe94d5fa049eb96a50c2cd8 : TIMED_OUT in 1h 14m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/17ad164d53074143b7633ce8d508fd63 : FAILURE in 1h 10m 42s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4f72f532edf1440c80c5b3a4f87ce270 : FAILURE in 1h 55m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cfcb147101504013a8947ae409950c9a : FAILURE in 1h 13m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e0c4b1cba15247c3845fd972ec617a0d : FAILURE in 1h 11m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4805aeb1e5b74e48a5364db6dc2aaa34 : FAILURE in 5m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9efaf2829d7471e9ed6753490cefd2a : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/186b282cfdb1410ca9a9e85f7ccb663b : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6747ed347e57426bb73659377b082662 : FAILURE in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38e32d3af7194407801ed20bfd520dd6 : FAILURE in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4326f626aa5a44109b7b65f4bb3358d6 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9bdd1bdb85d4a4b9b4edb1a3684db05 : SUCCESS in 4m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f0d71b3e78484f609341d740c33b0190 : SUCCESS in 27m 19s","accounts_in_message":[],"_revision_number":3},{"id":"5f8de134723bb74de937bc1f31c3c84db475edc2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-25 00:47:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3edeb871f3a81c2046cf887ffae65bda4c5cea42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 01:51:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c68145d058554c3abcc58ee7293fa38f : SUCCESS in 41m 51s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/2eb655d2c3db457ea6783e90e6366469 : SUCCESS in 1h 03m 24s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/7327569c6dfb457599f3d586ca7bc983 : SUCCESS in 39m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fd6f5f95f9714ccca44697e25a3a779c : SUCCESS in 37m 59s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1b0cb234bb7241f78f54d18377e362ec : SUCCESS in 58m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6be417a03c9444c2a084d9e2fb8fce5c : SUCCESS in 37m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/541b6b1c5f3c4b2896a3258dc997b5e0 : SUCCESS in 41m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73e6085615804be9a03ce0ab8690b849 : SUCCESS in 5m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/beaf94052f0045afa863283c4a2d9d5f : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0092b504a5f45da8e51489a725f4a99 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ba47bd1aa004fb9b9077e918df9d33d : SUCCESS in 7m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29c0ed6e9bb04124a6932ba28fb66cd2 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4a51179cbb84ef5a142e5d52b528efb : SUCCESS in 8m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aea101069b5b4df9a23767c93cdc81ed : SUCCESS in 4m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fb5ec779ae5a4dd4920202ea5250c573 : SUCCESS in 36m 40s","accounts_in_message":[],"_revision_number":3},{"id":"0da76b4198d5260eaf458a67c0d9d3289c5afe8b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-25 06:40:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"48e7ce0fff0000710536390b97c7312063c4c959","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 06:40:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b6f15d7b8954f17af9eb4f73d883087c7cc63906","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 07:37:16.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c62002b1724645649bc7fc791446c283 : SUCCESS in 32m 03s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2c5408576f354b22879a5084affd1d85 : SUCCESS in 28m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/32aad25a2d9941529fe6d731bebcb90f : SUCCESS in 32m 12s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/62884e09c29f435b9070e21a6c1c6bfd : SUCCESS in 55m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5fbc3da5c42d4d2ca43b7bb87fe33e22 : FAILURE in 21m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c74faf0e8af04d01a721c3562d40ab40 : SUCCESS in 32m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/075b7c0de09d4647b390176383b1ee3a : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6ea8702fb9b4bd6ab20fb715b24fc48 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d92576bac42a42cc9b1de17faa3c1955 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84b602ac1a604dedba017ca2d838f042 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3df2fe4230e0462a81945e0797eaf20b : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4dd1a37f66c4427590924d49c4c03897 : SUCCESS in 4m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bd319c0d9a3d43f5852e280b3f938e37 : SUCCESS in 26m 53s","accounts_in_message":[],"_revision_number":3},{"id":"50e551b7899042abfbbd89f9debd12d153ddc2b8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-25 07:50:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5d589c96600e76734e408b091db1c81386210740","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 09:03:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fed24262d92e494ca5776f366c44ea1d : SUCCESS in 27m 51s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/cdfa2f130ad84e908e0e8657283dda7a : SUCCESS in 1h 07m 35s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/b646328996e946ffb6b2afbf2c19c701 : SUCCESS in 37m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/459c5ae117af41538241eeeb59287be4 : SUCCESS in 40m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/181e20867ae24cfbbe7df6d51b43a8f8 : SUCCESS in 54m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fd2dce20fe9c471c9b4def250d39deb6 : SUCCESS in 37m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/dae0b7ac3b1e4e038db49f69c1115852 : SUCCESS in 49m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62bd1df66e20493090539ee3906ae987 : SUCCESS in 4m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6eb3440520240acb1f58f5b0b290725 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b4743854325405d8355023d7484a63a : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79f13dd4c5f04c9cb344724149e2da4f : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1cc75b8e95bb4fc4b41daf6b99ae389f : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08ca13de7a9940d3a4851b5b04e27b26 : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab82c8da3fab465d83dc0841968f95ea : SUCCESS in 5m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/957aba5188c2447183263c5165fc0c91 : SUCCESS in 35m 37s","accounts_in_message":[],"_revision_number":3},{"id":"520356e3f06790fa9801bea56027f3835d953ceb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 09:04:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a78450c9bf5ba4f8278f1d45bf93691181685cf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 09:47:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/16b0fc10934246cd96c0d7a10868a9b0 : SUCCESS in 31m 28s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/6221b02b254e456b81d32bacc8802fba : SUCCESS in 39m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c4a28854e6044684b6b4517bd307b405 : SUCCESS in 34m 40s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b5cdf0a3424e4bb7ac96e4f13606a21f : SUCCESS in 42m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/08f7cb35a4ce4c11a697b41a15eb288b : SUCCESS in 31m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/89c9feb6c8454836bf509911f491755d : SUCCESS in 37m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b02e8d2e4863430c9d279c7ccc678bc6 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ae91890ea2f431096e786f2f71b274c : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37db89a943034c219c6613d4fe05ca50 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed457be95e0e44c0bc46c71d90d0cfe2 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c423a11bf694e0f98645706fdb8a766 : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0638eb67d38b424ea3802c2155d1d13d : SUCCESS in 4m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5a78958b08174a25a28ff36f24639cff : SUCCESS in 28m 32s","accounts_in_message":[],"_revision_number":3},{"id":"50f46c68496b4b7a7009e92d33b4a35615c8bfeb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 09:47:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"23e7d76840ec9c5fca91baac516536a927ed2800","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 09:48:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59325f8bdbd94e64afa453566bbd7f9c : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c5b3c559c0d41c9bcf9652d18e5b3e2 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1cef20edbc462fab75a919fc0de796c2d6edb30b","revisions":{"13a92b7b97538655e99a647801ae635e8868b631":{"kind":"REWORK","_number":1,"created":"2022-01-20 14:38:28.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/12/825512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/12/825512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/825512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/825512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/825512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/12/825512/1"}}},"commit":{"parents":[{"commit":"bff3d5f6e31fe595a77143ec4ac779c187bf72a8","subject":"Fix designate-manage pool update bugs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bff3d5f6e31fe595a77143ec4ac779c187bf72a8"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2022-01-20 14:20:35.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2022-01-20 14:38:00.000000000","tz":60},"subject":"Allow email subadresses to be used within mail url of CAA records","message":"Allow email subadresses to be used within mail url of CAA records\n\nThis patch allows for +subadress to be used for the local part of mail urls\nfor CAA records. There are more characters allowed in email addresses, using\nsubadresses for incoming email to role accounts is quite common.\n\nCloses-Bug: 1958533\nChange-Id: Id265fa1dfa5c0703d8e95b5de1334d4ca716fbc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/13a92b7b97538655e99a647801ae635e8868b631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/13a92b7b97538655e99a647801ae635e8868b631"}]},"branch":"refs/heads/master"},"43218f06a878b6e1ba51aa918064008d42eada89":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-01-20 14:43:02.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/12/825512/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/12/825512/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/825512/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/825512/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/825512/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/12/825512/2"}}},"commit":{"parents":[{"commit":"bff3d5f6e31fe595a77143ec4ac779c187bf72a8","subject":"Fix designate-manage pool update bugs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bff3d5f6e31fe595a77143ec4ac779c187bf72a8"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2022-01-20 14:20:35.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2022-01-20 14:42:04.000000000","tz":60},"subject":"Allow email subadresses to be used within mail url of CAA records","message":"Allow email subadresses to be used within mail url of CAA records\n\nThis patch allows for +subaddress to be used for the local part of mail urls\nused for CAA records. While there are more characters allowed in email\naddresses, using subadresses for incoming email to role accounts is quite\ncommon.\n\nCloses-Bug: 1958533\nChange-Id: Id265fa1dfa5c0703d8e95b5de1334d4ca716fbc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/43218f06a878b6e1ba51aa918064008d42eada89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/43218f06a878b6e1ba51aa918064008d42eada89"}]},"branch":"refs/heads/master"},"1cef20edbc462fab75a919fc0de796c2d6edb30b":{"kind":"REWORK","_number":3,"created":"2022-01-22 19:24:35.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/12/825512/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/12/825512/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/825512/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/825512/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/12/825512/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/12/825512/3"}}},"commit":{"parents":[{"commit":"bff3d5f6e31fe595a77143ec4ac779c187bf72a8","subject":"Fix designate-manage pool update bugs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bff3d5f6e31fe595a77143ec4ac779c187bf72a8"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2022-01-20 14:20:35.000000000","tz":60},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2022-01-22 19:24:12.000000000","tz":60},"subject":"Allow email subadresses to be used within mail url of CAA records","message":"Allow email subadresses to be used within mail url of CAA records\n\nThis patch allows for +subaddress to be used for the local part of mail urls\nused for CAA records. While there are more characters allowed in email\naddresses, using subadresses for incoming email to role accounts is quite\ncommon.\n\nCloses-Bug: 1958533\nChange-Id: Id265fa1dfa5c0703d8e95b5de1334d4ca716fbc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cef20edbc462fab75a919fc0de796c2d6edb30b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cef20edbc462fab75a919fc0de796c2d6edb30b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
