)]}'
{"id":"openstack%2Fdesignate~827248","triplet_id":"openstack%2Fdesignate~stable%2Fwallaby~I2b69025fc11125bb73a4e0f8c0dedad951399cbf","project":"openstack/designate","branch":"stable/wallaby","topic":"bug/1934252-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"5572":{"account":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"last_update":"2022-06-04 21:27:15.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-06-04 21:27:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2b69025fc11125bb73a4e0f8c0dedad951399cbf","subject":"Checks for invalid denylist regex patterns","status":"MERGED","created":"2022-02-01 18:49:40.000000000","updated":"2022-06-04 21:28:49.000000000","submitted":"2022-06-04 21:27:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827248-bug/1934252-stable/xena-stable/wallaby","cherry_pick_of_change":827246,"cherry_pick_of_patch_set":1,"meta_rev_id":"d876197fb4f7b940f93cb13f9897365105ad6c9b","_number":827248,"virtual_id_number":827248,"owner":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-04 21:27:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2022-06-04 21:28:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2022-06-04 20:43:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-03 17:39:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2022-06-04 20:43:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-01 19:52:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-12 15:12:05.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2022-05-26 01:44:53.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-05-31 22:51:34.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"44a0d08bc1ace19a0e0fd26939d1f91549a51c9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-02-01 18:49:40.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"1afb199602f49e0e23695dbd6b0d5766d4ecacb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 19:52:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/464f86e0966b434283c132b33aa77ef6 : SUCCESS in 34m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9a00e5d59cde40e78322d066a744b0f6 : SUCCESS in 41m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ab27990539a349d7a8337e4e7db8df6a : SUCCESS in 59m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f2517beb0c754149a9851e4d40418dd9 : SUCCESS in 36m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d9b172b0acf64747b6bdb2e7bc5973a8 : SUCCESS in 39m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34a39559b2e24205a9f34821594480ea : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2519374e548d4fc5836f24110ccb5f4b : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/334c72c29be0413285403b38943101c4 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d614ebdab38f45b998c30291bab21b13 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea4f909a36c6471787c03fe1570ee1a3 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5adc38e4684441aa3918b1bf6b3b38a : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f419562f84aa4ce6a8ffb27d02d10dea : SUCCESS in 4m 33s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/b4fea50b42bc4038b03c376823cf9ede : SUCCESS in 25m 55s","accounts_in_message":[],"_revision_number":1},{"id":"94097e29fa0c3fe80ef5baa37d6e36d6e7ff9e09","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2022-05-12 15:12:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"38c6c08349550b2534c5d291d3bc80571b3331ee","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-26 01:44:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b4e4276a5819b8bad64de57d4190bab0c345b2d8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-31 22:51:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af1365a0649a880157120ea5be5d02324d7ff174","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 22:52:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"744287b72fb8953e8027857b8cc904ff43266613","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 23:47:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/26a28fb853154450acf4460407e44cab : SUCCESS in 39m 10s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/9a2ce90a61094b198980da0da1c87e82 : SUCCESS in 40m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/31636c92edc643ad8cc1fa0b775f6cda : SUCCESS in 36m 36s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6f01c9b8cace44309567782a38d78909 : FAILURE in 52m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d68cdbeb2b4f4577aefb6e328168b3c5 : SUCCESS in 35m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/979a215462324e73b586cc1f4df873a6 : SUCCESS in 40m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f12556f55e3484cbe2cdf1c0ad7b9bf : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fda6317f2575497a837674508a35a7e5 : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e86ca0e9c1b49129a40983ff20d523e : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8174ab5a3744d5b83dcde0f8b157581 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/241e214572794bf39457db51e089a892 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"33ae898eeceecb8ad428f8fe16e502d7a6d6be60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-02 04:34:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"84a1a017cb3ae9917ef69102ab2cce53b17d7365","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 05:26:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1f90c3aa05fa41bfac6a3f71e3ecdb60 : SUCCESS in 34m 33s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/f0f60a0ed01c4afda618778d99650458 : FAILURE in 19m 15s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/953f0fd52fcc4869b391224f818fba29 : SUCCESS in 36m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f18222ac305d444ba54fbfd220e019c4 : SUCCESS in 37m 48s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3b8def67f7174ebdbcec252adb73c49b : FAILURE in 48m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/08e908f49b99450ba9653279a19c6887 : SUCCESS in 34m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cb348f459aa94f1191998dd9cfb453fe : SUCCESS in 32m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aac9803920944b388762b8f01e2d0b46 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3647355ff6344b81ae30347442cf18ff : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb744528210447f8801c3d2897ba583f : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a456c7f70024e6ba9b22f771b7f8725 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7d47b849db8468ca70faa36932d1fa9 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/998e10c30634468c9b037f99bb2e90aa : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75d9704ae55e480086c6e044a31cfadc : SUCCESS in 3m 53s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/f8bf6e9a6b0e48b7b7bbcc189ee8a558 : SUCCESS in 29m 48s","accounts_in_message":[],"_revision_number":2},{"id":"8ab301e3ea474105a0eccde7db93cfc8b010c47a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-03 17:39:19.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1a5ba04a125fdbb71e2b2f85a6629d92285b4c18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 18:39:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6855dcc685fc4f7388fd3333719ee5ea : SUCCESS in 47m 40s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/044672f2d2e4465086c7752defa95ad2 : FAILURE in 32m 13s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/34c59016d8cb4c37ae364e09b2a72ad8 : SUCCESS in 50m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a547ec8b816142df863e5e9cb0c045a3 : SUCCESS in 53m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/382b20933c6742d297a3a9446055ba82 : SUCCESS in 46m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/59413eb1c2b54b5ea1956328944a8115 : SUCCESS in 47m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25cbfad31c904c13b9b7c17351c66704 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ff1f35c0c214b2a8d13420180ff1753 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b9b87a96bdf48439aeed6bbc39fc837 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6be179bc4ffb4c3fbcaf4bf52a74698f : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fb5bf63edcc4409b82c936f65369302 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b373923e5c11456ebf584e5d89114f94 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34a0dac100124c6980f0146a45194598 : SUCCESS in 4m 01s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/049d0c44543c4a34bd00f0d78e4bae73 : SUCCESS in 35m 29s","accounts_in_message":[],"_revision_number":2},{"id":"74029ca09b12de013e9bdb1bfb5ef1e17f300955","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-04 20:43:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"19d0c55d209e463525b15371610ec4cf9ca850fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 20:43:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"509fe0ab2e251efc980ad4355b61dbcd34090119","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 21:27:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dad443e7e4954e1da919dad8ead76138 : SUCCESS in 43m 11s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/5501375791484e75b1081b87e78a99dc : SUCCESS in 39m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/388a60252da141daaea3f1cdd96cc967 : SUCCESS in 42m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/80592e02dd21415191ef13b7f36491af : SUCCESS in 39m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/858eb8cad83c4c379c75ca43cc5c98dc : SUCCESS in 39m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bc1e0c53e34480ead995848bcec71c1 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8dd03a8b072846e480e90a3ec24a9b62 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b6bde474af6439696acdbd21b37bc46 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dceba02bcdf4685a1533b00f479c999 : SUCCESS in 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/356ddd8e37af4fe9abc1f28b69e335fd : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":2},{"id":"bdb55f29cbe130ffa7a52adc002400b763fc6333","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 21:27:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7ad6f3e4204f93ddd4954e53c01700144e8b2b76","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 21:28:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fbdfd8bb8d1427ea505351ca4515044 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38caff7ca82d4c4d99d56175f71d2dc2 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a2ff328cd7c47dd53d02bcecf3e16473bdb8ba67","revisions":{"b8fa0a6f356fa5ff7335debbe82c6bb064989f0f":{"kind":"REWORK","_number":1,"created":"2022-02-01 18:49:40.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/48/827248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/827248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/827248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/827248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/827248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/827248/1"}}},"commit":{"parents":[{"commit":"914b55a631de7e1623cc635edf8dab5168328752","subject":"Remove lower-constraint job from stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/914b55a631de7e1623cc635edf8dab5168328752"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-01-20 20:38:06.000000000","tz":0},"committer":{"name":"Don Kehn","email":"dekehn@gmail.com","date":"2022-02-01 18:49:40.000000000","tz":0},"subject":"Checks for invalid denylist regex patterns","message":"Checks for invalid denylist regex patterns\n\nAdds new field check method DenyListFields to validate\nthe pattern string. in addition, check for a zero\nlength string as well.\n\nCloses-Bug: #1934252\nChange-Id: I2b69025fc11125bb73a4e0f8c0dedad951399cbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b8fa0a6f356fa5ff7335debbe82c6bb064989f0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b8fa0a6f356fa5ff7335debbe82c6bb064989f0f"}]},"branch":"refs/heads/stable/wallaby"},"a2ff328cd7c47dd53d02bcecf3e16473bdb8ba67":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-06-02 04:34:44.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/48/827248/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/827248/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/827248/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/827248/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/827248/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/827248/2"}}},"commit":{"parents":[{"commit":"2e55c3e24bfaf1b03241aa1a4a538d03c29d69ad","subject":"Improve quota API validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2e55c3e24bfaf1b03241aa1a4a538d03c29d69ad"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-01-20 20:38:06.000000000","tz":0},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-06-02 04:34:44.000000000","tz":0},"subject":"Checks for invalid denylist regex patterns","message":"Checks for invalid denylist regex patterns\n\nAdds new field check method DenyListFields to validate\nthe pattern string. in addition, check for a zero\nlength string as well.\n\nCloses-Bug: #1934252\nChange-Id: I2b69025fc11125bb73a4e0f8c0dedad951399cbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a2ff328cd7c47dd53d02bcecf3e16473bdb8ba67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a2ff328cd7c47dd53d02bcecf3e16473bdb8ba67"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
