)]}'
{"id":"openstack%2Fdesignate~827989","triplet_id":"openstack%2Fdesignate~master~I0c8e6a3a474a5cfa029b741b1a26b7d48d244206","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-17 23:43:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0c8e6a3a474a5cfa029b741b1a26b7d48d244206","subject":"setup.cfg: Replace dashes by underscores","status":"MERGED","created":"2022-02-05 16:42:19.000000000","updated":"2022-02-17 23:47:31.000000000","submitted":"2022-02-17 23:43:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827989","meta_rev_id":"2c3917879715ba0890f5863127f6791cc852e67a","_number":827989,"virtual_id_number":827989,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-17 23:43:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-02-17 21:19:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-05 18:31:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2022-02-17 21:19:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-05 17:41:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-05 18:31:05.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-02-17 21:19:18.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"9aca25a13ea092aa5614869768e9cf4b3d52c363","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-05 16:42:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57f6ddba71ff540ded479710b2c5eda93eb44329","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 17:41:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dec648cf35b04598922a5101f9dfbd84 : SUCCESS in 33m 50s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/e484e54aa8dd497aa98969d55abcf285 : SUCCESS in 41m 59s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/43e5bb3cdfbd49928d3d7379a7b3c9ca : SUCCESS in 49m 56s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ae5ab6b3ca224cc09e49fb2bc9aac654 : SUCCESS in 45m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/92a62c6ac5fc4657af54ea693af60ac1 : SUCCESS in 35m 59s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a838cab68bee4ad2ac41d00e4619d09f : SUCCESS in 47m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cb6ed1a2ce96407cb75f6c504fb734bc : SUCCESS in 35m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4dbd49e03f7e44b3ba00dcfcb88d749b : SUCCESS in 42m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73ca5c97f0a246b3b47d3b5e488678a0 : SUCCESS in 5m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b63e148e500c430082577000df0a4705 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b84eb05b4e824608baa2dc34eca61b02 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c489dbf40573477682d323d4caff4bbd : SUCCESS in 7m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/685d72a951574d8aa69808b213ee4b20 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9423ee88cf542ddb78e507dee15e293 : SUCCESS in 10m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e31579628ac44c4b813c9d09c15939a2 : SUCCESS in 27m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/875deeb649914f2c9ddd1b2e180c8b61 : FAILURE in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd41827c08b4a3bf6ec67e0ab6707d552e00281f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-02-05 18:31:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c3a7d6a62c8e658bd3d4a23726c9dcd9573df465","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-17 21:19:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ccdd32d31d55abdc769c72b8372065e009e92472","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 21:20:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c39d30226c2afdb6a49c5b872a1ef2068b7a616","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 23:43:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0d6fd3f643e343969ac4d2e52e1a4a5a : SUCCESS in 47m 16s\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/141e5f5735b846fa99a6e21527d77f41 : SUCCESS in 51m 21s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/d022cfd7240241958d5174ad302c4f97 : SUCCESS in 44m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8a56015738f34458bb26a01485bb3d48 : SUCCESS in 47m 08s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b4a70d2c651d46da9aa9b682c99a38e7 : SUCCESS in 1h 03m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2eecf84fc97c446aa53b197abf7a9b55 : SUCCESS in 40m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/dbbbd14c88464b50a328b9c6d6a9fce0 : SUCCESS in 53m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9023411f8794d2dabcc7ed9074a1a53 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/810a606129254683801c02464bac0abd : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdca4a9330ea401a80c2490d06686f75 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6733f33f4034cbd882ed18b23f1f6f2 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9ee9d2d613e4b64aea96a71320301ca : SUCCESS in 7m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/851bb9268e2e4739abfba7146e263d8c : SUCCESS in 29m 46s","accounts_in_message":[],"_revision_number":1},{"id":"5dc5b31d1739fdb748b1f2eb0ff10f4ed6a54990","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 23:43:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2c3917879715ba0890f5863127f6791cc852e67a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 23:47:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55e872ac99a749f6ae299b9e5ea05c22 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30b9a0853c007eca511dabd21022b5d10215fbfa","revisions":{"30b9a0853c007eca511dabd21022b5d10215fbfa":{"kind":"REWORK","_number":1,"created":"2022-02-05 16:42:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/827989/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/89/827989/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/89/827989/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/89/827989/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/89/827989/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/89/827989/1"}}},"commit":{"parents":[{"commit":"bff3d5f6e31fe595a77143ec4ac779c187bf72a8","subject":"Fix designate-manage pool update bugs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bff3d5f6e31fe595a77143ec4ac779c187bf72a8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-05 16:42:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-05 16:42:08.000000000","tz":540},"subject":"setup.cfg: Replace dashes by underscores","message":"setup.cfg: Replace dashes by underscores\n\nSince setuptools v54.1.0[1], the parmeters with dash have been\ndeprecated in favor of the new parameters with underscore.\n\nThis change updates the parameters accordingly to avoid the warnings\nlike the example below.\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: I0c8e6a3a474a5cfa029b741b1a26b7d48d244206\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/30b9a0853c007eca511dabd21022b5d10215fbfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/30b9a0853c007eca511dabd21022b5d10215fbfa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
