)]}'
{"id":"openstack%2Fdesignate~829150","triplet_id":"openstack%2Fdesignate~master~Ibd01f6ad3046ca29eec02032e9115183f89a6a4b","project":"openstack/designate","branch":"master","topic":"quota_errors","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-02-21 23:42:31.000000000","reason":"Erik Olof Gunnar Andersson replied on the change"}},"hashtags":[],"change_id":"Ibd01f6ad3046ca29eec02032e9115183f89a6a4b","subject":"Add proper quota error messages","status":"MERGED","created":"2022-02-15 02:18:11.000000000","updated":"2022-02-22 00:42:57.000000000","submitted":"2022-02-22 00:41:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829150-quota_errors","meta_rev_id":"b1a6700ca3e9c02ea71d21b832b4c4cf93bd8392","_number":829150,"virtual_id_number":829150,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-22 00:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-02-21 22:17:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2022-02-21 19:23:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-21 23:42:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-15 03:46:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-21 19:23:51.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-02-21 22:17:04.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"3bcf7ac9b9444fce635ddc2847dd02d2c620866a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-02-15 02:18:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e4a17ad2560bf9085de2b4eef0e4d6845b660a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-02-15 02:21:03.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"01679330d4a881fa67b12bf7c0d4fca5c695bbf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-02-15 02:45:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"98cadb4f14db84c2edad65c107cfec28ff6b1b4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 03:46:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f727d6963e0d41f7aa63be4068c6bfb0 : SUCCESS in 31m 40s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/7960f0809b054c70b7fab180c709d6a9 : SUCCESS in 38m 54s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f5d184052914950ab6395a639df8fb8 : SUCCESS in 33m 39s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/24b3470764a845ed86c89734cd1312f9 : SUCCESS in 28m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/99cb59c562bb4f21a119d7fb72e0e5ef : FAILURE in 39m 54s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/669e334f424c4183a1a1698390beb297 : SUCCESS in 55m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/248ca0e4177548a9bf4a53d76e35be25 : SUCCESS in 36m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b3b5270aa56243158e53bf8b4ebbfad5 : SUCCESS in 40m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/868f9699e4e7462b9b0e21807557e907 : SUCCESS in 8m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b69eae0dbbad4b67b392e3e58ce01309 : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea55f8bb2ed643e1971714bbb7943e6a : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/776da1c8ee634855b6e3c0f32c8f2d70 : SUCCESS in 7m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/093ab419b6b04d089bd0e60c73b9c3f1 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc08c35de95a496784d758acac39496a : SUCCESS in 9m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4eecd70caa694edbbfe640ecf97eb6cb : FAILURE in 22m 03s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ef0e6d3961c74fe2b1443c095859bd87 : FAILURE in 6m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a1a1356b9aac42fb7147d62287eac72dd2b28891","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-02-15 04:51:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ebce4b91d7335bcd8b26cd5239875e25326c3d1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 05:47:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8e4e026ebe40448cb25bd632f21b529e : SUCCESS in 37m 47s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/59d0fd11c3e5418abadc0d09d92dbd46 : SUCCESS in 51m 51s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e64ca1aa07a94631bd245339cfeaeee8 : SUCCESS in 44m 50s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4a2a304f773a483fbe05b8afa3cf6eb5 : SUCCESS in 37m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/13b45ea6588342fa864d48a3fe8f140f : SUCCESS in 35m 02s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9e1537c9d8c945a5b8a7425c974fd1f3 : SUCCESS in 49m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/26b534299a2c4bddb048ed14969428df : SUCCESS in 37m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/81eb6a47164a4f879f4623ceaa1878f9 : SUCCESS in 42m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9030b43caae24e2ca115b13d3d2ea1fc : SUCCESS in 6m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed116184ce694dc49a853166d8309288 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d509ae923054c39bc0214f46a3a2763 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/954aca52bc5d4dbeb6af1112bb3f2e13 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/621b4ffcd0a640cf81f35d99153b0944 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75954824090b4c4b917e7939e2ef40da : SUCCESS in 9m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5daef89959084248855f73a7dc41ddb0 : SUCCESS in 25m 02s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0a93923ec4d64f76a3d75e20e65c0ac0 : FAILURE in 4m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"22fe91eeef579a9116116bdf891a84216b9d5a0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-02-18 17:49:14.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"2a849566fa3c87ded62629a4ccbb46f70efee6a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 18:43:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5ce69ec4d0704d14ba69bbfd617c6691 : SUCCESS in 40m 33s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/56e564977269436b94ec74f82d98e0d6 : SUCCESS in 40m 06s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/643ca39358514ad7b6bbfea431b865c7 : SUCCESS in 41m 38s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3b0a4914e45c4fb68d8e45d07898d0c2 : SUCCESS in 37m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7fbda77921404f9e87dde3207cd1adf5 : SUCCESS in 32m 44s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1e737f39f70f4eee8ef4a024355ea90a : SUCCESS in 51m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4aae891ccdeb4c039f3006133e8219c5 : SUCCESS in 38m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/69c3b1838eeb4042bea4a72641ce7881 : SUCCESS in 40m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b8c5b140eae42dcabff53d6dd0420ec : SUCCESS in 5m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2213235b0d95496cb78f775bc1e2b85e : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bbc2e023e0241d7ad22438e17a57225 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44fe13d7c8ad4f189fada769c0ee8e8e : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/709ea3583c2445959b24dfc88d1e95bd : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c12750b64a944429d8a26a25f306920 : SUCCESS in 7m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fc77bcc5e29f46a399ec8236c858715f : SUCCESS in 27m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/20ebf78ced0645a6875bdd162a854d1c : FAILURE in 5m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2591434620951c1031a07a5fa02e385dc087e98b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-21 19:23:51.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"32ffc9ee87dac90de6ce473af87c35a5ae6a73b5","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-02-21 22:17:04.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"af90da3eaaacc40d21411f0a8575e556ada60ec5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-02-21 23:42:31.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e425bc0417e2dd5242bfd4a034c0cb49bd0b7b4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 23:43:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2299a73c66dc3cb633b700a21f0cb1fa543c7abc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 00:41:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4b150e91875d4bfb8d273df310fbb74b : SUCCESS in 39m 21s\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/67caf08bac994908b38a134cf7778c77 : SUCCESS in 33m 18s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/6502e96cc85249288504dc42fad123c9 : SUCCESS in 37m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2931bf109e5748789714ca7b21567fce : SUCCESS in 35m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/07a2e6e92dcc45dc9c38001210ee050d : SUCCESS in 57m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/51e4f0927a8849e594e8bb54a928352f : SUCCESS in 38m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/21e708f0ec61456f8ed427c9bea76869 : SUCCESS in 40m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36a161fe7114461fb0e95e69916586e7 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c13ce43af2d34af4a2b3633ad3057df7 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa6ba59048804f3fb5f2870340597aea : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39413024a1454fb4a7a97148d9b71112 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f193fdeb57d04feb8b84a70d8d94eeef : SUCCESS in 7m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8c945c3a5f7046c8a25a734c42a0cd35 : SUCCESS in 25m 49s","accounts_in_message":[],"_revision_number":4},{"id":"da112e87e40f2822d1ac01ab26374455516dbb83","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 00:41:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b1a6700ca3e9c02ea71d21b832b4c4cf93bd8392","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 00:42:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dd6108206b04091b625c0c05fda0e80 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"907a029625a7f95369ca5a2f5173efefdb04d28f","revisions":{"1ae0fc62851a15dc54b14e8b5798e3cc1c2d374d":{"kind":"REWORK","_number":1,"created":"2022-02-15 02:18:11.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/50/829150/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/50/829150/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/829150/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/829150/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/829150/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/50/829150/1"}}},"commit":{"parents":[{"commit":"057329deb123d981405a61043510998ad9f795d1","subject":"Merge \"Integrate OSprofiler and Designate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/057329deb123d981405a61043510998ad9f795d1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-02-15 02:17:12.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-02-15 02:17:12.000000000","tz":-480},"subject":"Add proper quota error messages","message":"Add proper quota error messages\n\nOverQuota never actually contained a description\nof the error. This patch adds the actual quota\nhit and fixes a minor bug with QuotaResourceUnknown.\n\nChange-Id: Ibd01f6ad3046ca29eec02032e9115183f89a6a4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ae0fc62851a15dc54b14e8b5798e3cc1c2d374d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ae0fc62851a15dc54b14e8b5798e3cc1c2d374d"}]},"branch":"refs/heads/master"},"da8227c45c5ee755ef492cf0f19e3cd892c1fa4b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-02-15 02:21:03.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/50/829150/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/50/829150/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/829150/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/829150/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/829150/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/50/829150/2"}}},"commit":{"parents":[{"commit":"057329deb123d981405a61043510998ad9f795d1","subject":"Merge \"Integrate OSprofiler and Designate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/057329deb123d981405a61043510998ad9f795d1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-02-15 02:17:12.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-02-15 02:18:41.000000000","tz":-480},"subject":"Add proper quota error messages","message":"Add proper quota error messages\n\nThis patch adds a message to OverQuota containing information\non the actual quota reached and fixes a minor visual bug\nwith QuotaResourceUnknown.\n\nChange-Id: Ibd01f6ad3046ca29eec02032e9115183f89a6a4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/da8227c45c5ee755ef492cf0f19e3cd892c1fa4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/da8227c45c5ee755ef492cf0f19e3cd892c1fa4b"}]},"branch":"refs/heads/master"},"dc3b8cf87f7a041a3d0f56b2480cd913d11f833e":{"kind":"REWORK","_number":3,"created":"2022-02-15 02:45:17.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/50/829150/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/50/829150/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/829150/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/829150/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/829150/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/50/829150/3"}}},"commit":{"parents":[{"commit":"057329deb123d981405a61043510998ad9f795d1","subject":"Merge \"Integrate OSprofiler and Designate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/057329deb123d981405a61043510998ad9f795d1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-02-15 02:17:12.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-02-15 02:45:11.000000000","tz":-480},"subject":"Add proper quota error messages","message":"Add proper quota error messages\n\nThis patch adds a message to OverQuota containing information\non the actual quota reached and fixes a minor visual bug\nwith QuotaResourceUnknown.\n\nChange-Id: Ibd01f6ad3046ca29eec02032e9115183f89a6a4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc3b8cf87f7a041a3d0f56b2480cd913d11f833e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc3b8cf87f7a041a3d0f56b2480cd913d11f833e"}]},"branch":"refs/heads/master"},"907a029625a7f95369ca5a2f5173efefdb04d28f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-02-18 17:49:14.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/50/829150/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/50/829150/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/829150/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/829150/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/829150/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/50/829150/4"}}},"commit":{"parents":[{"commit":"7e543c2d22516254cbaab111e9dad87a408ab80d","subject":"Fixed incorrect message when zone import failed due to quota","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7e543c2d22516254cbaab111e9dad87a408ab80d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-02-15 02:17:12.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-02-18 17:49:14.000000000","tz":0},"subject":"Add proper quota error messages","message":"Add proper quota error messages\n\nThis patch adds a message to OverQuota containing information\non the actual quota reached and fixes a minor visual bug\nwith QuotaResourceUnknown.\n\nChange-Id: Ibd01f6ad3046ca29eec02032e9115183f89a6a4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/907a029625a7f95369ca5a2f5173efefdb04d28f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/907a029625a7f95369ca5a2f5173efefdb04d28f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
