)]}'
{"id":"openstack%2Fdesignate~830529","triplet_id":"openstack%2Fdesignate~stable%2Fwallaby~I4e3e51b32ab01efc4202c297708eff5a2e2b4985","project":"openstack/designate","branch":"stable/wallaby","topic":"bug/1595265","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-03-24 06:56:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4e3e51b32ab01efc4202c297708eff5a2e2b4985","subject":"Allow TXT record over 255 characters if split","status":"MERGED","created":"2022-02-22 22:46:03.000000000","updated":"2022-03-24 06:58:02.000000000","submitted":"2022-03-24 06:56:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830529-bug/1595265","meta_rev_id":"2bd7e2c3fe716ad5d399a2f2f834d69bca168733","_number":830529,"virtual_id_number":830529,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-24 06:56:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-02-24 22:41:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-24 02:46:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-24 02:46:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"CC":[{"_account_id":30678,"name":"Emanuel Andrecut","email":"emanuel.andrecut@fleio.com","username":"andrecutm"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 22:46:03.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":30678,"name":"Emanuel Andrecut","email":"emanuel.andrecut@fleio.com","username":"andrecutm"},"state":"CC"},{"updated":"2022-02-23 00:00:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-24 22:41:13.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2022-03-24 02:46:57.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"f17d8c7041c85d9bc407111d25241cbed36a54fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-22 22:46:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22c2439a1f43e7a69352c9e86baf566e96638857","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 00:00:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d6544b4998194e76963b147b57f87da0 : SUCCESS in 45m 44s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/95d805d4bd1446f09bd1c0da9b3588eb : SUCCESS in 1h 01m 28s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1447d022f02c484b93d24d7dd90d4756 : SUCCESS in 57m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/74a6b417307a47a6b92854c766025da1 : SUCCESS in 43m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/91701c57ec4b4b2c9e919c97c9e384d3 : SUCCESS in 1h 07m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cc472e00064948e19fac14d61d06f9b7 : SUCCESS in 44m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1ec19e4a14fd499992ad26bbd4432670 : SUCCESS in 48m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b507518f7e947fd998bca136878bdce : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7af02b9f87ca43b0b9efaad685e1c3af : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0f094756e8349178e880db750bd614d : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fc86bc079574d4f81dbe1c5cfc16d11 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45ab9d1fcfa54042b304fe5e2d0937cb : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a1e960f8d6849f48be2ba572e39fd05 : SUCCESS in 7m 59s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/6e59777ae75b40bf91e047b27bef6d64 : SUCCESS in 26m 25s","accounts_in_message":[],"_revision_number":1},{"id":"14d1399d02ccce58f00b63548cc87afcc54c9c62","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-02-24 22:41:13.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dffdc7291e34c1f5141426a5c12004f07e19ff51","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-03-24 02:46:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"312d7ba0cb321781229a3dd4a5e04565b7f43cb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 02:47:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70c395c072fd40ae3da43f43ff1004bcfdd0398c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 03:57:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/107681e752ee47aebd46647293b0431b : SUCCESS in 48m 44s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/0025d6a5984a41cdb87a46a50f5aa498 : POST_FAILURE in 10m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/165245aedaf54aa29212caae4a4061ca : SUCCESS in 42m 04s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/267b1af092f4469da725b02c41bb2b7f : SUCCESS in 1h 08m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2aeebf88781748499f5e89fca769eb87 : SUCCESS in 39m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/317772828d7348929b51b5ee7de8352d : SUCCESS in 43m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ead703a421e44a3b6b16219ca601f8e : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b8ee088a94c4e6aa29d1ffe20946ae8 : SUCCESS in 8m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6bbdacd2f704cf29c1c364674b36e77 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00de121560f9429a9c43ad20769a3bef : SUCCESS in 10m 14s","accounts_in_message":[],"_revision_number":1},{"id":"5abc066115176fc266a2e6cc841b3669b59e1269","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-03-24 05:09:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"efc0c705b91c0ddaa97e41090eda15a2dc33d43e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 06:09:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e4a4515291074665a74b8a87f54e2a98 : SUCCESS in 35m 23s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/3e0639c135e84e4195ae269c102e2773 : SUCCESS in 43m 13s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/a707127386e648a4b6517749d53ebdcc : SUCCESS in 37m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/de4d1343c9394c1d8f8e99e85a29af9b : SUCCESS in 37m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/64dbd16ab06d400cacedb346a85cb06c : SUCCESS in 57m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/61f8d9c928084ea28ef72ee1bc539fa2 : SUCCESS in 36m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cf2c00aa369f47cf98b05bd61adf1b81 : SUCCESS in 40m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2c99df6c1f54ec6a00f8979c85bf5b3 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d57733728cae4f7ea9f9178c0a36c1ba : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7665983c01547ad92c4b09ceff2510d : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9aa59dbd080b43719c4b259bd3ea5ea1 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2ef3f8d2fef4baead54d034f996b502 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8790d79b14643ccb9c8eee1ab2611b8 : SUCCESS in 9m 06s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/0cfd07f4f6a349c39aed11d33753f1bd : SUCCESS in 37m 34s","accounts_in_message":[],"_revision_number":1},{"id":"0810c3182d6d2d4cccf2b3b959d9ad709022598b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 06:09:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4900c8decfeda422d3c50c02c9e85bd010e84fa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 06:56:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c091bc433aab479fa245fe04340927b7 : SUCCESS in 32m 12s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8899bdbb4ca341829796247dc2ad50de : SUCCESS in 34m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/be5fb4ee2f98489dba7a6c8f1b4c4e27 : SUCCESS in 30m 02s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d8cf1081697b4f3287c79b3d0a84fd11 : SUCCESS in 46m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b454bf6801f945f88883ea4e92703a16 : SUCCESS in 32m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5d3bfca2401349bd836358a9e784bd2d : SUCCESS in 38m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eab50e1cf00145a6a20f464a3e74f227 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/201e96905d524c6984028312be258338 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2416eefc1004752a16c2a5f96e2b2f0 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62a1130e153b40acace1a82feb11bd39 : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":1},{"id":"00b0ebb5df29c6c9e939ba8350e852a1001ef8eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 06:56:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2bd7e2c3fe716ad5d399a2f2f834d69bca168733","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 06:58:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dfc6362408e4b2b8dee0f0636aa4a81 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0323c4244929aaf199ecbb7c8d32138764029cc2","revisions":{"0323c4244929aaf199ecbb7c8d32138764029cc2":{"kind":"REWORK","_number":1,"created":"2022-02-22 22:46:03.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/29/830529/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/29/830529/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/830529/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/830529/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/830529/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/29/830529/1"}}},"commit":{"parents":[{"commit":"63d6672befb2225d844388fe43f857b0f3dbb8b6","subject":"Merge \"Fix support for scoped tokens and default roles\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/63d6672befb2225d844388fe43f857b0f3dbb8b6"}]}],"author":{"name":"Emanuel Andrecut","email":"emanuel.andrecut@fleio.com","date":"2021-10-20 08:48:18.000000000","tz":180},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-02-22 22:46:01.000000000","tz":0},"subject":"Allow TXT record over 255 characters if split","message":"Allow TXT record over 255 characters if split\n\nAs defined in RFC1035 section 3.3.14 TXT-DATA\ncan be one or more \u003ccharacter-strings\u003es.\nBefore this commit Designate threw errors\nwhen saving TXT records that are split\ninto multiple strings because validations on\nthe field did not consider allowing this\nin a single DNS record as per RFC7208\nsection 3.3.\n\nThis patch allows longer TXT record data\nbut only if it is split according to RFC\ndefinitions mentioned above. If data\nis made of more \u003ccharacter-string\u003es,\neach one is individually validated with\nthe same validations as if the data was\nnot split.\n\nCloses-Bug: 1595265\nChange-Id: I4e3e51b32ab01efc4202c297708eff5a2e2b4985\nSigned-off-by: Emanuel Andrecut (emanuel.andrecut@fleio.com)\n(cherry picked from commit 03a5d5d74eb5de3f37b2e7ac5f28ae139ae436a5)\n(cherry picked from commit 5e9cca1b02a8e115e0f4cb471b180427db4c3993)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0323c4244929aaf199ecbb7c8d32138764029cc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0323c4244929aaf199ecbb7c8d32138764029cc2"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
