)]}'
{"id":"openstack%2Fdesignate~834975","triplet_id":"openstack%2Fdesignate~master~I55ca76ef7c2cbeb5fdae1aed1dcbe58b7acddc34","project":"openstack/designate","branch":"master","topic":"bug/1966128","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-04-01 15:27:00.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"I55ca76ef7c2cbeb5fdae1aed1dcbe58b7acddc34","subject":"Fix set-quotas for non-project scoped tokens","status":"MERGED","created":"2022-03-23 21:06:48.000000000","updated":"2022-04-01 16:43:06.000000000","submitted":"2022-04-01 16:41:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834975-bug/1966128","meta_rev_id":"4d878941e5e2e91a44a6d6206de21402d62e6cb2","_number":834975,"virtual_id_number":834975,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-01 16:41:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-03-23 22:01:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2022-04-01 15:27:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-23 21:49:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2022-04-01 15:27:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-23 21:49:49.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-03-23 22:01:36.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"CC"},{"updated":"2022-03-23 22:01:53.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2022-03-23 22:08:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf4f81dbfaa7732355cbe6b317d616c5cf218412","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-03-23 21:06:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58a6cba0efa074ba5d7bcbe53f600f4b8cf4ca5a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-03-23 21:49:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3a49787a23f5ccce665f30a786a939ab93c7dc85","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-03-23 22:01:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79a0806ee9e0e613d2f5e080ac694daa5712b8b3","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-03-23 22:01:53.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4cb56e43ecc27df20c33cc9285030856bf8dac90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 22:08:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b8bea4dbddee40de98f7dd564ab397a0 : SUCCESS in 33m 31s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/d04242c4b37a4e7ebe5e19d0bb7834d0 : SUCCESS in 56m 30s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0192b14359c24f2797dfca8d42423022 : SUCCESS in 35m 38s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1bc805f65c484362831320455a9223ed : SUCCESS in 40m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/947a26a33705486591324c6f1b690190 : SUCCESS in 41m 46s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ce02967fb1a643969ae858a4e99d575e : SUCCESS in 58m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/36a3b1b92a7141a8aa5f724c283dd045 : SUCCESS in 41m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d77aac28e17c443da77edaba6ca6e3c2 : SUCCESS in 41m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c146f9467b441d9a9842a15bb9f8c19 : SUCCESS in 7m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f23f498472645de980ff90d7252a4ff : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ad82560aefd43f1a271f9501f69b200 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2349edbf6d7424ba9157e7bd0f01b80 : SUCCESS in 7m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03735d23e4194c5d8fa49c4690866b6f : SUCCESS in 5m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d1541f8102d4066b0d9c60ef99a32fd : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f8d04bce1b54e8aafa1c4ce80c01400 : SUCCESS in 8m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/308d0795234f4faf90b179dedf72d7be : SUCCESS in 5m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/40ba811063ff4ba281d23d0287ead279 : SUCCESS in 31m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7f3ff72dd8534b9c88ca6661b6c67489 : SUCCESS in 6m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9420009cbc7e157bea50632ec80d6137a9718271","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-01 15:27:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ee70a7180510bf5289d8888e4e004f3c8a25a224","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 15:28:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ffedc62823083579ef6002769021cea2451ba983","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 16:41:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/52a4f2f48d274626a7e39cc307db7775 : SUCCESS in 44m 33s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/dbc64a9ab2114f0ab855a017d17d8c74 : SUCCESS in 41m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b494b824f0b9457b9ad31956d485f4b6 : SUCCESS in 42m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0ca5b9b1279046deb6bb63b8d596b23f : SUCCESS in 1h 05m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4831ebb689fe49e7931df11033c7a91b : SUCCESS in 41m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/82e38ac04a4749799675b33bf8daef0d : SUCCESS in 42m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/429e838e35b64e61a1036ac03ad3ab02 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/608a5ee25836403b95fc24a1b4a8f987 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccd47427a2804a0d91c35e3fe4e9f59d : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25f0b128c0d042d09823e5081525c4e7 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/668d2049e1e04c52b231118207a7ffe3 : SUCCESS in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9129c0e1ef8140a094bb9a9ace256164 : SUCCESS in 4m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/84158610492740dcb818fa2cf1376623 : SUCCESS in 31m 04s","accounts_in_message":[],"_revision_number":1},{"id":"af19cd7551e810906528ec353deed522842600a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 16:41:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4d878941e5e2e91a44a6d6206de21402d62e6cb2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 16:43:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8292baef63d3433aa4084dedec1c767e : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a43dc79567848be952d0ee09604a95b : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"158e017be489eea5402ef258ea260183067bf1ca","revisions":{"158e017be489eea5402ef258ea260183067bf1ca":{"kind":"REWORK","_number":1,"created":"2022-03-23 21:06:48.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/75/834975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/75/834975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/834975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/834975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/834975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/75/834975/1"}}},"commit":{"parents":[{"commit":"c998ff690058ea7319e6bc92b3f841a77dee8d6e","subject":"Remove unused downgrade functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c998ff690058ea7319e6bc92b3f841a77dee8d6e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-03-23 20:52:35.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-03-23 21:06:46.000000000","tz":0},"subject":"Fix set-quotas for non-project scoped tokens","message":"Fix set-quotas for non-project scoped tokens\n\nPreviously, if set-quotas was called with a non-project scoped token and the all-projects flag was not set, the quotas would be updated but the result returned\nwould always be the default quota values.\nThis patch changes the API to require the all-projects flag when set-quota is called and the token is not project scoped.\n\nCloses-Bug: #1966128\nChange-Id: I55ca76ef7c2cbeb5fdae1aed1dcbe58b7acddc34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/158e017be489eea5402ef258ea260183067bf1ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/158e017be489eea5402ef258ea260183067bf1ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
