)]}'
{"id":"openstack%2Fdesignate~835317","triplet_id":"openstack%2Fdesignate~master~Ic7b8fee4d4702b0632774d32542b23d7d2a8c253","project":"openstack/designate","branch":"master","topic":"bug/1966517","attention_set":{},"removed_from_attention_set":{"34181":{"account":{"_account_id":34181,"name":"Manish Honap","email":"mmhonap@gmail.com","username":"mhonap"},"last_update":"2022-11-23 01:23:51.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2022-11-23 01:23:51.000000000","reason":"Change was submitted"},"20104":{"account":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"last_update":"2022-11-01 15:22:07.000000000","reason":"\u003cGERRIT_ACCOUNT_20104\u003e replied on the change","reason_account":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-10-20 15:48:04.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-10-20 23:40:08.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Ic7b8fee4d4702b0632774d32542b23d7d2a8c253","subject":"Add option to force delete zone-files in delete API.","status":"MERGED","created":"2022-03-25 20:15:23.000000000","updated":"2022-11-23 01:24:55.000000000","submitted":"2022-11-23 01:23:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":57,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"835317-bug/1966517","meta_rev_id":"19204f4381ca320f777af6da43889cdc58ffcc41","_number":835317,"virtual_id_number":835317,"owner":{"_account_id":34181,"name":"Manish Honap","email":"mmhonap@gmail.com","username":"mhonap"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-23 01:23:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2022-10-21 20:27:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2022-10-24 15:57:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2022-11-01 15:22:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-23 00:25:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-10-20 23:40:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"value":1,"date":"2022-10-22 06:19:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-23 00:25:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-20 15:48:04.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-25 20:21:00.000000000","updated_by":{"_account_id":34181,"name":"Manish Honap","email":"mmhonap@gmail.com","username":"mhonap"},"reviewer":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"state":"REVIEWER"},{"updated":"2022-03-25 20:21:00.000000000","updated_by":{"_account_id":34181,"name":"Manish Honap","email":"mmhonap@gmail.com","username":"mhonap"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2022-03-25 21:19:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-27 20:01:04.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2022-04-09 20:52:32.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-05-31 12:20:04.000000000","updated_by":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"reviewer":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"state":"REMOVED"},{"updated":"2022-09-21 09:33:57.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"state":"REVIEWER"},{"updated":"2022-10-20 23:40:08.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-10-21 20:27:09.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2022-10-22 06:19:48.000000000","updated_by":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"reviewer":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"state":"REVIEWER"},{"updated":"2022-10-24 15:57:17.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"da08661102c72c2e3418c5843017068375d6228a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34181,"name":"Manish Honap","email":"mmhonap@gmail.com","username":"mhonap"},"date":"2022-03-25 20:15:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ec0ba39163bedc1b74eef0253d34c89af0c4ebe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 21:19:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1a56249ef1d14d02b7ab9f670a2bf66b : SUCCESS in 5m 10s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7a1a874de3e44805abd203be80fb0313 : FAILURE in 46m 05s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/23b5ff25d31d4f5dab1c41e8afe867e3 : FAILURE in 52m 08s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/cab9513ae349436ab477751a3e505d9b : FAILURE in 38m 46s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ebf8c4f1977c468a9fd14f93052b1933 : FAILURE in 41m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c3118171410f4f0eb2392f952297d6fb : FAILURE in 44m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2333ba6712394c1796e32e599c56fdac : FAILURE in 59m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8298951eff0b4383800cd3e686ddd0d0 : FAILURE in 43m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5f769eb4abb94df785e31e2de31ef2d9 : FAILURE in 40m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3c841bc9aa543e9abab1ab8c9957de1 : SUCCESS in 7m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/231a5a4ce81b4d7da05394e70d42cf55 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efebafbf2a514fef8cfe2f43715a337d : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abc68160c9a84959bda9c2af063f31df : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4f4e246e115434b8fc98a7bae6b90ad : SUCCESS in 6m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e52c1b1284d447a5a3e8b04f2cc794bd : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41fa6d102e024cefbfe5f214f2735372 : SUCCESS in 9m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39f710bb77ab47a195847355c56ed87d : SUCCESS in 5m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/89161343bcef4513b6bb589cc98bb2c6 : FAILURE in 38m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ebc39c9530b34be6a992efddc85ae41a : SUCCESS in 6m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bcd8f79c357ea0d93e5bca9bee1ef9f863f7d292","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34181,"name":"Manish Honap","email":"mmhonap@gmail.com","username":"mhonap"},"date":"2022-03-26 13:49:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e4401e98bcb2d33151d117cb169ea37c03c8b993","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-26 15:02:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/30d8bf0a1cab44a0b75f9fcf7a16c5fa : SUCCESS in 5m 11s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/faff1456c4d04f5d8f94c9841b6b7a80 : SUCCESS in 51m 52s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/8ddcf4935acd4a6e9856f702d1cb72a4 : SUCCESS in 1h 08m 38s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/516b3aebf33045458764e6143eb1c1fd : SUCCESS in 33m 12s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/c1467329c75542f7918d0a465683098b : SUCCESS in 48m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/321cef0da2554f94a890fcf34775a7b7 : SUCCESS in 41m 53s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6ad7a70e6e794e279edae4f0efcdcbb1 : SUCCESS in 56m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f8c42a190de544bbb9b94d2193c58c97 : SUCCESS in 43m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e9de7b78f0af45d991a79a9c6cb581e7 : SUCCESS in 38m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42e87ae9c42a4349a547dbd0c26049ee : SUCCESS in 7m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7736f74445ce4a2b8b245ddaca8f5ea7 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3bc4f2e91534916bb77a00cda066cfa : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5566f64e52714ed695daccf72282f90b : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d703c62aa1c8421ebef3052182c3fdd5 : SUCCESS in 6m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98ea55430a1c42cebfa49fb6c87fb0a1 : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b86d946195544939ca2cea14d708785 : SUCCESS in 10m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/542b8ff1f9f246c79595b98398f6af8c : SUCCESS in 5m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cb183993b3d34965a9a4934d8456b9e5 : SUCCESS in 41m 27s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/01f0c7e544494593b1d35d85b1da13a6 : SUCCESS in 7m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c370c037bb56dbfe8a68209e64f3af0f2139d1bc","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-03-27 20:01:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"65ea75a023a6cb9ff58250d502f00ebe54def25a","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2022-03-28 09:38:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d2790ac52fc27c6e74298b004b8814e9d49287c9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-03-28 20:56:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"61635bd2cbb12a134037ab9bf982882ebf2669e9","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2022-03-29 16:44:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e7674b57e1ed6707d9e5bf779e049cb24179358a","author":{"_account_id":34181,"name":"Manish Honap","email":"mmhonap@gmail.com","username":"mhonap"},"date":"2022-04-07 19:16:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4f8658d5c465ba3807750f6978c6261c9b762740","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-04-08 12:45:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1ddcc03f99e899dc084e699993fe7bf9780e0e2e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-09 20:52:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2f0ea20f20779edd9a0bc931bd917674d94565c5","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-04-30 12:41:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9fd7dd417e92dacd1285d6262fbdc8a53a39a1be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-04-30 13:12:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aad026392df640045ffd08050fe70f3d2da13e86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-04-30 13:16:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ce7c85071f2c625cb59cc3f60d024f12b45a0f2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-30 14:03:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c6eeb2d8cbc9438fa495f97f6f649de6 : SUCCESS in 3m 23s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f4e01a3bd35540c095dea8732ebbc2aa : SUCCESS in 34m 01s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/255bf88a45384a84983eeb0bd2149be4 : FAILURE in 12m 54s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/913fe4755cbb4556b8786022e03e39f0 : SUCCESS in 34m 08s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/da40f0014e5b456e8bcbee31fd11bc3e : SUCCESS in 34m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/24811bb536694ae2824f39d1e848a144 : SUCCESS in 31m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c7e0de915b424bdb8a51b1bc444ddb64 : SUCCESS in 46m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/db24cb2aabdb43aa98f1da7247f12ccd : SUCCESS in 32m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/390c49ba7e054e849e1aa37d1e790aac : SUCCESS in 34m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fa14b0252b94bcabb02cabfdc7f92de : SUCCESS in 5m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70f1f528ef184531a279e1b764245eae : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1a1207e85b04b3c8b6e507079d7d9cc : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/259fe2be35f945868ab452fd46cf472e : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5700a0adcdc441659dde1921910b8843 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7bd8dd6e21840e6b4d884d859824d5a : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fbe2516ec424ffc8de9a3da9af196f6 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aeff018f4bce44908f8773963d5bf1eb : SUCCESS in 4m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6870a647e05c4c24a105ba69ac823509 : SUCCESS in 24m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8b2bbb79f8fa45e3a9ad215d5dd12cd2 : SUCCESS in 4m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5dd9debbc7698b9b237f73755f6c9f93abcfd09d","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-05-07 10:52:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ddfb15d4e8184ed5f6c318d98d8672f19e5cb6cd","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2022-05-31 12:20:04.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e3dbf632146db960c83a68ecfa433cb27effc598","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2022-05-31 12:20:04.000000000","message":"Removed reviewer Dmitrii Galkin with the following votes:\n\n* Code-Review+1 by Dmitrii Galkin \u003cgalkindmitrii@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"e772a330d8250200308120cc098bb9579a69685a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34181,"name":"Manish Honap","email":"mmhonap@gmail.com","username":"mhonap"},"date":"2022-06-14 17:50:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3811502276c1607dbe4d6c088342260acdd82b46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 19:29:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e05b0dbe0fa8467187a0f97ec41df076 : SUCCESS in 4m 23s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/10fe57182e894373ada9696f19d1ba54 : SUCCESS in 38m 34s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7685c2d57b234bf78c4a1f97efd64758 : SUCCESS in 45m 54s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b8de67ed80d466fbb463426356d42cd : SUCCESS in 1h 08m 17s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/05635f68f4334a9b94d8cf4262955798 : SUCCESS in 1h 08m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b61059addb234464888679f88a37350e : FAILURE in 1h 08m 34s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d173bece524745849df0700748916bb1 : SUCCESS in 55m 50s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f6d0906f3b074f02888c7bd0b5cfeec3 : SUCCESS in 1h 34m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/17c1f199ba7e44f394fc356fefa294fe : FAILURE in 43m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/263b40954ea34a7e9501cca89e0d0875 : SUCCESS in 39m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d8eec74480a43c28443189325ccf46d : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10d0d9e5a3ee4cff9cdf673fee389b15 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1794a27969b94a58b3c3c508bdb10e3e : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f2bcbfa06754205963497450a1dc48c : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1c8000e507e435e99fbd362f1c23471 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f414e7508e1b440a907c575dca2abec6 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4d50e780bbe420b90cd13e5c2dfd51f : SUCCESS in 5m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6ad3f9bcb5704fb4b928acc8b1a9dcf5 : SUCCESS in 29m 35s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8313ba50bb134ead9edbc120ee84cb97 : SUCCESS in 6m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b97c00f39495aae12b8b624693fbd61166e981a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34181,"name":"Manish Honap","email":"mmhonap@gmail.com","username":"mhonap"},"date":"2022-06-27 17:01:09.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"cef3757eec4fe3de44ed13ea30e5367d24964b62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 18:07:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/294eaf1599f145e0a7271641017d09d1 : SUCCESS in 4m 03s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3455cba1c3674be48b1fce1b475677b9 : SUCCESS in 51m 53s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e9840752cd5c451da3452d7eb4631e4c : SUCCESS in 52m 07s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0422c3240e64928b4d54558e7d85dd1 : SUCCESS in 44m 30s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ace83fce0c154e89a25004137dadc023 : SUCCESS in 43m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/30fb5d800d984f6fa2f3b73c1aacb672 : SUCCESS in 40m 24s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e808d703e0904684ac0c4fa3acf957c8 : SUCCESS in 1h 03m 32s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c9e8f7073a6b4ca2931c2a76ed829992 : SUCCESS in 59m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0086c93693054030a2af9f645e1e0515 : SUCCESS in 41m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fc1dd92ee3154bf2abf31b88363030de : SUCCESS in 43m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f6a070a6506410e9b44beaf55bbb09e : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c44bb6063b34dada715373d36f4c801 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0aafe3b6b5964b7d9fd657c855ba3e4a : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/874c1c3b3af14b91a574beaf261b5fcb : SUCCESS in 6m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3211ef38d23c4dfda926fa87294b67ec : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a1bf13082c54610b784bb2666efc95c : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3e6e42298c24167a8dcf889d002bd7b : SUCCESS in 8m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5f633b8c9c3e4b47b6dca687b9f3ab49 : SUCCESS in 30m 47s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d0e47088243d48efa7f3ffea53bd70e9 : SUCCESS in 5m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"96a1e3b86d296ef4ee993f074edc574019eda54a","author":{"_account_id":34181,"name":"Manish Honap","email":"mmhonap@gmail.com","username":"mhonap"},"date":"2022-07-19 18:45:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1e0a621daf46435c39882618859cf9ca8e24503f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-08-10 09:28:51.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d23c90112aaac1541b1a0a388938d45c91698f85","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-08-10 09:31:56.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e8ff25997d6637216c4310009dfa6687dd5ebf00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 10:35:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cf339471a084236aab8e8c95bc0459b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/511e667f5e5c4973bc161cd8d5a39eb9 : SUCCESS in 4m 51s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c2c0787d0d8b4e64a50f4f83ff75059f : SUCCESS in 47m 55s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7b787e0f493e45fba39318cbb3e27a72 : FAILURE in 17m 55s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b7cac4937ff40e68601e73a3b8b9ca2 : FAILURE in 14m 56s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/88b05705b7f64bb98736e3a1bfbef7c4 : SUCCESS in 45m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f14683bb41594211aed37fde741c6d08 : SUCCESS in 52m 47s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0bbb1481b17647fa8e5b46fa5cfcd0c7 : SUCCESS in 1h 00m 30s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7b35dd9f57af47788b01165487b04892 : SUCCESS in 1h 04m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/804b8693ecfc4863817cf870a8407cc3 : SUCCESS in 54m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e88c2facdcb04337a44f021572c6b972 : SUCCESS in 45m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c20cc174234416894ca6989fdedbe97 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98c7ef88f950430face5453ee5476c27 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1284bf9f1fea4200a78720958919cf91 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cceeecd71e148239e4e3c11bc0b8899 : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d616952fc562408d8046d9509701c8bb : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/979b0789d9e6461fbbc8e2a3577b6aeb : SUCCESS in 9m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f38e889729f4eaeaa15a09d24a3896a : SUCCESS in 6m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c5ddf5272fb04337965c6cd885d08fa8 : SUCCESS in 41m 27s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a46eb946020247d8a5ac8751aa6ce40b : SUCCESS in 4m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5f8a508f34dfc451efb1b08500b5f29dfc5349af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-03 08:30:51.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4dd0c39af91fda6cfbd7fe472fe3de3098ffbbb7","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-03 08:31:00.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8d58337982ddc6bdbd1c9668ce39195549b22129","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 09:29:21.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33714a6607fa49d690eb92ca9ea3e4bb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/28d9e8b82ab84b98926271a79a3b5afa : SUCCESS in 5m 59s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/21f73352e5034ffdb8df326d51db5483 : SUCCESS in 49m 02s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7a4de98e7fc64010b11edca4a78e2827 : SUCCESS in 57m 37s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b68787330231498f9ec17981db49f10c : SUCCESS in 44m 06s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ea2ea5d084e540be92adc26bacbc0310 : SUCCESS in 33m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/33377356d81c4cd1b1d92f65d6acb95a : SUCCESS in 49m 10s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/613f42235e104a729797ea280b654124 : SUCCESS in 48m 38s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6c70e6f4a536414383dbf44d01b8d2f6 : SUCCESS in 54m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6b36bd5337e8490c8fec958bb7d06687 : SUCCESS in 49m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/204bba6097084d649f14be842f300481 : SUCCESS in 44m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4e28f305d84416cb24e597a2d6c2c84 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68b3ee03690e47bf8f38039aa8b43787 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efce290696264780b0c6557193adf602 : SUCCESS in 7m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/515e9df1ce5845ddaf7dc72e0c2ebc57 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4deaf7d845514f3cb31a12d882b9489c : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f8973b878374960aa0a42f8e2b8e0be : SUCCESS in 4m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/21bac92009f14342aafecf6e5c91ebe1 : SUCCESS in 36m 47s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9157b0aba7fb4858b1b9ac64d5dab05c : SUCCESS in 6m 49s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"31002dd79266d77c3267c420f7c683de667f1d2e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-10-03 22:08:48.000000000","message":"Patch Set 8: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":8},{"id":"9d9c65b927dc8deff5fc3698e59f4a03b1f09066","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-03 23:28:43.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"026806105769599a6cc2f71edd906a06659c3d0e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-10-04 01:24:23.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2a772105007a475bae5022cbbf8a2f5448940cd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-06 10:29:34.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ae7ca897beaca45332cc3920661a92315ca49d1b","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-06 10:31:08.000000000","message":"Patch Set 8:\n\n(11 comments)","accounts_in_message":[],"_revision_number":8},{"id":"b699b075f2082a44b76e4a43af3455f765348dab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 12:03:13.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/35396b24674a48e18a8ecde99052fed2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/37754cdf11d340f09e2bd2cbfafdcb20 : SUCCESS in 4m 35s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dfa34a2cb4554118a30b0f6ea783842c : SUCCESS in 1h 01m 15s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/cb1a0440ba124c8586ff0cc3d3e1adce : SUCCESS in 1h 06m 38s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a112c322b294452bf4f42c3257efe09 : SUCCESS in 57m 46s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/9accb64e01a943e497a12f46274c6310 : SUCCESS in 42m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1fff889cc4f34522852c5ebb34b07809 : TIMED_OUT in 1h 15m 13s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/94fe02caa16c4c6aa102a6077ff9dfdd : SUCCESS in 1h 00m 13s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ee0b0a31854146d98874cb785d8ff38b : FAILURE in 1h 32m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/56b4749fec7d4a8481c199d42897b932 : FAILURE in 1h 18m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/11687a7042ae4c1dbb8d85a07840bbe4 : FAILURE in 53m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38d108862ba343d8aba48f975b23d991 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0268566d93f84cc3963fc797292e11bd : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/727dcb761d094af095e4a786c885e675 : SUCCESS in 8m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f345c49b66cb4772b16e12246f2de849 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d35aa9dbfcf34e2d96b6fa4136c0efc4 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a23869feccb048e5a834b59f14d45b27 : SUCCESS in 4m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b4eb5a37edec4c6480309ad523a042ab : SUCCESS in 40m 40s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/678fb411d37b48a6b4bdabb2efd6ad7e : FAILURE in 4m 12s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"744a6a398bd738f29d309107de4cccdfffa83dbd","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-09 13:12:26.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ab4572e2a81ca0d163044dfa15e3a6de8ffcbaca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 14:36:20.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d35eadac91149d2b2d1d926c3209c1a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4d2b9d17822b4360881035450ab78b5d : SUCCESS in 3m 09s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bb2fa365b94f44f682301c05ca823a34 : SUCCESS in 33m 01s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4fcac455acd04d8cbe1d2e2ca878b903 : SUCCESS in 51m 52s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6822442d1d94fe1b1235d10753ff1ff : SUCCESS in 36m 22s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8f98d3f375a44fb1b6c5948b4889b811 : SUCCESS in 37m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9e3f04de7b5b49eaa2e8b2f9756d73be : FAILURE in 51m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/84734fb25a93416b83761503d036d2b0 : SUCCESS in 51m 56s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f132a3f33f684689bb9c20fe390d2167 : FAILURE in 1h 22m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c5d53ecd6f6e4dbe8ab55b35c1e435a6 : FAILURE in 55m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/85fa65d7cfa74e4aa16341b84253fb66 : SUCCESS in 43m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5836c43aca6442fba12dcb4af76fb780 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/497241d1f1994551a2371b6eaae4d4e2 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3573a86d60ce40ceb52d59cecefff8e7 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c4048bd4d994c6698a2a91a31af37f2 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d662b20580d4cfab9f3881609ad6490 : SUCCESS in 7m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e09c69459ee4e14ad101b2acc08d821 : SUCCESS in 4m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d72b659d2a1748fe94ca6c0736c7a9b7 : SUCCESS in 33m 01s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2528a3aec8bb4f4f9be365e120f46d65 : SUCCESS in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4dff4c9853d5961699e36bf8c822ff9b584a2c6d","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-10 09:24:40.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ee2f9eea1370e667a7fcc8ce62b346e9dd66d976","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 10:56:18.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/55ba0a30fd584a859fe1771278961079\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5bef95aada2c407e8386632116312cd2 : SUCCESS in 5m 24s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/84dbd841a0c54f0db277cab17c802945 : FAILURE in 18m 53s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/49a85c3693ca4117876c38b5a16690aa : SUCCESS in 1h 02m 38s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd7638449ae2489383c818a3bc3a9bf6 : SUCCESS in 57m 27s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/264256c413cd44b1b71271ac1c966adb : SUCCESS in 48m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/56ade4ab68734ffeb5a7f386bf8d9745 : TIMED_OUT in 1h 13m 00s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d60df15130094831bc6a069ab675d9ad : SUCCESS in 1h 03m 50s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/caf05f11f668474582189bc7c37b7e99 : FAILURE in 1h 25m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7a6363fca0bb4eb7b7bb13f2ddb990a2 : FAILURE in 1h 15m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/96dd8dcc599041fa81a8dece7404baad : FAILURE in 18m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d63ce0a9ca8749399c6f754b5797f2b5 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f0e5b2a60e146338f5f5612964342fc : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a302923e77a4887a07e18ad2947d47a : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f5a52b08c5a491aa091e04ee8c070a0 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81f795843e5241548df8cef1630a5433 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36f402a7af804364a483c74a01fc7e98 : SUCCESS in 6m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/075cc9e15ba543ce8fef40af63c8eb68 : SUCCESS in 38m 57s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b8c8668f679a4d51b7edf6db819b62b4 : SUCCESS in 4m 32s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7fd0a87fec046492be201ad4221a93a8a9a7b1be","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-10 11:00:32.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"607d506b9eb09a9426e0cc27a0d392ac31267ea9","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-12 12:11:07.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"010dbc9dc46ac85bd86cf6d0e48670995880c7e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 13:42:33.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9af140b520b4bbcaa9ef9847f4be54c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/eaf1a6a473f84da4b17296be7d8b03b9 : SUCCESS in 3m 20s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6823c624a9e749e8bded777204d4c2df : SUCCESS in 33m 14s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/82432ea28c0045e290bea191e621ae71 : SUCCESS in 58m 14s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b02f07fdfdf430caa7b3930dbc5f3ae : SUCCESS in 40m 02s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f56d40480169435d95926d49d53d6434 : SUCCESS in 36m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4d267789dc8c49fea9696c5294679c16 : FAILURE in 57m 11s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5fcee5207f07477fbafb4fab8c83fef8 : SUCCESS in 46m 14s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2385d8514b29463798b006330f471ba0 : FAILURE in 1h 25m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b475e766c4a34a1982dbf2c1486218ca : FAILURE in 1h 01m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9202b287f5b94184b08ab1461158acf0 : SUCCESS in 36m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e330b2872ef44e958f587bd07ecfdd2b : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99c58d8e7d084bf19880dc052c2cd770 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc7439c103d74f4db63446e0039180a8 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/345120821766481aae185a95a4421539 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49e720ef62734b3585753f9719ffa05b : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e63cab7606524447ab197d6e283b53e3 : SUCCESS in 3m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/682631499f624077969a9571a9b3b66f : SUCCESS in 28m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/41d66ff4dd444e21960199d7f42ad240 : SUCCESS in 4m 25s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e9789ea0abfab6ebbb5a9301bb520c20f1830909","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-12 16:24:52.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7f1ece2fecd46cb99b692457c085b5d2e269343d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 17:35:40.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/27cb70fbeeb84caf8a6f9d072a44f416\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/27e9f18d3951469f888c78ec75f54e0e : SUCCESS in 3m 17s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0cd420d090cc4c4a83707a59075d7750 : FAILURE in 49m 02s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/02fdc296d0a644289de8de12db339714 : SUCCESS in 1h 05m 05s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/50bb563f623b4633b08729f481526428 : SUCCESS in 1h 05m 11s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/682e1380e1d146c98eb878babf324c67 : SUCCESS in 48m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0185b98d65e94824b8f3f109a054c6cd : SUCCESS in 43m 29s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ceedb274e2984679bfb83d71fa86bb8a : SUCCESS in 57m 20s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/170582bb865d46edaa68930c129176f8 : SUCCESS in 1h 04m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/79b9b9efa857420db643bbb413ab9698 : SUCCESS in 39m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ab9043de78384450b24cb5842874ef02 : SUCCESS in 46m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93df862b86e947028ac07434be6a2db9 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7057c87142a4cfa9a7b6087141a8c6e : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9019728e215c4fd994f5c6edd74d05c3 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86ec8d9d3cf94afa83c82d0bc7e9f5fb : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/304b6a06104040dfb8b612947267ff4f : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b542f3eaca1482a888c99e5bacd9184 : SUCCESS in 4m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c548e98667b4460392a53bfdb90c187d : SUCCESS in 34m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/16cb0439fcb64d758c4fedd584d66971 : SUCCESS in 3m 47s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"663d1781689d3681ef52022c9d65acb789fd7e3f","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-12 18:06:04.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b9e49dae03d49d55baec691b3c46b0166f4fa31c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 19:01:42.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a665e3213bff4246a8607b245b82e887\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/27fd686fdb9e42c3b5aa80bb0e015119 : SUCCESS in 4m 11s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4a292325d90b46b085d18e4ecbae7ed0 : SUCCESS in 32m 06s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/312f99a837454ee4bb5da2b561efbd31 : SUCCESS in 52m 09s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/851bf21d3c2343d9b52582ff61edcbc2 : SUCCESS in 42m 33s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/a37930fda6054352af9b5a155e653023 : SUCCESS in 37m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/363c7ce18e3a47e9a94671269a29d90d : SUCCESS in 36m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5c7efabf950f4f86b63b1d276a6220e9 : SUCCESS in 48m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/dc53a657dfa141b8ae1ef56765001b46 : SUCCESS in 52m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/77a3349427af4976926728038cd796d4 : SUCCESS in 38m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/10707bc8b678477eb2e7d1495ddc7632 : SUCCESS in 43m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6528a3272f684b72a65f386032afb7af : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13ef0b0329134bee818c79731c6bf5b8 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b3cb95c3df04145ad3b6be107c16d09 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6152341bc99d4854a38677ed6378e767 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c921d9b0a4264b968abd66939c4097bc : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83de727feac84e97bba4dad5843e93fc : SUCCESS in 4m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c686abfe57834113b486b08b015b4655 : SUCCESS in 26m 21s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9dd3b397f29f420ea8ecec44699b84d9 : SUCCESS in 4m 06s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"73de85b82cafa8e0f1c07928d735ba9a4508d0b3","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-15 14:51:28.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e6a1adc918aadf70382c39a86f1366bba4ed0805","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-19 19:39:40.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2f2f1f0d68c6c185a8038d7ee7dac462616e9d96","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-19 19:47:36.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"1906d9d6a013d2f6c421d74f81d610e681b30292","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-19 19:49:00.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"8edf5e9f4775ef6bd451a709f6577704d79a5f19","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-10-19 19:51:14.000000000","message":"Patch Set 11: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"7c17fb12a17e7157e833c527031a55b9aad6b9fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-10-19 20:12:44.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"20e165877684ac01c14d992a8834a1294d0d9833","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-10-19 20:26:26.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"2db0871ab7107880110d1ee6938e462dccf45787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 21:37:53.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9649297166c4147863aaebdeceab907\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c88b2989c7b74e65bdb62d0e1f83fb07 : SUCCESS in 3m 42s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/918d6f9b9488490bb21c4fbed7ca7c6d : SUCCESS in 40m 52s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/24464534f02e439f9d757bfc92d8a64f : SUCCESS in 53m 34s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8289120a4d594873a1f637e041e458ef : SUCCESS in 34m 17s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/eb868156ce4447369f3bd7b438b8fc94 : SUCCESS in 1h 07m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f8c8ec5653014783843448887901595c : SUCCESS in 29m 47s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9fab0a81de2f426498e568c8aa222c76 : SUCCESS in 46m 51s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/77aa6411c37f4f7e81d4514e8d236b2c : SUCCESS in 1h 21m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ecd0342d56224429bd46605ad5831771 : SUCCESS in 55m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a159769ded984e36b44ac89b24d6e386 : SUCCESS in 55m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30c6813574c647d5af59b8a87afbea79 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d7e1ca35b834b4face31ca51bc7d432 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9daf7a981aea496b94beb019df032344 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba823e1719bb4e129dff629ff5de482f : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6adc6e394874c529a295304674c6aff : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25823a06d21c406782c51ec758bc92c5 : SUCCESS in 4m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d8ca2d97e18141418d48091da0c6c2bf : SUCCESS in 30m 04s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/41d76e63edd54111b1d813a8237a7730 : SUCCESS in 4m 59s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"954d0282289ead2c8f69f6d92610cf7433a0028b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-20 06:42:05.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"abfa9340b951bb6fdf89c0815814baf8ef3e9139","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-20 06:42:18.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c587736dfc112065a7e0617300ffdfc0c323def1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 07:58:54.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/283160806ffc43dd9a9234e6642885da\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2c597db0b78c4bf38ba47dc429991ecf : SUCCESS in 4m 06s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ed7750bac9b24caa8cedaf631cabb12e : SUCCESS in 43m 37s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5470bfab9a1d4355b01abfa5ca975b1a : SUCCESS in 1h 06m 08s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/31378e8d418548f6b84414757f9ddf14 : SUCCESS in 48m 27s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/a2dc6eb70fcc4838b2f7672bd5ffbd88 : SUCCESS in 48m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7646de725a5a4339a8181ee764e7f577 : SUCCESS in 50m 12s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/444d63cfef6641e9b45f8d2be15d55d6 : SUCCESS in 1h 11m 49s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f0c05aeffaf4418796495e6ff7af7a74 : SUCCESS in 1h 02m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4465d5a4a4064305bd051bbd8212b5ca : SUCCESS in 36m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/15d892567a7b4d80bfd80e628e4df98c : SUCCESS in 34m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fcffb5a9eba4dad8a88f7acd6c6384e : FAILURE in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56de58289af1444786f916a52c732cdc : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93656c9a8bde438e9ac5ce75dafe2c7a : FAILURE in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92b3b5c52d4f4192b75ff68a3da35f17 : FAILURE in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d7857509568425aa20b0f94d0d6e190 : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f05d596129a04c14a76240462450ff8e : SUCCESS in 3m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aabac984deb249c4997214dad803e43d : SUCCESS in 33m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9d3aa5bb7e7c42fe8590dceb29953d92 : FAILURE in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"482dfe15e57de1a90dc8bd799c745523752e2b4c","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-20 08:47:34.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"516fe5282b51b1c9d7126c76e23e28c3d9af780e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 09:45:59.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/203d62a03fcc42d2a7f90132cdb52817\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e30545b36a2045c28968fd8440d1525a : SUCCESS in 4m 37s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7d40a446859744f3914c1cb41aca1df2 : SUCCESS in 41m 00s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a6d2bf34ed4f4533a49b0c7d5280d429 : SUCCESS in 50m 03s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1244668c395140dda6ef79f8c6cb8e55 : SUCCESS in 45m 15s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/bf7f64b6e42047199af25f14e998cdb1 : SUCCESS in 38m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bedf9fc9c0164fe3a07a453fbf7caf9e : SUCCESS in 38m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9a2cca555b0446028c2a1471a4d8c815 : SUCCESS in 52m 28s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b55195c343fa47f9907c5872e4fb99bf : SUCCESS in 52m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/23879016daff48838429cfe2742bc32b : SUCCESS in 38m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3b7bbad8f2fa4cbd8f7adee8aef3bbcd : SUCCESS in 41m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a89fd2e8ed374bf880c0b68e25bdcaf7 : FAILURE in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f80d7a05acf4176aad99a90391bddaa : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df0fb2508da243a3b8100a437d0b3572 : FAILURE in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/001189f26d504a408f02b4c7544c19c0 : FAILURE in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2264a8dfd54543ccbd8eed56a50fef99 : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6def286f65ec4b22bf496ebb0482632b : SUCCESS in 4m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/77738202fd874a0f82371ff0110f256a : SUCCESS in 27m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/96aa5e0682b744caa681d87731cc3a0e : FAILURE in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"e27e674da877211ca1216bbb4b1e4d168435a02c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-20 10:01:30.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"eecd1f45aff4cfc8ec4fe7cdf86902e1b0bcd1f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 11:10:38.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93db0b300bcb4d3d8301dc6ef8e4012c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c436da12c5854f5892085a8fa4b5a895 : SUCCESS in 3m 40s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/39ba292e60d94431abd70a81b7ca031b : SUCCESS in 33m 45s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/dfcc9792f99d44b9a7d0771032eb37eb : SUCCESS in 47m 33s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e57347a3012d4bc29924ee1faa1068c5 : SUCCESS in 52m 31s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8c24978ad5194759be3d00692868914d : SUCCESS in 35m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4e14afa5fabf4e4ab108f767fbe04542 : SUCCESS in 43m 13s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/453250160f7c4fe79e1d0df56e56833a : SUCCESS in 46m 19s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fba291249d334f7e8b50cdcf18fc3e62 : SUCCESS in 1h 02m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7ed2f9db51624045b3f13d8366db3777 : SUCCESS in 34m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/25930d9e7b4946af8de3e5b42c576704 : SUCCESS in 46m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6be5c98084a4583a77ed1eaccef08df : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f19b54d3eb9a4f3b9c42be2fa68de40c : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3240e0f42864f30b737ed9729a055f3 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c726c9525b2140ddacd3849a532fc441 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d895ad195c2d4240a35d27017a1b47ca : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5e77797fa8c4e9882d11a08273357af : SUCCESS in 3m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/469b4d3af4c24c57b9771e013903799c : SUCCESS in 29m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d08a8dbf7d9843aa806264b6e25362a7 : SUCCESS in 4m 22s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"48abd7bb67423a192f15784deb8a88248fb813d6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-10-20 15:48:04.000000000","message":"Patch Set 14: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"f24978eaec8e15b105e33fba0cd3a4f036bc3459","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-10-20 23:40:08.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"77873629e57c7dd76b01546cd2272bdc29e5e44a","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2022-10-21 20:27:09.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"42e3176a6920219410f2dc8d2f7f290708b81e57","author":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"date":"2022-10-22 06:19:48.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"783ed954de91e56ececb7bb3db15b0797a493f71","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-10-24 15:57:17.000000000","message":"Patch Set 14: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"32548286fdc2fbc22bc70b0e7f3df2087149161d","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2022-11-01 15:22:07.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"8ee5fbcb7cdc1833b6e25ac01c033b5085a5fc4d","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2022-11-18 11:57:15.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"e2d24581ed0b455329487ea442a837ad7cf16563","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-11-22 10:15:41.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"01a5a0ab35423c83be638d3048949b8d378ceb75","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-23 00:25:25.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"253306e98c0cf4ad5f01f9243e8d3e744b2d8e33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 00:26:01.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"b1dd89d42f0c8b0155fd6b266bc6d9a651798603","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 01:23:51.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f3e999cb5f54d2fae334f15a15085fc\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a2297e6b089f44e48320c0fd9715ad7f : SUCCESS in 3m 12s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/91139525aee14469a2a8bbd8d1821b31 : SUCCESS in 28m 35s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/454accdcab3a4d7fb226743efa0b3131 : SUCCESS in 28m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e55765bb4c9f4e8eb30a514bd9936a49 : SUCCESS in 35m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5625158163dc4093bb6a749f4a552695 : SUCCESS in 57m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bb6c105a17654fe89221c2c48b9d445a : SUCCESS in 34m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/271fcd9e31c94226812f6116b2d27556 : SUCCESS in 36m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6832dc52aeaf4e4eb1492cef72ec7a40 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/589a486464e740158fa7f586538201f2 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/425cf6f2fb9a47b7843ce7337ce42676 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dd662736c4443bc9960c9cee2c1bd6b : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f3fe8306b7e445cb8292fbbd26d9248 : SUCCESS in 3m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d7daa3c3879c4eb9905af2400b7bc608 : SUCCESS in 25m 06s","accounts_in_message":[],"_revision_number":14},{"id":"909f5bda29881b2f43a1d18398009028c2c6fe7c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 01:23:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"19204f4381ca320f777af6da43889cdc58ffcc41","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 01:24:55.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c5b62f960ba407cac1fd6c1b1585bf7\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/da808e2b46854d6ab6d0c14b1b13297a : SUCCESS in 41s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd04b434214a4567a839eb4dc6a5e535 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bca17e2c0a944039a635e1c9e9c9bf55 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"d193b0c70c8ee1c12310f8e211307088287561f1","revisions":{"4cc8d2c227e7cdb36bf33549b78cd96fa1354fbf":{"kind":"REWORK","_number":1,"created":"2022-03-25 20:15:23.000000000","uploader":{"_account_id":34181,"name":"Manish Honap","email":"mmhonap@gmail.com","username":"mhonap"},"ref":"refs/changes/17/835317/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/835317/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/835317/1"}}},"commit":{"parents":[{"commit":"c998ff690058ea7319e6bc92b3f841a77dee8d6e","subject":"Remove unused downgrade functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c998ff690058ea7319e6bc92b3f841a77dee8d6e"}]}],"author":{"name":"Manish Honap","email":"mmhonap@gmail.com","date":"2022-03-25 19:58:02.000000000","tz":330},"committer":{"name":"Manish Honap","email":"mmhonap@gmail.com","date":"2022-03-25 20:15:20.000000000","tz":330},"subject":"Delete back-end zone-files when zone is deleted","message":"Delete back-end zone-files when zone is deleted\n\nWe do not delete the zone-files on the back-end (Bind) when the zone is\ndeleted in Designate. This results in tens of thousand leftover zone files on\nBind.\n\nThis commit provides a separate option in Designate Zone deletion API to\n\"force cleaning zone\" on the back-end. This option is restricted for admin and\nowner roles.\n\nCloses-Bug: 1966517\nChange-Id: Ic7b8fee4d4702b0632774d32542b23d7d2a8c253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4cc8d2c227e7cdb36bf33549b78cd96fa1354fbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4cc8d2c227e7cdb36bf33549b78cd96fa1354fbf"}]},"branch":"refs/heads/master"},"913b8ffeeeac0cbee7178cc30a7e5272d09d5606":{"kind":"REWORK","_number":2,"created":"2022-03-26 13:49:15.000000000","uploader":{"_account_id":34181,"name":"Manish Honap","email":"mmhonap@gmail.com","username":"mhonap"},"ref":"refs/changes/17/835317/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/835317/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/835317/2"}}},"commit":{"parents":[{"commit":"c998ff690058ea7319e6bc92b3f841a77dee8d6e","subject":"Remove unused downgrade functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c998ff690058ea7319e6bc92b3f841a77dee8d6e"}]}],"author":{"name":"Manish Honap","email":"mmhonap@gmail.com","date":"2022-03-25 19:58:02.000000000","tz":330},"committer":{"name":"Manish Honap","email":"mmhonap@gmail.com","date":"2022-03-26 13:48:15.000000000","tz":330},"subject":"Delete back-end zone-files when zone is deleted","message":"Delete back-end zone-files when zone is deleted\n\nWe do not delete the zone-files on the back-end (Bind) when the zone is\ndeleted in Designate. This results in tens of thousand leftover zone files on\nBind.\n\nThis commit provides a separate option in Designate Zone deletion API to\n\"force cleaning zone\" on the back-end. This option is restricted for admin and\nowner roles.\n\nCloses-Bug: 1966517\nChange-Id: Ic7b8fee4d4702b0632774d32542b23d7d2a8c253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/913b8ffeeeac0cbee7178cc30a7e5272d09d5606"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/913b8ffeeeac0cbee7178cc30a7e5272d09d5606"}]},"branch":"refs/heads/master"},"49549c34ad4677d5124afada2849d74ad01cba26":{"kind":"REWORK","_number":3,"created":"2022-04-30 13:12:04.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/17/835317/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/835317/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/835317/3"}}},"commit":{"parents":[{"commit":"fc905bf76031a88d1164fc613295c9b6acdfe8fe","subject":"Removing more unused rpc calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc905bf76031a88d1164fc613295c9b6acdfe8fe"}]}],"author":{"name":"Manish Honap","email":"mmhonap@gmail.com","date":"2022-03-25 19:58:02.000000000","tz":330},"committer":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2022-04-30 13:04:20.000000000","tz":120},"subject":"Add option to force delete zone-files in delete API.","message":"Add option to force delete zone-files in delete API.\n\nDesignate does not delete the zone-files on the back-end when zone is deleted\nvia delete API. This results in thousands leftover zone files on backend e.g.\nbind. Add option in designate zone delete API to clean/delete zone files on\nthe back-end. This option is restricted for admin or owner roles.\n\nCloses-Bug: 1966517\nChange-Id: Ic7b8fee4d4702b0632774d32542b23d7d2a8c253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/49549c34ad4677d5124afada2849d74ad01cba26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/49549c34ad4677d5124afada2849d74ad01cba26"}]},"branch":"refs/heads/master"},"33e1c6e32784e4a5d7887a455ca245a1d628ab76":{"kind":"REWORK","_number":4,"created":"2022-04-30 13:16:29.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/17/835317/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/835317/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/835317/4"}}},"commit":{"parents":[{"commit":"fc905bf76031a88d1164fc613295c9b6acdfe8fe","subject":"Removing more unused rpc calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc905bf76031a88d1164fc613295c9b6acdfe8fe"}]}],"author":{"name":"Manish Honap","email":"mmhonap@gmail.com","date":"2022-03-25 19:58:02.000000000","tz":330},"committer":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2022-04-30 13:14:29.000000000","tz":120},"subject":"Add option to force delete zone-files in delete API.","message":"Add option to force delete zone-files in delete API.\n\nDesignate does not delete the zone-files on the back-end when zone is\ndeleted. This results in thousands leftover zone files on backend e.g.\nbind. Add option in designate zone delete API to force clean/delete\nzone-files on the back-end. This option is restricted for admin or\nowner roles.\n\nCloses-Bug: 1966517\nChange-Id: Ic7b8fee4d4702b0632774d32542b23d7d2a8c253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/33e1c6e32784e4a5d7887a455ca245a1d628ab76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/33e1c6e32784e4a5d7887a455ca245a1d628ab76"}]},"branch":"refs/heads/master"},"d79466311eb3ba2df4b01f63ddb9695603c8efd5":{"kind":"REWORK","_number":5,"created":"2022-06-14 17:50:35.000000000","uploader":{"_account_id":34181,"name":"Manish Honap","email":"mmhonap@gmail.com","username":"mhonap"},"ref":"refs/changes/17/835317/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/835317/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/835317/5"}}},"commit":{"parents":[{"commit":"f41a73c6e5fefb1f08d3fd54afa9c6e02df76d31","subject":"Merge \"Do not declare queue at a pipeline level\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f41a73c6e5fefb1f08d3fd54afa9c6e02df76d31"}]}],"author":{"name":"Manish Honap","email":"mmhonap@gmail.com","date":"2022-03-25 19:58:02.000000000","tz":330},"committer":{"name":"Manish Honap","email":"mmhonap@gmail.com","date":"2022-06-14 17:49:43.000000000","tz":330},"subject":"Add option to force delete zone-files in delete API.","message":"Add option to force delete zone-files in delete API.\n\nDesignate does not delete the zone-files on the back-end when zone is\ndeleted. This results in thousands leftover zone files on backend e.g.\nbind. Add option in designate zone delete API to force clean/delete\nzone-files on the back-end. This option is restricted for admin or\nowner roles.\n\nCloses-Bug: 1966517\nChange-Id: Ic7b8fee4d4702b0632774d32542b23d7d2a8c253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d79466311eb3ba2df4b01f63ddb9695603c8efd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d79466311eb3ba2df4b01f63ddb9695603c8efd5"}]},"branch":"refs/heads/master"},"d4074b46ac12b7f6b41c4673dad2ef2ffd718a65":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-06-27 17:01:09.000000000","uploader":{"_account_id":34181,"name":"Manish Honap","email":"mmhonap@gmail.com","username":"mhonap"},"ref":"refs/changes/17/835317/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/835317/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/835317/6"}}},"commit":{"parents":[{"commit":"d05232fc075ffe6bfaf7a537334fe081bb41a65c","subject":"Merge \"Removed RPC calls from MDNS and moved them to the Worker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d05232fc075ffe6bfaf7a537334fe081bb41a65c"}]}],"author":{"name":"Manish Honap","email":"mmhonap@gmail.com","date":"2022-03-25 19:58:02.000000000","tz":330},"committer":{"name":"Manish Honap","email":"mmhonap@gmail.com","date":"2022-06-27 17:00:27.000000000","tz":330},"subject":"Add option to force delete zone-files in delete API.","message":"Add option to force delete zone-files in delete API.\n\nDesignate does not delete the zone-files on the back-end when zone is\ndeleted. This results in thousands leftover zone files on backend e.g.\nbind. Add option in designate zone delete API to force clean/delete\nzone-files on the back-end. This option is restricted for admin or\nowner roles.\n\nCloses-Bug: 1966517\nChange-Id: Ic7b8fee4d4702b0632774d32542b23d7d2a8c253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d4074b46ac12b7f6b41c4673dad2ef2ffd718a65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d4074b46ac12b7f6b41c4673dad2ef2ffd718a65"}]},"branch":"refs/heads/master"},"ffa7d6ecadb892e588faf5c1af063f8b942084f2":{"kind":"REWORK","_number":7,"created":"2022-08-10 09:28:51.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/17/835317/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/835317/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/835317/7"}}},"commit":{"parents":[{"commit":"f3c29e8bceea832e1cb787740fa676b10a410848","subject":"Merge \"Fixed issues with __repr__ and __str__ on objects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f3c29e8bceea832e1cb787740fa676b10a410848"}]}],"author":{"name":"Manish Honap","email":"mmhonap@gmail.com","date":"2022-03-25 19:58:02.000000000","tz":330},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-08-10 09:28:08.000000000","tz":0},"subject":"Add option to force delete zone-files in delete API.","message":"Add option to force delete zone-files in delete API.\n\nDesignate does not delete the zone-files on the back-end when zone is\ndeleted. This results in thousands leftover zone files on backend e.g.\nbind. Add option in designate zone delete API to force clean/delete\nzone-files on the back-end. This option is restricted for admin or\nowner roles.\n\nCloses-Bug: 1966517\nChange-Id: Ic7b8fee4d4702b0632774d32542b23d7d2a8c253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffa7d6ecadb892e588faf5c1af063f8b942084f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffa7d6ecadb892e588faf5c1af063f8b942084f2"}]},"branch":"refs/heads/master"},"c57cf5cbed167b7f619c4c4c11fef3d2cd90bdc1":{"kind":"REWORK","_number":8,"created":"2022-10-03 08:30:51.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/17/835317/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/835317/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/835317/8"}}},"commit":{"parents":[{"commit":"6ca038f350f36ee13ea45f50ddf3624e43e32e6a","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ca038f350f36ee13ea45f50ddf3624e43e32e6a"}]}],"author":{"name":"Manish Honap","email":"mmhonap@gmail.com","date":"2022-03-25 19:58:02.000000000","tz":330},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-10-03 08:30:41.000000000","tz":0},"subject":"Add option to force delete zone-files in delete API.","message":"Add option to force delete zone-files in delete API.\n\nDesignate does not delete the zone-files on the back-end when zone is\ndeleted. This results in thousands leftover zone files on backend e.g.\nbind. Add option in designate zone delete API to force clean/delete\nzone-files on the back-end. This option is restricted for admin or\nowner roles.\n\nCloses-Bug: 1966517\nChange-Id: Ic7b8fee4d4702b0632774d32542b23d7d2a8c253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c57cf5cbed167b7f619c4c4c11fef3d2cd90bdc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c57cf5cbed167b7f619c4c4c11fef3d2cd90bdc1"}]},"branch":"refs/heads/master"},"3593de9561c793914792c1b7e72830aa3fe9d65c":{"kind":"REWORK","_number":9,"created":"2022-10-06 10:29:34.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/17/835317/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/835317/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/835317/9"}}},"commit":{"parents":[{"commit":"6ca038f350f36ee13ea45f50ddf3624e43e32e6a","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ca038f350f36ee13ea45f50ddf3624e43e32e6a"}]}],"author":{"name":"Manish Honap","email":"mmhonap@gmail.com","date":"2022-03-25 19:58:02.000000000","tz":330},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-10-04 20:07:19.000000000","tz":0},"subject":"Add option to force delete zone-files in delete API.","message":"Add option to force delete zone-files in delete API.\n\nDesignate does not delete the zone-files on the back-end when zone is\ndeleted. This results in thousands leftover zone files on backend e.g.\nbind. Add option in designate zone delete API to force clean/delete\nzone-files on the back-end. This option is restricted for admin or\nowner roles.\n\nCloses-Bug: 1966517\nChange-Id: Ic7b8fee4d4702b0632774d32542b23d7d2a8c253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3593de9561c793914792c1b7e72830aa3fe9d65c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3593de9561c793914792c1b7e72830aa3fe9d65c"}]},"branch":"refs/heads/master"},"09c53b04f3df51e20d252157f82f0ae111b8b97a":{"kind":"REWORK","_number":10,"created":"2022-10-12 16:24:52.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/17/835317/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/835317/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/835317/10"}}},"commit":{"parents":[{"commit":"6fee42ecac2ff1f3a8393ab069c1043ce891f4ef","subject":"Remove unicode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6fee42ecac2ff1f3a8393ab069c1043ce891f4ef"}]}],"author":{"name":"Manish Honap","email":"mmhonap@gmail.com","date":"2022-03-25 19:58:02.000000000","tz":330},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-10-12 16:24:23.000000000","tz":0},"subject":"Add option to force delete zone-files in delete API.","message":"Add option to force delete zone-files in delete API.\n\nDesignate does not delete the zone-files on the back-end when zone is\ndeleted. This results in thousands leftover zone files on backend e.g.\nbind. Add option in designate zone delete API to force clean/delete\nzone-files on the back-end. This option is restricted for admin or\nowner roles.\n\nCloses-Bug: 1966517\nChange-Id: Ic7b8fee4d4702b0632774d32542b23d7d2a8c253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/09c53b04f3df51e20d252157f82f0ae111b8b97a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/09c53b04f3df51e20d252157f82f0ae111b8b97a"}]},"branch":"refs/heads/master"},"8cb32aff4ff38c7c69cdea25842e0a2a7e63460f":{"kind":"REWORK","_number":11,"created":"2022-10-19 19:49:00.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/17/835317/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/835317/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/835317/11"}}},"commit":{"parents":[{"commit":"06b297eaf39b084a7f999e32c41f16eefc3ad859","subject":"Merge \"CI: Move queue configuration into project-template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06b297eaf39b084a7f999e32c41f16eefc3ad859"}]}],"author":{"name":"Manish Honap","email":"mmhonap@gmail.com","date":"2022-03-25 19:58:02.000000000","tz":330},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-10-19 19:48:34.000000000","tz":0},"subject":"Add option to force delete zone-files in delete API.","message":"Add option to force delete zone-files in delete API.\n\nDesignate does not delete the zone-files on the back-end when zone is\ndeleted. This results in thousands leftover zone files on backend e.g.\nbind. Add option in designate zone delete API to force clean/delete\nzone-files on the back-end. This option is restricted for admin or\nowner roles.\n\nCloses-Bug: 1966517\nChange-Id: Ic7b8fee4d4702b0632774d32542b23d7d2a8c253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8cb32aff4ff38c7c69cdea25842e0a2a7e63460f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8cb32aff4ff38c7c69cdea25842e0a2a7e63460f"}]},"branch":"refs/heads/master"},"814e78e37514c54997a3f44b5616c2904573d992":{"kind":"REWORK","_number":12,"created":"2022-10-19 20:12:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/835317/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/835317/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/835317/12"}}},"commit":{"parents":[{"commit":"06b297eaf39b084a7f999e32c41f16eefc3ad859","subject":"Merge \"CI: Move queue configuration into project-template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06b297eaf39b084a7f999e32c41f16eefc3ad859"}]}],"author":{"name":"Manish Honap","email":"mmhonap@gmail.com","date":"2022-03-25 19:58:02.000000000","tz":330},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-10-19 20:12:42.000000000","tz":0},"subject":"Add option to force delete zone-files in delete API.","message":"Add option to force delete zone-files in delete API.\n\nDesignate does not delete the zone-files on the back-end when zone is\ndeleted. This results in thousands leftover zone files on backend e.g.\nbind. Add option in designate zone delete API to force clean/delete\nzone-files on the back-end. This option is restricted for admin or\nowner roles.\n\nCloses-Bug: 1966517\nChange-Id: Ic7b8fee4d4702b0632774d32542b23d7d2a8c253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/814e78e37514c54997a3f44b5616c2904573d992"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/814e78e37514c54997a3f44b5616c2904573d992"}]},"branch":"refs/heads/master"},"28b7ee1b164b48e469d0b7b7830f4cb91780cca1":{"kind":"REWORK","_number":13,"created":"2022-10-20 06:42:05.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/17/835317/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/835317/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/835317/13"}}},"commit":{"parents":[{"commit":"06b297eaf39b084a7f999e32c41f16eefc3ad859","subject":"Merge \"CI: Move queue configuration into project-template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06b297eaf39b084a7f999e32c41f16eefc3ad859"}]}],"author":{"name":"Manish Honap","email":"mmhonap@gmail.com","date":"2022-03-25 19:58:02.000000000","tz":330},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-10-20 06:41:58.000000000","tz":0},"subject":"Add option to force delete zone-files in delete API.","message":"Add option to force delete zone-files in delete API.\n\nDesignate does not delete the zone-files on the back-end when zone is\ndeleted. This results in thousands leftover zone files on backend e.g.\nbind. Add option in designate zone delete API to force clean/delete\nzone-files on the back-end. This option is restricted for admin or\nowner roles.\n\nCloses-Bug: 1966517\nChange-Id: Ic7b8fee4d4702b0632774d32542b23d7d2a8c253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28b7ee1b164b48e469d0b7b7830f4cb91780cca1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28b7ee1b164b48e469d0b7b7830f4cb91780cca1"}]},"branch":"refs/heads/master"},"d193b0c70c8ee1c12310f8e211307088287561f1":{"kind":"REWORK","_number":14,"created":"2022-10-20 10:01:30.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/17/835317/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/835317/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/835317/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/835317/14"}}},"commit":{"parents":[{"commit":"06b297eaf39b084a7f999e32c41f16eefc3ad859","subject":"Merge \"CI: Move queue configuration into project-template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06b297eaf39b084a7f999e32c41f16eefc3ad859"}]}],"author":{"name":"Manish Honap","email":"mmhonap@gmail.com","date":"2022-03-25 19:58:02.000000000","tz":330},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-10-20 10:01:23.000000000","tz":0},"subject":"Add option to force delete zone-files in delete API.","message":"Add option to force delete zone-files in delete API.\n\nDesignate does not delete the zone-files on the back-end when zone is\ndeleted. This results in thousands leftover zone files on backend e.g.\nbind. Add option in designate zone delete API to force clean/delete\nzone-files on the back-end. This option is restricted for admin or\nowner roles.\n\nCloses-Bug: 1966517\nChange-Id: Ic7b8fee4d4702b0632774d32542b23d7d2a8c253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d193b0c70c8ee1c12310f8e211307088287561f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d193b0c70c8ee1c12310f8e211307088287561f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[]}
