)]}'
{"id":"openstack%2Fdesignate~835651","triplet_id":"openstack%2Fdesignate~master~If7d50cbd4fa0ce86e0ddf03068da36acd7d72cb4","project":"openstack/designate","branch":"master","topic":"bug/1950118","attention_set":{},"removed_from_attention_set":{"28609":{"account":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"last_update":"2022-04-25 20:58:03.000000000","reason":"Change was submitted"},"5572":{"account":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"last_update":"2022-04-25 20:58:03.000000000","reason":"Change was submitted"},"13995":{"account":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"last_update":"2022-04-25 20:58:03.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-04-25 19:43:30.000000000","reason":"Michael Johnson replied on the change"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-04-20 19:03:53.000000000","reason":"Erik Olof Gunnar Andersson replied on the change"}},"hashtags":[],"change_id":"If7d50cbd4fa0ce86e0ddf03068da36acd7d72cb4","subject":"Clarifies the zone import error message","status":"MERGED","created":"2022-03-29 15:34:54.000000000","updated":"2022-04-25 20:59:27.000000000","submitted":"2022-04-25 20:58:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":36,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835651-bug/1950118","meta_rev_id":"c0e6740fe66fca5f15066cecdef0c23dbba8b34f","_number":835651,"virtual_id_number":835651,"owner":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-25 20:58:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-25 19:43:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-04-20 19:03:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-25 19:43:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"}],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-29 15:36:07.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"state":"REVIEWER"},{"updated":"2022-03-29 16:48:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-29 18:38:53.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2022-03-29 19:57:33.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-04-12 20:38:14.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-04-13 15:48:54.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"}],"messages":[{"id":"315167085ad202f95618d741ee228551d6761458","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-03-29 15:34:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6cd855518a69476ad53eec7fe653fb1dbca0a10a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 16:48:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fb032c7b7e9845be930e3bcbd40d7f49 : SUCCESS in 43m 55s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/3275bcfc890c44228e288a6b78936be7 : SUCCESS in 55m 34s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/33831a92b3fd4454bd53cfbd79120d9d : SUCCESS in 49m 38s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/374f61ddc08d40549604780cecf28ab7 : SUCCESS in 48m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7f320d216f9842f9bd5e11f9f50b0276 : SUCCESS in 43m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d663b51ec48a4e5485aa83a6efdd87cd : SUCCESS in 1h 02m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7b56bb05243a49a5a536e54af719b146 : SUCCESS in 46m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4a0fa98545ea4ba9922233abd3153880 : SUCCESS in 44m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/638744ea83fd422483e9a94d78279332 : SUCCESS in 7m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ed3bbf67d0744b5a97b075b7ddcb29c : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/785271eeefa54f4ab5f963c7143c1119 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9d71b943b544b70890d86fbac63f765 : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6a70d2ea4cf4bf68841f4f54d983f0d : SUCCESS in 7m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a102fa0930bd47379dd5049340cff38c : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0edef7709d384626917dcdbdf7c4f4fa : SUCCESS in 8m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8483f83e935742709340ed09d0e26be5 : SUCCESS in 28m 17s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/62c4da6686c346acb632cf6a5de181be : SUCCESS in 6m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f80a2b2dc7a871b1dcd2abc923fbd294045b98b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-03-29 18:38:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"69516282168ee02c275e68ef4ae89a8224a90022","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-03-29 19:57:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8a2b9e842346ca9b879e33dd28c32245f719d0c4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-03-30 16:49:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8696451c496fc4641507667c94a6e39ef6c07401","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-03-30 19:52:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9dc6ba874149356cafd05c8bf5128ccd3b6155d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-03-30 23:27:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf78cfb0964f133534d705b42b8ff0b147b62b8c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-03 01:34:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"61f3ab29090669091e9d011a205694f4039176a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-12 20:24:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3d2af073fd7bb9bf9a41fee5161f0ea5ec74e558","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-12 20:38:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d82ae11ebf6d061b6c97ea55c2c0427a2e06b7b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 21:37:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/16a485c1d820477eaa8bf5d0c5f2d3f6 : SUCCESS in 47m 34s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/7562ca65f2b04eb48d101f35ff690497 : SUCCESS in 1h 06m 04s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/51519536b0bb43cca15e61d41de7a893 : SUCCESS in 57m 21s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/700cd98829f84145afec3d7227e59f6e : SUCCESS in 44m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d7439211ba3347a289a953b0af69e166 : SUCCESS in 46m 42s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2c1f5c436d204c6b9af4f36b1cbbc26e : FAILURE in 41m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/76b41349d0dc4681ba7062777d629089 : SUCCESS in 43m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/40fcef8527954109a034db0112d9c5c1 : SUCCESS in 46m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0746c15e5de44c5aa7ea43e3664d85c2 : SUCCESS in 5m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43269c3f108245f7b748961366bbaaca : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1b46c7236a64885ba108a5593ff28ab : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b45f998436a44427b9665a82e51afd88 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/585c9b3f578441a782973f40124add27 : SUCCESS in 5m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e75d9cc8bdd4f48ac41df93ace36b8c : RETRY_LIMIT in 3m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6ae4e55d1c740de9c7a62820c2b12f8 : SUCCESS in 8m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b6c9dd427985470ba820254eeb12053a : SUCCESS in 25m 45s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4f7cfa3ec7b2449eac515f2a4e868064 : SUCCESS in 5m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e1f8e880f6bc3b0dc69e2b1fd72eb93663085b5f","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2022-04-13 15:56:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c5a0f2410a5b8a2ee0f1fe0891606be3d0c62592","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-13 16:07:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0cc8906b223897e0e6aaa79a1c42941beb42d149","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-13 21:50:38.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5198bb823778d0fed90faf3195196dc1cd6fcc73","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-13 21:51:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d57116c4a9cea0c4fb8df19693a2a8073fd02bc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-14 15:04:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"689dbe73a1ff3e285ac7a9205d835703a6d77aa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 16:22:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8d452237726f428ab0a6c206131c9756 : SUCCESS in 43m 30s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/57e3a8088b724ab197f11303a143a12d : SUCCESS in 34m 25s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a87016f5bcf4dafa9a1c609a50c2130 : SUCCESS in 50m 29s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/db6011078b764fcc8f5c19dcb9722166 : SUCCESS in 48m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c649b159ef0741d28418188f9cebc08e : SUCCESS in 46m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0c5dba6783814b43b4b1d483f50e2c4f : SUCCESS in 1h 11m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4e4626ded2bf4e9e82805d6f3cc84b07 : SUCCESS in 44m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/79790daa73e04b9cb4e27a6c94efa6cc : SUCCESS in 29m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0fbcbd85ea54a4bbadb9ba28d9578b9 : SUCCESS in 5m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d62497fa1834c589fe5f6c02c46c973 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e7f5e55a01b439f9645210ca6ee37ba : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/427d16ef4d7a461d90f3bdd7cc947060 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d05bea57bd6429b92de267b5136b9a5 : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/440fbec0211f46d0925f76414b86ea01 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5469e5916c014a6589047eec056252d2 : SUCCESS in 7m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/74b97f4f7a0f47d2bbe5dfd9b95158d0 : SUCCESS in 27m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9a62afed57114ed89b8acd4ebdb98785 : SUCCESS in 5m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"21e809df1881569ef9f45f2236850422813543d2","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-14 16:22:38.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9dc5b13e3a6d5fb16b9936fb85bd5f3c7eb77a72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-14 16:37:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b2ac651bd5b645a94a8c29c7f1451720da60642d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-14 17:05:24.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e236cea24a8ec682bd9e492ef3ed4294fd1880dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 17:30:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0b8e057192174c36b9c898c6c0672510 : SUCCESS in 34m 33s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/2ea3d1b10038464e9767f2276470322b : SUCCESS in 39m 23s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a89021d7e293480c83348cf85a22237d : SUCCESS in 38m 05s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/36247a72a7e94e869209809752fa9935 : SUCCESS in 33m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bd8f074ea4d8456fbd52011f79f1bc37 : SUCCESS in 33m 52s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f1aa5e46309244f7a4e31bf3c2b29652 : SUCCESS in 50m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/12065a74557940f986a141abdfbbe43b : SUCCESS in 36m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/732ecf65636e4322b32708b89c1c5e6a : SUCCESS in 34m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03cd21955bd24e0ba58aecc8e325c270 : SUCCESS in 6m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4dff6b2fcc5407ab9b1b43cfa0ee4e7 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb243669c83942808b4c56bb4baaf91a : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f111d80998b949248a5b99860482e4e1 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42ddc1b9834e41e780fd481cc000c287 : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ecb612cbebd4f7ebb13f507e0fc4b9c : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/983b53e68f3e4c75a06011e01ac3d140 : SUCCESS in 9m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f3b4168425ed40d78040371d83f38a9e : SUCCESS in 33m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4698b3e792df4a8f8e690cb1735803d8 : SUCCESS in 5m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"18776c29ffb46a2a033a5e0df1ffb265ee51efb0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-14 22:32:47.000000000","message":"Patch Set 4: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"bbd450474ceb19873fddf49c9c3767091659c837","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-18 20:25:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d5ef72e598e66b8a42c4c36bd2993b0c84e2e18b","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-18 20:27:24.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"53f2d71acba9e3e1a773051a44859bdde5bf40d1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-18 21:35:38.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"07e12f0d596dd076e8f667c077ed57dbecdd7ca6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 21:43:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ee83213d76144c70a327dd7a716e0672 : SUCCESS in 47m 12s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/fff789135ad2493ab7376e5f36c4d2ba : SUCCESS in 58m 38s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d1346b55fe54d42922e57196a0de3f2 : SUCCESS in 49m 00s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/b11f7fbb694d4f9c829c6edb44e1075b : SUCCESS in 43m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c4c8f8a0f1cb428bb014acbc6c9a8615 : SUCCESS in 45m 19s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bdf7317c6fd64cc9a6cf0186bb960877 : SUCCESS in 1h 10m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4dd3b2c8561e4a5d84c2336a0b179d5c : SUCCESS in 45m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/21fe91dad43b422f9f06fc9e166f3ffc : SUCCESS in 57m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a71b7b7111da40daae5ea7df4f5949aa : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f54040275f547f59488042225369420 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6f24e3f1df8449ea993099a2455369d : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d65b0cbdbf3438981d2e88db645b02f : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7e0a2fbd3a94cfaa9995acaa5cc411e : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69192e8cc6454e94abb8be0ba53ba862 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b9813f49b0340e79f11cef6a0d587b6 : SUCCESS in 7m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/164fb7194d5344699023a02882422a49 : SUCCESS in 31m 24s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/00861afe313847f0a5ae9ca8dea303ac : SUCCESS in 4m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"49641bb25474ee8a71f0288c6724e03e13b13188","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-19 20:28:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"350dede7ba7de4fa2fe5e242de5dfc128d790265","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-20 16:02:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3976e328ed2301ae445f1322713066f11960b89a","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-20 16:02:49.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d4426ebe58909727610d1f3789d73ef0d082c3e7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-20 17:25:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"25963ab841ad76b87ba2b5b76fbc3bfffa3a2f8f","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-20 18:04:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"71576fa187b56a978475164853547b64232a682f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-20 18:04:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d4bb00218a523133d32fcc01b19ac1f7e4918ac0","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-20 19:03:53.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"dfab1445ca551af144b3b9bf1f042d1ebd4598cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 19:10:57.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2d025b9f690541f19caef2169e0d604a : SUCCESS in 37m 04s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/e655a64adee04db08bad4a1fae2324ea : SUCCESS in 59m 35s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e949a84563614c32bb08d387aef2e34d : SUCCESS in 48m 41s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/339a52f71ee64cf58deed1a5a6d9b8e5 : SUCCESS in 1h 00m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0245ecf917dd4e44828b447a8645f921 : SUCCESS in 45m 52s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c81bc031fb50498b82267fd85ca1896f : FAILURE in 51m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/321ee870b2c54e3588d9b59aea7daf60 : SUCCESS in 45m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/caf2266481bb4a64a81556aec9df85ba : SUCCESS in 34m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c684b027106947ff8f191046345099e2 : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7785c3c47ec545239b932de712feeca2 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/673ebf2278704b6fb7df267a340418c0 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a461895b9e5240e0a6e8a34dc8fdc7c9 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d2914ba3ec745798512da4f4c38339a : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/281fee59a05f4cf8b3b516f535b9a749 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a22c6a09b114e0688a170f066fb75ec : SUCCESS in 8m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/23dd4b87c22e416b8571c669be1ac7f5 : SUCCESS in 31m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5301468d082f4aad8ae4601f392de446 : SUCCESS in 5m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7485e3e983152f00fd54b9473efdf90e075e7766","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-20 20:19:27.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"dd7710784c169e8a26c02a83469c5e2f1cc90112","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 21:16:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3e80619c304b4538a8923584caae0afc : SUCCESS in 38m 05s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/e9f4c77901444c3a9cec0c0a8e51e19b : SUCCESS in 42m 29s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf5484e1259a40c58e8caccf23cc58d3 : SUCCESS in 36m 55s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/110ad20b8d3647989e82c025ed71063e : SUCCESS in 36m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0ee0c6b1fe094e3d998daace07aa4123 : SUCCESS in 31m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2ebb13637be44625b2a3a5b3b93651a8 : SUCCESS in 55m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9e887e53f888480c892b60263ea1b3d3 : SUCCESS in 32m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/87f6cd00b1b64f4a8f4aa2d5d2754e0c : SUCCESS in 32m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e82f1b7db9d84f74a01372b0d920f82e : SUCCESS in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/025fe2fe41144fc4b2c3ea527d5625b4 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d59880a98d364d56b34cb82a3ac83ba6 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecc1e5b1eec045b9923599fcdb6cf8ce : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8324cb70aeb14e14a867113c4c173c6f : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de288898a963453ab31fe5b95917b37a : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb40e23bf1534522afd75660de5791d2 : SUCCESS in 6m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a94cc2d98c0c44e98e84461634463ca4 : SUCCESS in 27m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3bfe80d92db643158e7c204a59293fd7 : SUCCESS in 4m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9bf72e11933f94d4ce10d78253dcc49e9db4321d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-25 19:43:30.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0c1304dde81dcbb3eda3955b915667af6ef9b531","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 19:44:28.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"30668945878ee7f15d7a7d15ac08207f826a84a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 20:58:03.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e7e7b3654ddb4de6b4aff508a22c9146 : SUCCESS in 33m 51s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/48dc9807ec4a452db5ba5893966c36c4 : SUCCESS in 41m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4014da50da2147de99aa2bd86dd42dbc : SUCCESS in 38m 38s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/50dd43526e2d4d448ce8cc7dcb3744e3 : SUCCESS in 1h 09m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/11e766083b85401587d44b19c94a65bc : SUCCESS in 40m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2301604a602840f797b154be3be17287 : SUCCESS in 40m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1e57ee9b9704390bfd9a5850dbd4333 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59d66fcfb7a5494c86bb39ffe64dd82e : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd24c01972b9493da53208335104e15e : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77e2a9410d0e4842b74b52f9a2892342 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ecba4ffeb4b4c868f8d5ee4b84bf4b3 : SUCCESS in 7m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/19a2f24f39254765a84f08caac5c824a : SUCCESS in 33m 30s","accounts_in_message":[],"_revision_number":6},{"id":"07ff8178f4d8d0d578d00bfd9680db608ef73913","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 20:58:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"c0e6740fe66fca5f15066cecdef0c23dbba8b34f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 20:59:27.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8c15ed5d3fd4499b8e6645127d8ed22 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"d705c5d6b861f319096afdd3ad50a2100651b064","revisions":{"e197f6b950b5294a29ec488586150aba10b99c1e":{"kind":"REWORK","_number":1,"created":"2022-03-29 15:34:54.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/51/835651/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/51/835651/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/835651/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/835651/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/835651/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/51/835651/1"}}},"commit":{"parents":[{"commit":"89864d458fcac3bb0399c0444adbb7906e1b2211","subject":"Add Python3 zed unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/89864d458fcac3bb0399c0444adbb7906e1b2211"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-03-29 15:09:29.000000000","tz":0},"committer":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-03-29 15:09:29.000000000","tz":0},"subject":"Clarifies the zone import error message","message":"Clarifies the zone import error message\n\nThis patch defines the situation where the exception handler produces\nthe message \u0027An undefined error occurred\u0027 to the exception during the\nzone import where an underlying exception occurres from the rpc dispatcher\nat the same time as a duplicate zone exception. Added an exception to\nclarify the error msg and excepton message to the rpc ExpectedException.\n\nCloses-bug: #1950118\nCloses-bug: #1964323\nChange-Id: If7d50cbd4fa0ce86e0ddf03068da36acd7d72cb4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e197f6b950b5294a29ec488586150aba10b99c1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e197f6b950b5294a29ec488586150aba10b99c1e"}]},"branch":"refs/heads/master"},"3fb3b5631fd02be9d1b404bcadd050fb72e7f7d9":{"kind":"REWORK","_number":2,"created":"2022-04-12 20:24:47.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/51/835651/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/51/835651/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/835651/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/835651/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/835651/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/51/835651/2"}}},"commit":{"parents":[{"commit":"89864d458fcac3bb0399c0444adbb7906e1b2211","subject":"Add Python3 zed unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/89864d458fcac3bb0399c0444adbb7906e1b2211"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-03-29 15:09:29.000000000","tz":0},"committer":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-04-12 20:18:35.000000000","tz":0},"subject":"Clarifies the zone import error message","message":"Clarifies the zone import error message\n\nThis patch defines the situation where the import_zone exception handler\nproduces the message \u0027An undefined error occurred\u0027 to the exception during\nthe zone import where an underlying exception occurred from the rpc\ndispatcher at the same time as a duplicate zone exception. This patch\nadds added exception interrogation into the exception\u0027s exc_info to clarify\nthe error msg and excepton message to the rpc.dispatcher\u0027s\nExpectedException.\n\nCloses-bug: #1950118\nCloses-bug: #1964323\nChange-Id: If7d50cbd4fa0ce86e0ddf03068da36acd7d72cb4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3fb3b5631fd02be9d1b404bcadd050fb72e7f7d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3fb3b5631fd02be9d1b404bcadd050fb72e7f7d9"}]},"branch":"refs/heads/master"},"9b422119b38a6078c44589a0f458d77d5e776f19":{"kind":"REWORK","_number":3,"created":"2022-04-14 15:04:06.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/51/835651/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/51/835651/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/835651/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/835651/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/835651/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/51/835651/3"}}},"commit":{"parents":[{"commit":"89864d458fcac3bb0399c0444adbb7906e1b2211","subject":"Add Python3 zed unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/89864d458fcac3bb0399c0444adbb7906e1b2211"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-03-29 15:09:29.000000000","tz":0},"committer":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-04-14 14:38:50.000000000","tz":0},"subject":"Clarifies the zone import error message","message":"Clarifies the zone import error message\n\nThis patch defines the situation where the import_zone exception handler\nproduces the message \u0027An undefined error occurred\u0027 to the exception during\nthe zone import where an underlying exception occurred from the rpc\ndispatcher at the same time as a duplicate zone exception. Kiall Mac Innes\nexplains this in\nhttps://opendev.org/openstack/designate/commit/2c9460505d07d0e46765a552b637e5a3296b667b\nthe problem here is in the way the _import_zone was written where\nhttps://github.com/openstack/designate/commit/9b809a11b3068552274340606eb76d2217411b85\nwas written with threading.local, so it only works if it stays within the same\nthread. Since _import_zone is created on a separate thread, the necessity to add\nthe decorator @rpc_expected_exceptions() at the top.\n\nCloses-bug: #1950118\nCloses-bug: #1964323\nChange-Id: If7d50cbd4fa0ce86e0ddf03068da36acd7d72cb4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9b422119b38a6078c44589a0f458d77d5e776f19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9b422119b38a6078c44589a0f458d77d5e776f19"}]},"branch":"refs/heads/master"},"2b39a386a7b49281007d3eefd88f3fcbdbb5cf53":{"kind":"REWORK","_number":4,"created":"2022-04-14 16:37:12.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/51/835651/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/51/835651/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/835651/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/835651/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/835651/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/51/835651/4"}}},"commit":{"parents":[{"commit":"89864d458fcac3bb0399c0444adbb7906e1b2211","subject":"Add Python3 zed unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/89864d458fcac3bb0399c0444adbb7906e1b2211"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-03-29 15:09:29.000000000","tz":0},"committer":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-04-14 16:25:23.000000000","tz":0},"subject":"Clarifies the zone import error message","message":"Clarifies the zone import error message\n\nThis patch defines the situation where the import_zone exception handler\nproduces the message \u0027An undefined error occurred\u0027 to the exception during\nthe zone import where an underlying exception occurred from the rpc\ndispatcher at the same time as a duplicate zone exception. Kiall Mac Innes\nexplains this in\nhttps://opendev.org/openstack/designate/commit/2c9460505d07d0e46765a552b637e5a3296b667b\nthe problem here is in the way the _import_zone was written where\nhttps://github.com/openstack/designate/commit/9b809a11b3068552274340606eb76d2217411b85\nwas written with threading.local, so it only works if it stays within the same\nthread. Since _import_zone is created on a separate thread, the necessity to add\nthe decorator @rpc_expected_exceptions() at the top.\n\nCloses-bug: #1950118\nCloses-bug: #1964323\nChange-Id: If7d50cbd4fa0ce86e0ddf03068da36acd7d72cb4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2b39a386a7b49281007d3eefd88f3fcbdbb5cf53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2b39a386a7b49281007d3eefd88f3fcbdbb5cf53"}]},"branch":"refs/heads/master"},"bbef41ce8a5506c91590bd8b8250755cee98ca4d":{"kind":"REWORK","_number":5,"created":"2022-04-18 20:25:47.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/51/835651/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/51/835651/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/835651/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/835651/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/835651/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/51/835651/5"}}},"commit":{"parents":[{"commit":"89864d458fcac3bb0399c0444adbb7906e1b2211","subject":"Add Python3 zed unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/89864d458fcac3bb0399c0444adbb7906e1b2211"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-03-29 15:09:29.000000000","tz":0},"committer":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-04-18 20:25:35.000000000","tz":0},"subject":"Clarifies the zone import error message","message":"Clarifies the zone import error message\n\nThis patch defines the situation where the import_zone exception handler\nproduces the message \u0027An undefined error occurred\u0027 to the exception during\nthe zone import where an underlying exception occurred from the rpc\ndispatcher at the same time as a duplicate zone exception. Kiall Mac Innes\nexplains this in\nhttps://opendev.org/openstack/designate/commit/2c9460505d07d0e46765a552b637e5a3296b667b\nthe problem here is in the way the _import_zone was written where\nhttps://github.com/openstack/designate/commit/9b809a11b3068552274340606eb76d2217411b85\nwas written with threading.local, so it only works if it stays within the same\nthread. Since _import_zone is created on a separate thread, the necessity to add\nthe decorator @rpc_expected_exceptions() at the top.\n\nCloses-bug: #1950118\nCloses-bug: #1964323\nChange-Id: If7d50cbd4fa0ce86e0ddf03068da36acd7d72cb4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bbef41ce8a5506c91590bd8b8250755cee98ca4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bbef41ce8a5506c91590bd8b8250755cee98ca4d"}]},"branch":"refs/heads/master"},"d705c5d6b861f319096afdd3ad50a2100651b064":{"kind":"REWORK","_number":6,"created":"2022-04-20 18:04:31.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/51/835651/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/51/835651/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/835651/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/835651/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/835651/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/51/835651/6"}}},"commit":{"parents":[{"commit":"89864d458fcac3bb0399c0444adbb7906e1b2211","subject":"Add Python3 zed unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/89864d458fcac3bb0399c0444adbb7906e1b2211"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-03-29 15:09:29.000000000","tz":0},"committer":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-04-20 18:01:11.000000000","tz":0},"subject":"Clarifies the zone import error message","message":"Clarifies the zone import error message\n\nThis patch defines the situation where the import_zone exception handler\nproduces the message \u0027An undefined error occurred\u0027 to the exception during\nthe zone import where an underlying exception occurred from the rpc\ndispatcher at the same time as a duplicate zone exception. Kiall Mac Innes\nexplains this in\nhttps://opendev.org/openstack/designate/commit/2c9460505d07d0e46765a552b637e5a3296b667b\nthe problem here is in the way the _import_zone was written where\nhttps://github.com/openstack/designate/commit/9b809a11b3068552274340606eb76d2217411b85\nwas written with threading.local, so it only works if it stays within the same\nthread. Since _import_zone is created on a separate thread, the necessity to add\nthe decorator @rpc_expected_exceptions() at the top.\n\nCloses-bug: #1950118\nCloses-bug: #1964323\nChange-Id: If7d50cbd4fa0ce86e0ddf03068da36acd7d72cb4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d705c5d6b861f319096afdd3ad50a2100651b064"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d705c5d6b861f319096afdd3ad50a2100651b064"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
