)]}'
{"id":"openstack%2Fdesignate~836158","triplet_id":"openstack%2Fdesignate~stable%2Fwallaby~I55ca76ef7c2cbeb5fdae1aed1dcbe58b7acddc34","project":"openstack/designate","branch":"stable/wallaby","topic":"bug/1966128-stable/yoga-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-06-04 01:56:21.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"I55ca76ef7c2cbeb5fdae1aed1dcbe58b7acddc34","subject":"Fix set-quotas for non-project scoped tokens","status":"MERGED","created":"2022-04-01 22:27:30.000000000","updated":"2022-06-04 03:00:08.000000000","submitted":"2022-06-04 02:58:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836158-bug/1966128-stable/yoga-stable/xena-stable/wallaby","cherry_pick_of_change":836157,"cherry_pick_of_patch_set":1,"meta_rev_id":"baaae099f19f132cd7e62f3eb445436249fb1dd3","_number":836158,"virtual_id_number":836158,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-04 02:58:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-06-04 01:56:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-04 03:00:08.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":2,"date":"2022-06-04 01:40:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2022-06-04 01:56:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-01 23:25:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-02 04:22:19.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2022-05-12 06:09:38.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-06-04 01:40:27.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"d058c0586912b12eeb664609dc1e3ceece5ac1f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-01 22:27:30.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"f083248da23732d5e78ef7e78a805f4bd466a8d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 23:25:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6a1ba05234e045e9b12118ed6b47fb96 : SUCCESS in 42m 30s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/cfeb4602dc75413680ab5d8ea592c3ed : FAILURE in 14m 00s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/10c04c9e0b22497fa4164e7dc24e1fb7 : SUCCESS in 39m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/790875d63ecf482b8662880170277b69 : SUCCESS in 39m 13s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/df69ed1b109d47ebb5ddf3eda364b5fa : SUCCESS in 56m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/191bcf806af746f8b52b26702c0d9502 : SUCCESS in 38m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/de94c51b68394f6eb7fe2ae7c5f5618e : SUCCESS in 46m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58cade85c29e4975a011f3195c16be5a : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4523e546f2b4ae8a95562fa0d47cb28 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3888f216cf042cdba4607e4a7d8dddd : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ba949539bd74dac9ac0d3747f388c63 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7fcf55273e940248e69c7b77dea163a : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3519647a6154309b8a07d078641f39e : FAILURE in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f395a2b0182f45c9bdeed5e6a8ea125a : SUCCESS in 4m 01s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/b58e8455f8ea42a8afdc1d50c59eecc8 : SUCCESS in 27m 28s","accounts_in_message":[],"_revision_number":1},{"id":"9f8023aa3673d95ac980bd2cd3eae85b983a8952","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-02 04:22:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"04e8b395a4db60035eb82fd054c69842e38309d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-02 05:08:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d1d9d21f45fe4e88be6dc47969992b96 : SUCCESS in 34m 01s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/11c0b9cd503849d1a3fb26e84525d7ad : FAILURE in 12m 00s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/a64efe3d462143a8a2bd05c330ac0840 : SUCCESS in 30m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9dc9f287f8b642dfb031615c100e4527 : SUCCESS in 28m 27s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/643dade798204844b2290acd2d316769 : SUCCESS in 43m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e12c501cef6046d291e54335eff317c6 : SUCCESS in 33m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/212dd09d9e024014b60b31434a914d99 : SUCCESS in 30m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c77da9d02a764e679efffaa4a34f51a9 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bf4f3dd2daa4841afc112ca364afae8 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a72e9a1d09854351a68d8133a10cef8c : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/465668274a4b4303b9fc7f873cfe7087 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c86e20f61fa54a6998c6a15b2f0ee6bb : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c72ae06e3b5e48c482590504f408111a : FAILURE in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76749f3cf24d45e8ab9f1fe3a0622d93 : SUCCESS in 3m 44s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/c19a5847950b43018cfb72a1796b9c3c : SUCCESS in 30m 07s","accounts_in_message":[],"_revision_number":1},{"id":"04edc6d33d3d399f80a646cf493598dd53d78d4d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-03 02:50:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"88a1bb4cd1cbd8b4f720048adccc056b91586612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-03 03:59:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/48e374aa89f34d39b40ecf22c4b17aa8 : SUCCESS in 48m 16s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/0f2944be457f482cb82b22c61dfd694f : SUCCESS in 46m 10s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3b9c31eda7674654bda41c7fa1ad34df : SUCCESS in 46m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/117a0f72ef4a455189cc6ef37385132e : SUCCESS in 49m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3cbd0f2ba08242369608daf1dcb3ee6a : SUCCESS in 1h 04m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/de9d82a6a3284232bf516ad603350348 : SUCCESS in 41m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cf1676ab3f5d4fd48da1f9ef80103363 : SUCCESS in 46m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/653de865316045d181180286ce34f273 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e49ab31ccaf048de9c13cb49bb580ba9 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ea8f567e76b448db04bf150aaa0e1a1 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4822153f7420424a9ceebb7cdf3c06ba : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f961212535d4bdaa72aa08932c57cf5 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d9848f1eb414e6aa6c529efe81ee748 : FAILURE in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3913efb99612418380289241c7a7e460 : SUCCESS in 4m 02s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/3b4ab0a538a84f6694cd622ab23682e7 : SUCCESS in 29m 36s","accounts_in_message":[],"_revision_number":1},{"id":"c0f74279f5de1568df11618a68806ee4a4afd016","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-04 17:42:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f1338fe33df9d82dac119fd3e9305608efe6c0d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 18:44:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3ddfebedf19a4c2390797302d695b2e9 : SUCCESS in 50m 08s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/1bdaf12c836047d2896187859c2fbd90 : SUCCESS in 55m 02s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/acddb3b3c46e40be98663fd2ea87335b : SUCCESS in 43m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2add63a92f7044dbb122defc74e9c47a : SUCCESS in 45m 33s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7fbbc52526994d229538b7956720976d : SUCCESS in 56m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7cac602b73654a65a4ce2f533b2cca0c : SUCCESS in 47m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2a568b94d3c54f0ab86d68619829bd72 : SUCCESS in 41m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/755499deb75a4421ae702bd5d70cbb14 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/565901ba7e4f4c7abd65003235fed6aa : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b74f99a42b4340499ff879f636131791 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a18cc355b9ec4c54a452406e14173dc5 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e3390f72dea4c15a69b470e2fef5c37 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64bded7be27c4b8e9c0a1c7c6ee87ace : FAILURE in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c89258bbc3e5407f9149589c2498bddb : SUCCESS in 5m 01s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/b790187e2a4a491d803e222fd9c63637 : SUCCESS in 44m 31s","accounts_in_message":[],"_revision_number":1},{"id":"a55bc922dc61fc90d729272a2c31beba266b8e86","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-05-12 06:09:38.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e61eff9d2f767e9462855e5b49aa0f6a09dfa24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 07:07:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8fd38d05c0b34dac9f22259be4953a5c : SUCCESS in 37m 37s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/2e3f15a210aa4b21b837758c70b80f28 : SUCCESS in 56m 02s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f03ffe3ca2b04bdbbb0e1dfe3d50a845 : SUCCESS in 41m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/86bdd06cb23f4cf7a0da13eb65b62ee1 : FAILURE in 42m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e256f68ecdfc4fce81246295ae134378 : SUCCESS in 56m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/25040afb39024cfe98e73a4ddc6ae890 : FAILURE in 50m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b5828e3f62fd4ae1809ac83bcdabe16a : SUCCESS in 38m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02a1c3d8de2a418a90314c61b2797c31 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0db9c5c74ea4f14b6af925b725febdb : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ec20cbb2d164a73b54056f52d5ded53 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7383069085064ebc8e40bb4159189545 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf813182e401472aa5271eb8108adfce : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df05d2bc3915462b9a1135b98cfe1328 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1c4c301750f4ddeb5e4a9364d4d6c5e : SUCCESS in 4m 40s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/e0311d880a804a0f8aac77b49a4c3d71 : SUCCESS in 32m 14s","accounts_in_message":[],"_revision_number":1},{"id":"9fb9d5d18627e173d7051ba83a4c883b8e02cd0e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-28 00:39:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a45713123f35a48fdaac8a6ecc671f369340d936","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 01:25:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a6bdfba9668d423b807d031c643eeed5 : SUCCESS in 32m 25s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/fc8c49f53ce34a0a8c5056fe743ca886 : FAILURE in 16m 58s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/62853fff22264524a07c96ad9595ba8d : SUCCESS in 31m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/94124a6a46304efa8ca45683a9345e4f : SUCCESS in 30m 33s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/30ddb348d5704477b755ce15bff9472f : FAILURE in 44m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1783313bc86e42e7b55e68598d4b73c4 : SUCCESS in 32m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3022a926310b4dee8ddd0a73fbbcca58 : SUCCESS in 32m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e506809df19045f892043663e4a236c4 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df6249861c144268bd0e1787e3ff5277 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a5f16626e1f4630b6ed339281a100ce : FAILURE in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19ac7e2f1207485abb52548010e2980a : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1670fa46d4b5445ab5432dd23f4800da : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e9067cbbf164c20bf7a61a77cb5dafb : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/139c908455df4778a9c806cbe2555633 : SUCCESS in 3m 29s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/1abfddeb87f34d25ac47018efd53793e : SUCCESS in 25m 01s","accounts_in_message":[],"_revision_number":1},{"id":"46dd4b6c208582a2a46bd57eebc3b16ebec73786","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-31 22:36:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ab6203adda2d0376dbc1e86b5699dbb7165ff04d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 23:27:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b33c79cf0dfc4ef6a3392a96124dcb55 : SUCCESS in 32m 26s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/f305130288374f088ad182f90abe0649 : FAILURE in 16m 34s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/5308f647bb56446e83f631bdb6501f17 : SUCCESS in 33m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3f280884e42a41f8bd546801a8a548ee : SUCCESS in 31m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/007476d37dc24feb99a656b239b34317 : FAILURE in 47m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bb4320229a984c778a3c4270bd90b1c9 : SUCCESS in 34m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4687ea193c234565913204cb548cd893 : SUCCESS in 32m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/383f443da1814a0ea434dd0a593231ae : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f813d11e38cd43189b8e470e14d04542 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4971801422a54dc384bc6627e7c7bd57 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47e137d4edea4df88ae92037b8897912 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b214ecd4fba4f3ebaa756563e466859 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25c18c3c6a0a4153a98f75792449813b : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50ab0f90ba04468fa8918869979a908b : SUCCESS in 3m 32s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/615eb1a4006a471daa155f9da2ac83ea : SUCCESS in 27m 38s","accounts_in_message":[],"_revision_number":2},{"id":"2183fe1bb84849c4193a21fe94e24140a4b81a5f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-03 18:08:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b783bccbbc265b6fcc75dd2bbbaa5ee451ec345a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 19:01:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/944f52de364842fdbe2fdf4dc90d6c8d : SUCCESS in 43m 01s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/babdeab720264878add839c551c3ee40 : FAILURE in 32m 30s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/675abadf9092423cbe075446d75a223d : SUCCESS in 43m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4f2d1b2b9f334976beac4c111db89052 : SUCCESS in 46m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a4059d89e4be45ff9f991158b9ad6ae7 : SUCCESS in 47m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a233cfbebd1348378ef1c2d35228a633 : SUCCESS in 47m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4173159226394d118e704593191cc813 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86ee5c7386224dc8b3af9b8066ae8ef7 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/332eb6fc19c949f3851e77f8d88001fc : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c46025c49eee4246816bd1bc7e4b5543 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7bce92f3226a41b4b08397f1b1cc8805 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63726ce63ffd4b9ea8c1fc78fc694a2b : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0526892a74634f5d982f7359897270eb : SUCCESS in 3m 32s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/dda9be7c19124754b7737a2d5bd3de17 : SUCCESS in 28m 00s","accounts_in_message":[],"_revision_number":2},{"id":"53b5c65396fbd30368396814fec5eda727e1f48e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-04 01:40:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a4c03024fc113970c130ac77be97d8ea5653ed77","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-04 01:56:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e7df68c0286dd8245cc722c136203969a8b6d0c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 01:56:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2a0ad404544851b3ae430db73a85fc410c74e6e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 02:58:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/76655d15ebe241a69b861592b2f0c840 : SUCCESS in 49m 49s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/94a8f22466bd4033a5a80b0e7e08aa22 : SUCCESS in 50m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/84eef40b42b64c2e9053d3a7a43c2be5 : SUCCESS in 37m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/253602dbfce2492b92acfcdbee7fdc96 : SUCCESS in 43m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/14ddf1d212d24b559a5fd91aa7c1be6d : SUCCESS in 55m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/014b5ffad141405a8b3708c919692d29 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa63c2cb528c4443ab381430fa97a539 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ad60aa34f9e4f06b69bbef9ce3120a5 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b3335076b204f889b67552e6a4faeca : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e59a56d89eb94c8991037b7e960f9305 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"83741a411915debdb948b2331a32fbf7fb1691fb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 02:58:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d6d112c220d5ffc4f61863e5c3f61e514dc8a592","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 03:00:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfd1bd0f33114971ab6467ca9045befa : SUCCESS in 1m 33s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddeb944ae0264a1b96decc15756a2188 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ea85d917a174f44269cf2d4b72ef57e14e785670","revisions":{"11c2a8af1855af4c63bc14ffd8b65664e257f0dd":{"kind":"REWORK","_number":1,"created":"2022-04-01 22:27:30.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/58/836158/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/58/836158/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/58/836158/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/58/836158/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/58/836158/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/58/836158/1"}}},"commit":{"parents":[{"commit":"0323c4244929aaf199ecbb7c8d32138764029cc2","subject":"Allow TXT record over 255 characters if split","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0323c4244929aaf199ecbb7c8d32138764029cc2"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-03-23 20:52:35.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-01 22:27:30.000000000","tz":0},"subject":"Fix set-quotas for non-project scoped tokens","message":"Fix set-quotas for non-project scoped tokens\n\nPreviously, if set-quotas was called with a non-project scoped token and the all-projects flag was not set, the quotas would be updated but the result returned\nwould always be the default quota values.\nThis patch changes the API to require the all-projects flag when set-quota is called and the token is not project scoped.\n\nCloses-Bug: #1966128\nChange-Id: I55ca76ef7c2cbeb5fdae1aed1dcbe58b7acddc34\n(cherry picked from commit 158e017be489eea5402ef258ea260183067bf1ca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/11c2a8af1855af4c63bc14ffd8b65664e257f0dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/11c2a8af1855af4c63bc14ffd8b65664e257f0dd"}]},"branch":"refs/heads/stable/wallaby"},"ea85d917a174f44269cf2d4b72ef57e14e785670":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-05-31 22:36:52.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/58/836158/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/58/836158/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/58/836158/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/58/836158/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/58/836158/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/58/836158/2"}}},"commit":{"parents":[{"commit":"2e55c3e24bfaf1b03241aa1a4a538d03c29d69ad","subject":"Improve quota API validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2e55c3e24bfaf1b03241aa1a4a538d03c29d69ad"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-03-23 20:52:35.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-31 22:36:52.000000000","tz":0},"subject":"Fix set-quotas for non-project scoped tokens","message":"Fix set-quotas for non-project scoped tokens\n\nPreviously, if set-quotas was called with a non-project scoped token and the all-projects flag was not set, the quotas would be updated but the result returned\nwould always be the default quota values.\nThis patch changes the API to require the all-projects flag when set-quota is called and the token is not project scoped.\n\nCloses-Bug: #1966128\nChange-Id: I55ca76ef7c2cbeb5fdae1aed1dcbe58b7acddc34\n(cherry picked from commit 158e017be489eea5402ef258ea260183067bf1ca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ea85d917a174f44269cf2d4b72ef57e14e785670"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ea85d917a174f44269cf2d4b72ef57e14e785670"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
