)]}'
{"id":"openstack%2Fdesignate~836217","triplet_id":"openstack%2Fdesignate~master~I4522796ac056630c63798830e8f8d92628f14a37","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-04-02 04:14:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4522796ac056630c63798830e8f8d92628f14a37","subject":"Fix delete zone transfer request with scoped token","status":"MERGED","created":"2022-04-01 22:26:18.000000000","updated":"2022-04-02 04:15:57.000000000","submitted":"2022-04-02 04:14:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836217","meta_rev_id":"bd83a1abcffbba7e8031b7ea25a772ffdc37e5eb","_number":836217,"virtual_id_number":836217,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-02 04:14:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-01 22:59:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-02 03:11:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-01 22:59:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-01 22:59:04.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-04-01 23:21:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"347d08455396b72ce6d27c6614c1ed64277b8560","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-01 22:26:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e926aa578851cd51fa4a1c9e8a06e3c6bdf43f6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-01 22:59:04.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"e427f364b0765c41e15562381ae4922fd3bac925","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 23:21:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c616686cdb47443d9d9449052c482b7d : SUCCESS in 36m 47s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/1a7616c5df89450d905b768ebb7540ec : FAILURE in 12m 57s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a645df2dd874138b8d12cfd47dd125c : SUCCESS in 35m 31s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/0b8f23155c304538b48e8644a3bf5d1a : SUCCESS in 39m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9dc896437da6418d9a1bff90a115e507 : SUCCESS in 37m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7efb523584c2475097a4531e670a80c8 : SUCCESS in 53m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bcca7fe9a80f4cd5a49fafccb263efd5 : SUCCESS in 36m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5261378588c24213aba7855bd0c6f4af : SUCCESS in 39m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d3640d4de6c4781a08258e2e1c8885e : SUCCESS in 6m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83e58aa61bda42e49f59cbbd4c3985c4 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b125a4f52a6046a391ee7dc1d7604e97 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85699ca6178f4ce49922cb142438cc51 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa7f97a3b20c4ad88c7a9121b5830186 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f791edd834f4a259ce545189638cb86 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc15488a58924c9f84028d49b1d0bd41 : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e564d8a61804489bf37fb2270d4bb78 : SUCCESS in 4m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cc894a26cd234d74bf770c0f366895c5 : SUCCESS in 27m 46s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b02ec06bff9945e8b57116d87cde1897 : SUCCESS in 4m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"42e3f6c053df624bf1a23afb5233a40466a89eb8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-02 03:11:14.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f2af97e024934827cfa90b093d0fa7dcd3b4383a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-02 03:12:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7400177fe7b0fef77cdc59a3b44b5ad6d66a432a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-02 04:14:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/48cf42711d404c329900fa8f52a76017 : SUCCESS in 50m 05s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/b226c646cb5849cd88023c3cc08335f3 : SUCCESS in 44m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1a5f1f96985a4268936c6135aa90357e : SUCCESS in 44m 13s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/07713f43045e44149812abfb04539bc4 : SUCCESS in 57m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d1f00d410c8e435caa758a846b948e31 : SUCCESS in 41m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/118bc125694e4603a548739c7ddd88d3 : SUCCESS in 42m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b65e1de59c14be9afc3c73fcbe7a157 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f42ba52f209a42ec837027f58e4db3a0 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97eb90764fd14056ad89f10bc1cf41ac : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea942f1632124b2481a55e167c649d43 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ce2f52ebe55493dae31ebd99ea83126 : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40625fc51601489792862f7ceb29ca67 : SUCCESS in 5m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c70c4edcb11649aa8ef6511fbc579147 : SUCCESS in 31m 17s","accounts_in_message":[],"_revision_number":1},{"id":"7d72718ff45813352dd96b59dfede778ce902a8e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-02 04:14:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bd83a1abcffbba7e8031b7ea25a772ffdc37e5eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-02 04:15:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54230876d33849a6b43f7429f0eed46c : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca0e6e9be6a249f1b61b2457042775a7 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6f847aeb5a309eb6e70dee58057fef01886573f6","revisions":{"6f847aeb5a309eb6e70dee58057fef01886573f6":{"kind":"REWORK","_number":1,"created":"2022-04-01 22:26:18.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/836217/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/836217/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/836217/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/836217/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/836217/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/836217/1"}}},"commit":{"parents":[{"commit":"dc4ce8a1c16d91d77b8765d8b293334314428806","subject":"Merge \"Improve quota object coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc4ce8a1c16d91d77b8765d8b293334314428806"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-01 22:25:07.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-01 22:26:17.000000000","tz":0},"subject":"Fix delete zone transfer request with scoped token","message":"Fix delete zone transfer request with scoped token\n\nThis patch fixes an issue were deleting a zone transfer request with a scoped\ntoken may fail.\n\nChange-Id: I4522796ac056630c63798830e8f8d92628f14a37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f847aeb5a309eb6e70dee58057fef01886573f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f847aeb5a309eb6e70dee58057fef01886573f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
