)]}'
{"id":"openstack%2Fdesignate~836238","triplet_id":"openstack%2Fdesignate~master~I172e628e52c57c0406237181c8543a982808b643","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-04-04 17:42:03.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I172e628e52c57c0406237181c8543a982808b643","subject":"Move Import Zone to worker","status":"ABANDONED","created":"2022-04-03 04:39:57.000000000","updated":"2022-04-04 17:42:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"99cb31dba4f5794c5d1bf30e79a66c13f6d15c30","_number":836238,"virtual_id_number":836238,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-04-03 20:35:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-04-03 05:46:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"23652994a9107848b9ddf7d0f694c04d018fd6f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-03 04:39:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eaf47a4fb4a5064e8ce4782b669bcfe1c3239095","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-03 04:40:11.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"ef1d77b65700a85e690832505353843cdd10394f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-03 05:46:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/63ebe5ee60bf4854b0475c7e148bbbfd : FAILURE in 36m 25s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/fe23b29d35364a30bbac8f4e04846e18 : FAILURE in 1h 02m 17s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/91306958bb4d427fbe6111c30862fc75 : FAILURE in 43m 11s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8623a3c2c75d446085d1123e3b7dcb6e : FAILURE in 43m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/945a5cc890264d4193ec3a0ab2e3cd46 : FAILURE in 39m 22s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/56408ba20788489ab5d94bc95ae38da1 : FAILURE in 49m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d3f49fc27a4e4eba904c0ec86348e55b : FAILURE in 38m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4e2a9d847b504612b562f0038b04faf9 : FAILURE in 39m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cde51ef95a74b8db6831c1986e98d6c : FAILURE in 5m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/460dee977d03459bb0e1e394b4bca858 : FAILURE in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49f46d07e6fd42a1b7658de5a3a62548 : FAILURE in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58c400f255bf411cb22c7ecc52446caa : FAILURE in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fce17c3a4774d01ad14508cad86ce09 : FAILURE in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10598482e54346dfbd160f79eb5bd19f : FAILURE in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5abee9cf81e4413fb2dad044c7d9269c : SUCCESS in 8m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/349fce666600454b82a444295ef604cf : SUCCESS in 24m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/35782727dbfd49e9b24192b8482020bf : FAILURE in 5m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc6ebaee4f444491cf77872136f31722b3dc8336","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-03 05:47:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"491dff03718eec38536da634f98a2e3acfeacf35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-03 07:05:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/de8b1a5a21c64ef78a3e5d8948b8fdfd : FAILURE in 55m 16s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/219997293c914485a4989401cc834370 : FAILURE in 1h 05m 37s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab7da17f1e2d4269b01023a56167597c : FAILURE in 39m 28s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3a8d67bb017b4502986fda3d995cf7c6 : FAILURE in 36m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/093d773dbc1044d2a80e3f8fd5fff9ec : FAILURE in 58m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fd75de9a92d44d64828a4b9c1e37b1bd : FAILURE in 1h 11m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5e5c8f72c2ab43f59b05d0600d4ea7d8 : FAILURE in 58m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c8725ec14f54433db4ddef34bef02be7 : FAILURE in 57m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e40c9a98c3ae4302a3e7a1fd035df686 : FAILURE in 5m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c4ad11fb3704847b27728a07d743aa5 : FAILURE in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbcbcf88a18543379e6bc5904eabce03 : FAILURE in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dff5ca4eddb0444eaa72548bae5d1faf : FAILURE in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6dc6d26e56eb4f31a445ca2a18d223e0 : FAILURE in 6m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fda2d821f1364da7b513b20f2cee7634 : FAILURE in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cbf75abf64f49a8808d9075f963be94 : SUCCESS in 8m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/000f2d38a97c4291bea70f23211fdffa : SUCCESS in 32m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/80b516e3901b41968ceacecb96d7c89b : FAILURE in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a6b66503d667ba653aafd136e2c7dcb23df290d5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-03 19:17:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"486d621b130b323d8d923ce2361db8ec6727136a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-03 19:18:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f7763f6fc2db5c9da003d339e53902a18715d2ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-03 20:35:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8772dac83d3047e0b11aff9cb7f66865 : SUCCESS in 46m 38s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/cfc0ee604e114e709ce679fe5efe447b : SUCCESS in 59m 25s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7416ccc7ed9a43a0aa6ff4ad772dd5d5 : SUCCESS in 36m 38s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4e6a7bbb36e44b3b8344729565e3675c : SUCCESS in 43m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7803d1c7a2ce4086a144e696b84045e6 : SUCCESS in 42m 33s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/971ecb1e090d47afba35c387d418d2a3 : SUCCESS in 1h 04m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c23ceba7daf54d0b8854be56507c27d5 : SUCCESS in 42m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/649d76bfb41642dd8b2d3b38c1b2f710 : SUCCESS in 43m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cedb4508fe004afc9d5dbc76524ab242 : FAILURE in 4m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43873517ff2d43d4b357268551d28958 : FAILURE in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/618a1e8305384eb2ba74ea0d131ca5f7 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a607c3a40aed4fa99ce81a268c5b7fb5 : FAILURE in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b4ddf6dec7744eda6566daf1d767a4c : FAILURE in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89f197319dde4277969ffab23fa6d1a8 : FAILURE in 8m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd7c3803d2e84a51aa7126bb3b742289 : SUCCESS in 6m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b7d1f6625a2d4adfa6afa5375b5642d9 : SUCCESS in 24m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b26925a310ed4022afafe18537d5d954 : FAILURE in 4m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"99cb31dba4f5794c5d1bf30e79a66c13f6d15c30","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-04 17:42:03.000000000","message":"Abandoned\n\nCould not figure out how to make the unit tests to work","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3cfe7b559d40749d4cbe632e602c6b14ea966652","revisions":{"e1d605fc537e0a4af57248cabc5ee538ebf7bab4":{"kind":"REWORK","_number":1,"created":"2022-04-03 04:39:57.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/38/836238/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/836238/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/836238/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/836238/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/836238/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/836238/1"}}},"commit":{"parents":[{"commit":"dc4ce8a1c16d91d77b8765d8b293334314428806","subject":"Merge \"Improve quota object coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc4ce8a1c16d91d77b8765d8b293334314428806"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-03 04:38:43.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-03 04:39:45.000000000","tz":-420},"subject":"Move Import Zone to worker","message":"Move Import Zone to worker\n\nChange-Id: I172e628e52c57c0406237181c8543a982808b643\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e1d605fc537e0a4af57248cabc5ee538ebf7bab4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e1d605fc537e0a4af57248cabc5ee538ebf7bab4"}]},"branch":"refs/heads/master"},"72560b9849cdc77c5453e24c5511f464d8ce5ca9":{"kind":"REWORK","_number":2,"created":"2022-04-03 05:47:17.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/38/836238/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/836238/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/836238/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/836238/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/836238/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/836238/2"}}},"commit":{"parents":[{"commit":"dc4ce8a1c16d91d77b8765d8b293334314428806","subject":"Merge \"Improve quota object coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc4ce8a1c16d91d77b8765d8b293334314428806"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-03 04:38:43.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-03 05:47:10.000000000","tz":-420},"subject":"Move Import Zone to worker","message":"Move Import Zone to worker\n\nChange-Id: I172e628e52c57c0406237181c8543a982808b643\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/72560b9849cdc77c5453e24c5511f464d8ce5ca9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/72560b9849cdc77c5453e24c5511f464d8ce5ca9"}]},"branch":"refs/heads/master"},"716e8f15dac5b56abbd2c84ec5d49665091c4c06":{"kind":"REWORK","_number":3,"created":"2022-04-03 19:17:36.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/38/836238/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/836238/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/836238/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/836238/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/836238/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/836238/3"}}},"commit":{"parents":[{"commit":"dc4ce8a1c16d91d77b8765d8b293334314428806","subject":"Merge \"Improve quota object coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc4ce8a1c16d91d77b8765d8b293334314428806"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-03 04:38:43.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-03 19:17:27.000000000","tz":-420},"subject":"Move Import Zone to worker","message":"Move Import Zone to worker\n\nChange-Id: I172e628e52c57c0406237181c8543a982808b643\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/716e8f15dac5b56abbd2c84ec5d49665091c4c06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/716e8f15dac5b56abbd2c84ec5d49665091c4c06"}]},"branch":"refs/heads/master"},"3cfe7b559d40749d4cbe632e602c6b14ea966652":{"kind":"REWORK","_number":4,"created":"2022-04-03 19:18:47.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/38/836238/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/836238/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/836238/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/836238/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/836238/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/836238/4"}}},"commit":{"parents":[{"commit":"dc4ce8a1c16d91d77b8765d8b293334314428806","subject":"Merge \"Improve quota object coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc4ce8a1c16d91d77b8765d8b293334314428806"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-03 04:38:43.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-03 19:18:38.000000000","tz":-420},"subject":"Move Import Zone to worker","message":"Move Import Zone to worker\n\nChange-Id: I172e628e52c57c0406237181c8543a982808b643\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3cfe7b559d40749d4cbe632e602c6b14ea966652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3cfe7b559d40749d4cbe632e602c6b14ea966652"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
