)]}'
{"id":"openstack%2Fdesignate~836410","triplet_id":"openstack%2Fdesignate~master~Ie49f99ba8e86913471ee6cd078f5afd28282c0f5","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-04-05 16:47:08.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"Ie49f99ba8e86913471ee6cd078f5afd28282c0f5","subject":"Fix tox docs env to have the correct dependencies","status":"MERGED","created":"2022-04-04 22:21:53.000000000","updated":"2022-04-05 18:44:19.000000000","submitted":"2022-04-05 18:42:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836410","meta_rev_id":"298ec47d9445e63633ccdde9dc2d327aa3adf125","_number":836410,"virtual_id_number":836410,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-05 18:42:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-04-05 14:19:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-04 22:26:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-04 23:14:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2022-04-04 22:23:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-04 22:26:28.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-04-04 23:27:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-05 14:19:53.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"f396d5e1c1b2ef83d45b56b4eb4a280dc8426cda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-04 22:21:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bda565eaa9a64fac862d178cd1c287b3e3b2f1b5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-04 22:23:08.000000000","message":"Patch Set 1: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c488275c6b5374a7af5a6cac3833f8f1fd6ec130","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-04 22:26:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"89989b2c371b54c9f9d3a988c1e0fd56278b5954","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-04 23:14:45.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"43b8f4097903dead2875a443427dab6cc807ba18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 23:27:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/605b20fa9a1a4f7487b72be1dc9506ac : SUCCESS in 4m 08s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3d36383f759a41a99fce20610cfd4875 : SUCCESS in 55m 10s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/40dafd46c1b741109253ec0ce8ef1b7a : SUCCESS in 1h 00m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/78baffc34d304f13ab74728ee290e726 : SUCCESS in 33m 24s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/0482bc0652db4b05afd2eb1e72496217 : SUCCESS in 34m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a8e5935292974124b2d6895df5a323b8 : FAILURE in 33m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f4f3f92469c241a989ba1d5dd69be81d : SUCCESS in 45m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a2529fb5e42142108859c0816dab7fee : SUCCESS in 41m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/92e7a1bffb7e44b1b0865ba57aab619e : SUCCESS in 34m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6628a0a77dfd4ff198d72220b1e089b9 : SUCCESS in 5m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17663979063447048315e0e18938e49f : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4385387fbe2d4521800299ed63d6eb7b : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b497ecdb5fed428a848737c81c0d0477 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c928fbc82cf44fe48d22e15118025c59 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fcafb966b4674e828f7a0a1507fbeb69 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f255857bbaa484ebf99013ead088af5 : SUCCESS in 8m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75ef6fef908b46ff914f74ed9a9afe2b : SUCCESS in 3m 32s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/acfeed732d2b45b98fd35e0c948da986 : SUCCESS in 5m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c58aade41b4cf12f3f2ea1f4dd9045fc8df56036","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-04 23:33:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"37dd44c5b0f82aa26c8cd4782c49762da37094a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 00:39:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/45555963a892455ea43222d2331a4f21 : SUCCESS in 4m 29s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/07d2f6d70f514f3da8b06feeeb2d68df : SUCCESS in 45m 19s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/d3aefdf583774c2f952480a2ce6c191e : SUCCESS in 58m 35s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/661ad0b984ce4d78805a98b43e1625a2 : SUCCESS in 50m 26s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/31c4af4e455545389fc0acc9731358e0 : SUCCESS in 46m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9901016b48f74786ba1a1085d43b6720 : SUCCESS in 41m 56s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/95bcf5116ed04c2c8c9795923bd774cd : SUCCESS in 1h 00m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1cc07a2c6f7b40eca62140e11a7f5d34 : SUCCESS in 45m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6d920c7b549643069e75d6dca8bbfbf7 : SUCCESS in 47m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4524e5d0baa44988d340c38276c0dd5 : SUCCESS in 5m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fde8547a291842c4868120eed1ecfb49 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcc875ac2c6d4081a29d7ff2f3222ea3 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93957cc3ee564af399624dd132de98ac : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2aeadff7140443c97b5ede99747d73d : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d8d7278fc6b4b55837328a5c9749c1e : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c3454a62eb64e09a4816bf5db3d5abc : SUCCESS in 8m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/169aeb9b780e44fb989ade9a2cb7c8c2 : SUCCESS in 3m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0947acae561a467eb0003865e377a9a2 : SUCCESS in 4m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"10971611daf71dc26bcaee132ee30fe245f6016f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 00:39:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f0fb9b2b29b13ef1014348dbbbc365d595e4b250","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 01:27:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/08ea69025eeb43ba93bddaa06fa93324 : SUCCESS in 4m 03s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/05ac3f0fe6ad4a6c805517d0c39256d0 : SUCCESS in 33m 50s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/8d8470711ce348c09f75a1c5dfb29305 : SUCCESS in 28m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/906ea7ad8e4b40449d44e980dbb5bb4d : SUCCESS in 32m 46s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/188d833d2e7c49718af2eadc81656b2e : SUCCESS in 46m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ca0241247fd44928a03291580a98d3b4 : FAILURE in 32m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/568f86674bd044c89dae52e1995cfb06 : SUCCESS in 32m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04d57feac20b41ee980cb47bb684f032 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7f94697b48d465884a7782168e8945c : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37bda74afe9344b0b840401801780c52 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5161a6f649f34d9bb0cb499520533dc2 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1ab1b759fcc45c88eddd6c78a6191d0 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d0110a358dd4061a63785a17d8c89ee : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"875a253eec6810e0cea54575f5be6126e47b8945","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-05 03:36:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d4b9b27fd00f3fc8df170ca40c79e5635492db4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 04:49:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/85c0d80feb234ad897a392197b9c3874 : SUCCESS in 4m 07s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a01dbbc7a97f476b9c03ec90c4099951 : SUCCESS in 41m 00s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/ddf4115b32194487852bc94aac0312b6 : SUCCESS in 56m 22s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/74e593d6cb074c01ad2199fd79517ba8 : SUCCESS in 47m 21s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4ce5cd665fa742cab65134d53a70c8a3 : SUCCESS in 32m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d02c822995c14cb295acbb036ab86560 : SUCCESS in 44m 09s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1b9bc83a60724714b198408342c6e774 : SUCCESS in 1h 06m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/647f335139ef4be7b55dc1c548c40030 : SUCCESS in 42m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fe82fca34afa49d8ba53334c66dc7fdc : SUCCESS in 46m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/002650cf0ab24d858253a97282d84e37 : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c55b52ed08b40e5a529a063fe132ffc : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/500bdc8e64814624960e3f7bef6aae9f : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd9b4d6fdc4f4f70996b082fe34eda9a : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c72889ce83064a8a9af501f8a85b766f : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38daa7c87d944b9ea7851b8509735ee6 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18e955c989db4ac8a0b7877ac6a679b0 : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d95a98954e134c21b4290e64f33b3fb4 : SUCCESS in 5m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/76978496e5d34337ad5c4565ee9b00e8 : SUCCESS in 4m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1c1e8b6e90cca4258796a702402a97cb7f4ab4f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 04:50:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"358db51ed3b9ed4a4eb39c1f9294c2ee1e25ceb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 06:02:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/054f9e743a684164873a1df3479be904 : SUCCESS in 4m 57s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/795efa8d8784415a8df65394b7026afc : SUCCESS in 42m 38s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/36f5ab4fdb4f4926a5734b5247bf1fec : SUCCESS in 44m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/23f3ab333f5d434680bfd93a38e604bc : SUCCESS in 42m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/84b61166fc644d71b96f5b08fdc4dff4 : FAILURE in 1h 08m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7052cb1e20d142388f41c884bc68ecf1 : SUCCESS in 41m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/05ba5abbdb70442f870f22ec43c9ae1f : SUCCESS in 39m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bc76cd02d9248d5a5c4f277902954c8 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/343a8aa22eac4175a571af7a6b625fc2 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6cbd4be73cd4fe2855b39e82db6398f : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6ef1761a29941a38134cfbc7dfd334f : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e753b57d1de04fbb9394c2d03c2c88b2 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04ad051e17bc43729f06e20695f1d660 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1f3b8c64f9bce1351cde78cea6b34cfb90d37d17","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-05 14:19:53.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e5c4a09fefd13d869924adad3cacee4fe208321","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-05 16:47:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab425e18c87886d4c01ea7257364a779d7a2d527","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 17:56:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1d1aa43514bf4c25afff13f0cee55502 : SUCCESS in 4m 06s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/40e4d9a14451416e8f1b8190b82abde3 : SUCCESS in 36m 00s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/ea40c71dacbe4e3497b89d5a5288ad07 : SUCCESS in 44m 35s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0582d2843ae54f7e99edb334142b88b5 : SUCCESS in 37m 27s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/6859ebcbb09b40d4b1cfe6d011083ca1 : SUCCESS in 37m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b80ef8124b30483abab03fbb346cf461 : SUCCESS in 45m 54s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fd5237cb2f744dc6bec2a16ffe5cbd6b : SUCCESS in 1h 02m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b47dcf9646e34c4b9aa2d9d42b5bbbc7 : SUCCESS in 34m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/069fe61360f74e51b785525c721c8426 : SUCCESS in 44m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9357acb35a584d98a2891dc92ea25f78 : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c85dda15e7ff4ef089c5fb22a7100a87 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7905e2194ed4b628f4e9a9f4562f521 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ce9c7d9172e40eb9ae9613c66536095 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8df63bc2b0de4cc0bd033455dae05c36 : SUCCESS in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d838e2266994195bfc8e19c67a06081 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5538d481ad664cb5abbf0e229170a0b2 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/824565f6b3da4ee7ab82c056116017e3 : SUCCESS in 3m 53s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/844e50e476d24c3ab18bf82e3185061a : SUCCESS in 5m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee1397f517439acab7f3524b8592f5b2410adf46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 17:57:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1153cde81314b12b2d3311943aa3bd19c23e6e2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 18:42:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/72aee7701b874a858d8f9d18f5a86430 : SUCCESS in 3m 34s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/16315080941840449b51f11fb56e1844 : SUCCESS in 39m 15s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/cff63d167c4540639f3c91536134d7f4 : SUCCESS in 39m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b2a6dea2994445479de6040a3151dc7b : SUCCESS in 29m 12s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f7e91398f04c442a8acbfe1d70fd2b5b : SUCCESS in 44m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4d670958b479438e9cebab45df3ffec8 : SUCCESS in 36m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b98f8460649f47f2977716c6ce187c9b : SUCCESS in 31m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1cc618d3c65f4ce098dea8e3dc60672e : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6df7b55b658b4c72aeef3cb445a95190 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5d0e795f4bf484cabecdf388f900c0a : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f34baa4d44334a4f8e74071681a4337d : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee90da23f35d4667a215a339596a3e62 : SUCCESS in 8m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f61e6eaa93a64e79a646916f401adca0 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"579a06930dc5c878892ea6d290b6b7d50dda1bff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 18:42:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"298ec47d9445e63633ccdde9dc2d327aa3adf125","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 18:44:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/4de54509cf9140c1b98ecdd3f7babe19 : SUCCESS in 1m 07s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9e2ea7a27f64292b895486df24c32a8 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/552e1994123149b99eebdd54960440e5 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30a03e1092b29457e4b79c9c7e33f835f534c857","revisions":{"30a03e1092b29457e4b79c9c7e33f835f534c857":{"kind":"REWORK","_number":1,"created":"2022-04-04 22:21:53.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/10/836410/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/836410/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/836410/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/836410/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/836410/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/836410/1"}}},"commit":{"parents":[{"commit":"6f847aeb5a309eb6e70dee58057fef01886573f6","subject":"Fix delete zone transfer request with scoped token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f847aeb5a309eb6e70dee58057fef01886573f6"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-04 22:16:23.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-04 22:21:52.000000000","tz":0},"subject":"Fix tox docs env to have the correct dependencies","message":"Fix tox docs env to have the correct dependencies\n\nPip is pulling in unstrained dependencies during the \"sibblings\" package\ninstall step. Normally this is fine as the dependencies have already been\ninstalled and it\u0027s just installing the package under test. However, with\ndesignate there are a lot of overlapping dependencies and pip will end up\nreinstalling an unconstrained version. In this case it was Jinja2.\n\nChange-Id: Ie49f99ba8e86913471ee6cd078f5afd28282c0f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/30a03e1092b29457e4b79c9c7e33f835f534c857"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/30a03e1092b29457e4b79c9c7e33f835f534c857"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
