)]}'
{"id":"openstack%2Fdesignate~836416","triplet_id":"openstack%2Fdesignate~master~Ie948931c1d31444c67829767f9ccb6c5eb6e6601","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"5572":{"account":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"last_update":"2022-04-14 06:43:24.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-04-14 04:27:17.000000000","reason":"Erik Olof Gunnar Andersson replied on the change"}},"hashtags":[],"change_id":"Ie948931c1d31444c67829767f9ccb6c5eb6e6601","subject":"Improved logging and minor cleanup of worker api","status":"MERGED","created":"2022-04-05 00:51:22.000000000","updated":"2022-04-14 06:44:52.000000000","submitted":"2022-04-14 06:43:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836416","meta_rev_id":"a4cec038b3dbca6742838fee2bf3452beb2cf01f","_number":836416,"virtual_id_number":836416,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-14 06:43:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-04-07 22:04:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-12 20:44:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-04-12 22:13:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-05 01:41:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-07 14:07:27.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2022-04-12 20:44:08.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"793a471e24e4174ea09c5692ffdb5822bc068940","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-05 00:51:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2858e671214b9c4e1946f11a6c05c2878a6a8ae5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 01:41:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4b2a8eb98abe478ca06f64017d6d2bec : SUCCESS in 34m 51s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/7c29ac7bf2bd4aaabfeece066daa028f : SUCCESS in 38m 35s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c289b17e805b4ac0b52ccfce3c5bafec : SUCCESS in 32m 39s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/b78ffbb25658461fb2a76fd16f657fb6 : SUCCESS in 35m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/dd680ae57c914b7db7cc48c8819ed123 : SUCCESS in 33m 23s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f5d4be5e28cb4779996ab7e6a04d2fb4 : SUCCESS in 48m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a6231f29856c4422b995d5149d36b57f : SUCCESS in 32m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bf579b1fd7b743caa782742cfff2d906 : SUCCESS in 34m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d33f374c180463ebce1818ab2a97703 : SUCCESS in 5m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49ae77d1f3854bde904f0f8b16052b4d : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b449b87e4214160ad02d73a6392f7bb : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb57baac08a0418c922a3deccc45cca3 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f8b1ff72b35425182a39b1fa8437a62 : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6930b4d8c954a949025cf0f6f48f609 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c925353829bc4f3f9b65c925126883cc : SUCCESS in 8m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d5e01d0b0b444683a58fcde60afab4b4 : SUCCESS in 29m 29s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1f3e11819f544fcebc12ed497cf53a56 : SUCCESS in 5m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"60970f9ec0ffe0810bca8dc15f66365ac47cfaef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-05 01:51:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"881c56dfd016eb97c54d4fd27b9e7b9dbdc8c0f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-05 01:53:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6ac523dba1de4e5f4afa08bf72d29d22269afce6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 03:00:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d11042012f344c078810a72075c0202c : SUCCESS in 52m 09s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/5c0b390cdf594ce78ec5c553d8f39121 : SUCCESS in 1h 00m 56s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8f28acc27684a299f68e683fbd6cf72 : SUCCESS in 50m 49s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f222da17153f4d14ad158d799aaec396 : SUCCESS in 50m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fd193b67c7d343d88ae7581b98eac733 : SUCCESS in 46m 40s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a0dabd48b49a44f99bc62fff69e687fd : SUCCESS in 1h 00m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dd13cd67260642bbadee11bed147db36 : SUCCESS in 47m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e111e22db5d1458fb5c66a8fd7a0a2ff : SUCCESS in 48m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/980fb9229744440090850832e9ef0105 : FAILURE in 5m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c892411110dc4ba58707fefbd999df1a : FAILURE in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d88ecb055f884ee88166f71d64514fd3 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47492789b38747f59fb39cd9e83cf9f3 : FAILURE in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6b04c9d64794de8aaa1888ba1f94508 : FAILURE in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/233252392246448d93c3024e007a10b4 : FAILURE in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e8e4667e93c41dc82c296da839d7aa1 : SUCCESS in 7m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/52669636e750445082a45c081f7594df : SUCCESS in 34m 11s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/36086219c6084166a7899828353fa68b : FAILURE in 4m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"182678435bc15508806edafd1b2cad5fe40d36ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-05 03:03:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"23853d6b76857989779ca4c207395288a3e57256","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 04:14:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/642f8906637f457a8427cd632f2e4c61 : SUCCESS in 43m 49s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/c359ed21d32a4e85953c32d9bcfa27d4 : SUCCESS in 57m 19s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/009d641dc30348dbbb275addf8c6fe4e : SUCCESS in 47m 42s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/554951e9d69944ec891691e87b0fc7af : SUCCESS in 43m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/880d39e9b2fd4403b4367d73e130f54d : SUCCESS in 47m 23s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3712becab35d44d6b59009f1cab825aa : SUCCESS in 1h 04m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/62d69d5da4ba4774841d44c320d726c6 : SUCCESS in 46m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a719ed0d786444298cf383135ed256aa : SUCCESS in 48m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e19005b42f7437c97f26f8a0d382dd1 : SUCCESS in 5m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3624b7f8500942ec8a5f8daf0658b46e : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8ec17c32d7c418494214db8132a66cb : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5312a8f317554181be03022362b3163a : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7efea21e20954f958bd9c7933443aecb : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eac8cc273c4c4852918166ac16cfd294 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1abbf070b7054c1cbb575aabfc7fdd76 : SUCCESS in 8m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a6d2bc07d18e4e89a3fd1771fa78989f : SUCCESS in 30m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/df5becf66f4349c4be8c51606813d249 : SUCCESS in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cee036b8a68b5a0b869f38228974a995cf2a7296","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-05 05:46:14.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"d4c4f63baabaaf542d199e591322900ae60dc797","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-05 05:49:58.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"7df1901b849e8d2d7ccab2efa37b7ac530f750ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 06:48:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0322e5b3af4f45d497881bf3a048ac26 : SUCCESS in 40m 20s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/087bfd2cc1b34adc817754aece8d906d : SUCCESS in 45m 38s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6fbbeace192442359509149bc5274f07 : SUCCESS in 42m 24s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/7ded9feb240044aeba947bf2ddfd4295 : SUCCESS in 39m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/da95a1476a4f4130977b4db53d7e959d : SUCCESS in 38m 03s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3fad0310f05c4a57b637ec41f9dac2a4 : SUCCESS in 56m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1d36a862b215472aa8e4f85649036511 : SUCCESS in 39m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/40a785d17bc94e088b4cf38915012048 : SUCCESS in 42m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3461ea7435ee4298ba9020ef03e7b19d : SUCCESS in 5m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bda122a2daf4818bd0239fd4ea1859b : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1f20273bf944275b574c205f83a532f : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe302030d2b0466798d40e9aa219b3ea : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/077d37bd9c3f46c38a5251a62f49c0e9 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b47b020f244f4560b098f2c333591562 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe0e7799a2124d0089a351d493de4c97 : SUCCESS in 9m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a447dc01b8dd426386aa1d6191698589 : SUCCESS in 32m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e4d60e224f23472c9adcc6f2c126644f : SUCCESS in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6e66e05fc5213a77da443b4155ecc0f44a64926d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-06 03:41:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"600634419d8ccf996336b4bbda5e0bb93742726b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-06 03:52:47.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"43c8c655710379a61c5a63969b1a94595354093a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-06 04:35:42.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"2616465b55bd338864a8e75fb23aaa165a477cff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-06 04:37:35.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"7df9c7b89a6f522bb825132cf26c056ae81d9473","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 05:37:08.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3bfd16ee0e004fa2a93d466303842ae1 : SUCCESS in 39m 35s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/2d8367fa9cea451eb89c48c103008525 : SUCCESS in 45m 00s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/671dc57c579446349607667d0f1e00a9 : SUCCESS in 37m 35s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1b74772e245d44159ba0b2bf6f2d606c : SUCCESS in 30m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e857669cd5624110866faafb7e8fd99b : SUCCESS in 29m 45s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8bdd33215e804cb9a9fafa0043b49c5b : SUCCESS in 58m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ee9f764e9dba43049b429592084a9174 : SUCCESS in 29m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/97a95259e23c4226ba8ce2b11d08949e : SUCCESS in 30m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60799d7e10fb4672bdf1a196d30e93db : SUCCESS in 5m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59bb3227b206426686395b04e409b128 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b1e65e856244dfd9dc28e91dfaa969e : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90b97488ef6b4aaebfdc6d2dc035439e : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07fe121020174937b8719c80788bdc30 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a452dab803a470bb4258a9c2ec34af5 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/185387a880324a96b5f9f665ecc9c327 : SUCCESS in 7m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8104a38bc08645cf829c3936eb0abdd1 : SUCCESS in 32m 42s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/327b2daa270d486b9e927804ad120a41 : SUCCESS in 5m 17s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"bae6caf590953c07926f302962feb36472689bf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-07 01:38:43.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"d9fe86089cab7b04aa3551a04b028c23d1a0c45e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 02:53:03.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5cf7a9e6e0114245a23450a2a9895e5b : SUCCESS in 57m 48s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/1b7ef3f647eb41f68664373d9f1b88b1 : SUCCESS in 53m 34s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2df37fc3a1614b16b606a58b48d5ade8 : SUCCESS in 48m 14s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/07ad64ae95574074b9dc1293bda79607 : SUCCESS in 58m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a0678d32d5214c398864f28328ffc94f : SUCCESS in 43m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/65ac20d225924829a7c6858e19846d9a : SUCCESS in 1h 09m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dadb7d6850ae4b83a9e720603ad5b347 : SUCCESS in 45m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d54d0e1560884469a2f87b19f128a05b : SUCCESS in 46m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2a72f0205964a83bff04fc2150d8ae7 : SUCCESS in 5m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a262c7282724a50a7422db1253b9162 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2e1a3a935594a2eb7ca3fcd6182dbd0 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c34a24b762e3465f94c360b78a275f02 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/991a038a117941b9bd54a09ea4f4ba67 : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9eee60a3fe074b039a62b68eaf642164 : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e02775529a054251ba4ec506d2614d4d : SUCCESS in 9m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2f30c7ffd029418fbbb074b108d73269 : SUCCESS in 35m 01s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ee7a847b570840d497cf05b5eb367c1f : SUCCESS in 5m 57s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"2097002dbcba9a7025d5852122eefcae554d2add","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-07 04:12:46.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"54cf433f10bd3e655a7d6a07a44e9de2807d7458","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-07 04:16:47.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"12114e029b292efb9e223ad4c3f5c6a467619763","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 05:22:45.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a06731662fed49769efb1e957a7a2ca6 : SUCCESS in 27m 36s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/f43ca479110c41c9b6ce3e9a1a46c235 : SUCCESS in 1h 05m 02s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3958d7211ec488c91a42dd2ec645dc2 : SUCCESS in 34m 24s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1a52292efc6d4e31b997da9b6927c744 : SUCCESS in 36m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b6c4353993944ace98bf3b23b22cb389 : SUCCESS in 38m 18s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6245f33e6fe942718f53e714df52a757 : SUCCESS in 41m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/768c7ea8d5ed43e18a28b30c5f883769 : SUCCESS in 34m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3a5b58b2138649bd8363afd594047413 : SUCCESS in 32m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c56cfd96c32b4dfaab6f47ddd24b1e4f : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f3969f607e6444295be5e722a4d5437 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90a7331c2d03425393afa421d9cd1786 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/481d678b428644de820ff2379553438f : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1fc3acec2546490e98f0ef63f82eea24 : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec79686b48d6489a9af3933524ded789 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b0c2068de384f5e87a27d32cd3131bc : SUCCESS in 9m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d195f10541714b26bd325041b86174de : SUCCESS in 32m 57s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/851a4c7d51524d7f9478324133530cca : SUCCESS in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"777889a858dedb645dae5f55654babd6c65281fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-07 05:42:55.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"7d3662345336357140192e7502e045e873379c94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 06:57:47.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/16fa13a687d24193bc01254cdcf63371 : SUCCESS in 47m 50s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/da3d8941ef144ba18b08142a850cc700 : SUCCESS in 55m 20s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e582d1de4a64c4c92cde8405d1567f3 : SUCCESS in 57m 44s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e176ec75ec5a4570b9f56da807673cf6 : SUCCESS in 49m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/057e5d54166f436f85944ab9ac9b756b : SUCCESS in 45m 32s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d824b9f1af634a1b879c9b74473fa190 : SUCCESS in 1h 08m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/523ea5c71863481b99af19227b2b87c1 : SUCCESS in 43m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/df92b8c8135f40f0b380897eb8af232d : SUCCESS in 43m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/225cbfd0223d4a3e8ab65c8662c6acc5 : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05c9eab6b7c04a6d8bbf546019e7ec21 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04952a31c8df4540afd2f7878064229b : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a78bdaec7e534230b21d55749a264c5f : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/805d57bfa54c4b969b780aa0ddc010b7 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a17783eb9384485a151f8b4b50f3251 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1111d2930fcf4729b97c5ddc7eedf1e3 : SUCCESS in 8m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/73daca59ca36413fbcccb24bce758cc3 : SUCCESS in 29m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a310eb79b91e4aae97db4d31c6497552 : SUCCESS in 4m 23s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"9c1c723186d268e89292752c6b241665f192b624","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-07 14:07:27.000000000","message":"Patch Set 13: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"d373491bb03fa71a795a38baf7a1b9054e5cb62a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-07 17:05:53.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"736406507260463e8c9269c6a428988152bf8dc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-07 17:16:21.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"613fde409e211f2ec19b81f7cb25544636ce5772","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-07 17:16:43.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"8c34e8ab4db5765a740ed45fae8e0f902ca42e1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 18:17:58.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e2cf9e1ac94f4ee8b2312cfafc4b9416 : SUCCESS in 32m 14s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/e99ca8b70fae4c3a9f9b3116ef2414b0 : SUCCESS in 56m 39s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/33860c87fecf4e73a4e2bb2943d5bd08 : SUCCESS in 46m 58s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4c1b3f6a30f94149b65361fc53169887 : SUCCESS in 53m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9b484207c2bc49dea3a89a2fad5e7eef : SUCCESS in 43m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/dbf48c92d63443c9afd3cc39143e6cdd : SUCCESS in 49m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3e79c20f5d2d4287983416bc395d6889 : SUCCESS in 43m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f359edbfc36b4616881f4091d92309f3 : SUCCESS in 39m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d3218ca447d49f89c2d1aa50f3cc532 : SUCCESS in 6m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7692485f4d64764a0f609505dacaa46 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02e4604fb43b45688e6a3a36ba806be0 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96e0b0bbb4ad4de8bb0cf1f2ebc5fbeb : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5f0001c96c64a769fad50377703e969 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/890d82acdb1048fc8b3447172050b2ab : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/559e0dd385944173844ae4d3a0873c88 : SUCCESS in 7m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6042bc32ab4c4069b304cf51c6147957 : SUCCESS in 26m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6278db4d227e4edbbd2a73f60148f6e0 : SUCCESS in 4m 45s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"56c39079154577be2836662692b805944a576438","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-07 22:04:58.000000000","message":"Patch Set 14: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"a7cebcc26920857797b6225b6fcf24e65dee59de","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-08 01:17:23.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"065b041f8698f8db2d2bbbcb1c2dd8710413823e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-12 20:44:08.000000000","message":"Patch Set 14: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"658ec6d9c9dfe783b72e39dabdcb04691b2c2934","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-12 22:13:24.000000000","message":"Patch Set 14: Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"1e12f5a1400d1562cdc8a9e47470752e98d55869","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 22:14:17.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"d31bb5352db005b26a16520d8edbf01456d75fa5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 22:28:38.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0af159308ce348e48e0f227180293887 : FAILURE in 12m 06s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/6f15826ef84c469da5ba52a3236d5448 : FAILURE in 12m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/39f44893feef45e3a969d13a7a3f7eb6 : FAILURE in 11m 22s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d236a1c27c9d441788e55398ab5f3830 : FAILURE in 13m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/141a43e0c5cf45b99aeb898c7800d781 : FAILURE in 11m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/68fdcf164be2471daf0c399e99f127ae : FAILURE in 11m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f0013dc2b7b40b0aa09fdc45fd46662 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09a7123b3cc849139db7478fcddcb396 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fb83913da5d46f3a7082128b40bc0c0 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30f2717b656a42cba1e23ea86bf9c1c1 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c5136eb00ac4c28b3e78371647466c5 : SUCCESS in 8m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/05f6fe718c0f41f5b2969b692815c255 : FAILURE in 8m 48s","accounts_in_message":[],"_revision_number":14},{"id":"0216e8078f59712d8dfdb3e5f240aa2f8724ba7f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-12 22:33:16.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"4bb485d3978b14d0c1df2e6bfc1c56a09541f1dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 23:11:05.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7206f045ca8b4848ac1fdc6b8c3a1a27 : FAILURE in 9m 23s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/70328bf1826d4116b9d1b0ff972842fe : SUCCESS in 36m 20s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2150d9ab29764029aec9e6c50b07cce8 : SUCCESS in 32m 20s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4add33c03ffe41c9a44996f4eded9407 : FAILURE in 9m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ad67d49b275e4cb581c46537a3621386 : FAILURE in 9m 57s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cfb81c57690642c89847b15b290a7913 : FAILURE in 10m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/69af9d6897c8433d93947ff120124a15 : FAILURE in 9m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d068025d06b64a25869058c012e7e090 : FAILURE in 9m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f613419c49b3437bae0cb1d3b6e6e16b : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c45bd8e5d6544459ed0633fd5541897 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c476360c864462d88f1252c5dcd7f66 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48c1a4e17ed1486bbe40dcda4d848235 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d46d989f42d843a68b4896fc728ce1c4 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6651472b8c6e4eb6a6116f1a92d2650a : RETRY_LIMIT in 2m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/669d7077e931430886168a8bc77b793b : SUCCESS in 7m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2fa850c2ead04dcca93a55479651d76e : FAILURE in 9m 44s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5e609f1309514c51b6b5a5b29aa64ea6 : SUCCESS in 4m 44s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"06a03a1c7037326e287acc6a5871b133169fe254","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-13 23:31:44.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"eb66fa2658b7b4c4953af44b013b8e4572df1288","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 00:44:50.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e80b5a66c29d46248f07a9da6f140f10 : SUCCESS in 41m 29s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/058b2c73a45646adb4d17f735c74f6c6 : SUCCESS in 40m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/17b45aa3ee9e491c9ef128f156d6fae9 : SUCCESS in 48m 45s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/9eb004b1ec1641d0b4cd5cc3f0ff0fa5 : SUCCESS in 47m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9175d70ab6d949439c4b9f7dbdc77fe2 : SUCCESS in 45m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/694e7f91eaf348e69f2efb6708c8d5df : SUCCESS in 1h 07m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c997201bb3e34b0eacf87d201af74f81 : SUCCESS in 34m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/54f2f7e044274272b68fc215a8daa189 : SUCCESS in 35m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55692627d3da4ea8aada72e076f1266e : SUCCESS in 5m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff31994541ce4f9f98b2339fe4fd0c31 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ca5bc9fefec43269f6dc53a8f6ac31e : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d51e722ffb84077aaed97be1fc1f980 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1e1503f7e034fa1afff66dbb7c4d05c : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51d763fe30d34a1f87ddeabccf62a8c0 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24c1656e8ca64a9b893fa3a949a2205f : SUCCESS in 7m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/16e97bbdf54944f7920aa0bf05171e6e : SUCCESS in 26m 35s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b27b98bcff1b438cbbd6dbfd352ee4bb : SUCCESS in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"1c64cf8ec90e7b9417c145610f54f2c516dfecca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 00:45:49.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"b8291ce4a73b3d0b867baf4368b18c049b6ff279","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 01:34:05.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d87e32647aaf46fcbdeb37fe43f1a576 : SUCCESS in 46m 19s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/fc082e2c1b534ea18e9aff20fe43a0cb : SUCCESS in 48m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2f8e69b8aca4483ba3fefbe12304267f : SUCCESS in 42m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2133276c264d417c83bfc03229395404 : FAILURE in 27m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/270f97d2401143b5ae983deb33e10f7e : SUCCESS in 34m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/652782d93160454b818f24fe3699a65c : SUCCESS in 42m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/335663f8462d44c58969cd4ca77c5f18 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1c6caac6ffc4d89a8432800e7379784 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6323cefcf77459a97e634a07fcdd7e1 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed8314a24e484f7e911724964e7602db : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f1e0b39b5f04e499eefe26493763e87 : SUCCESS in 8m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d3939cd3a7a448b796ed432b10fb5525 : SUCCESS in 33m 16s","accounts_in_message":[],"_revision_number":14},{"id":"ef2c919c458399d06068826badedf6a57bf1aa44","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-14 04:27:17.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"32c53e04575d6bdccd13c50c222279a009d6f7e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 05:30:58.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4b184acda0ed4db4a776586cff59d1e9 : SUCCESS in 55m 47s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/9a04deeacbfe4fc7b3f7a9476d2ab25f : SUCCESS in 55m 10s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f21ac0f5a714437ba0749b99e7e75b0 : SUCCESS in 46m 16s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1779e077411c4c8fb2c6a7caea72c1e2 : SUCCESS in 47m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/036ebd7002574efd91141fc421a63de2 : SUCCESS in 43m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7b309895c3dd45cca6629e3117640467 : SUCCESS in 58m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2eea6173163e45da962fa628923c708f : SUCCESS in 44m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e62f75c7b81b456ba7afea5d5172b450 : SUCCESS in 43m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e70e6266e59489c8d1a52ffdb833cd3 : SUCCESS in 5m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/610e682ef6b84135a2e1795865cfbb30 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d74c8922d4bc4f25bb4627bc0bba34d5 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/382385ed05b348028113ba1652b63a1e : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/003b99cc4778471197848719b6f6a440 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/486c40b313354ebd8a2432f2d44fcd87 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dde694b709364127a158b6f20423a3dc : SUCCESS in 7m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4bc6f6c539a342eb8f10ec98732fcdb8 : SUCCESS in 30m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f8140b23ca65460e9d49ff401ecfd355 : SUCCESS in 4m 43s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"912f27698a26a36c0201a20c5ac722ce12011fc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 05:31:48.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"99264e698e264c22a85ef2c04f6e96259c1d60fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 06:43:23.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a63024d59dc8440ebc89f188f00bc92b : SUCCESS in 49m 59s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/0c9b8494dd7548deb47fd6cd4acbeeb1 : SUCCESS in 46m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3df359dd5ea14a3996873e8d3c800911 : SUCCESS in 43m 34s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c4285516578b46e09653f6bda5e31f74 : SUCCESS in 1h 05m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/923464f0342149479bc3c5102681a115 : SUCCESS in 41m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/85f47f4cf79e4c6e8b7defb86b826147 : SUCCESS in 48m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a27dcdeece24f318f3ab13885896051 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/216483b1b9854654bbf283a65fcdba56 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/032e5ff4afaf43728d5bfd54faaf166b : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43d2cd32a2564c2e977ae3fbb811f567 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2052a8e88a5b4880bf9a430e852a76a3 : SUCCESS in 7m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0b02cbcc878940749df5a24bc4334435 : SUCCESS in 30m 12s","accounts_in_message":[],"_revision_number":14},{"id":"f12bf7492f621457cbddee4799008f21a2661d17","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 06:43:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"a4cec038b3dbca6742838fee2bf3452beb2cf01f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 06:44:52.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/805e82fe0b0a49cc80f437de457d736d : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"147e00f44febebf8089a93b58d739dbe174c9f4e","revisions":{"d076d10ef28108493b13d3228c0b3a0324e10fb4":{"kind":"REWORK","_number":1,"created":"2022-04-05 00:51:22.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/836416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/836416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/836416/1"}}},"commit":{"parents":[{"commit":"6f847aeb5a309eb6e70dee58057fef01886573f6","subject":"Fix delete zone transfer request with scoped token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f847aeb5a309eb6e70dee58057fef01886573f6"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-05 00:51:00.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-05 00:51:00.000000000","tz":-420},"subject":"[WIP] Testing some new logging","message":"[WIP] Testing some new logging\n\nChange-Id: Ie948931c1d31444c67829767f9ccb6c5eb6e6601\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d076d10ef28108493b13d3228c0b3a0324e10fb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d076d10ef28108493b13d3228c0b3a0324e10fb4"}]},"branch":"refs/heads/master"},"1f5c3ec3db75b7063fb2146a5e1b8d8f42c6e9a5":{"kind":"REWORK","_number":2,"created":"2022-04-05 01:51:03.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/836416/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/836416/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/836416/2"}}},"commit":{"parents":[{"commit":"6f847aeb5a309eb6e70dee58057fef01886573f6","subject":"Fix delete zone transfer request with scoped token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f847aeb5a309eb6e70dee58057fef01886573f6"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-05 00:51:00.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-05 01:49:37.000000000","tz":-420},"subject":"Improved logging and minor cleanup of worker api","message":"Improved logging and minor cleanup of worker api\n\nUpdated existing logging to be a bit more consistent and\nadded some additional loggiong.\n\nIn addition to logging improvements the following changes\nwere also made.\n\n- Replaced legacy endpoint zone_api with worker_api as\n  the zone api no longer exists.\n\n- Removed some legacy mixing code no longer needed.\n\nChange-Id: Ie948931c1d31444c67829767f9ccb6c5eb6e6601\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1f5c3ec3db75b7063fb2146a5e1b8d8f42c6e9a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1f5c3ec3db75b7063fb2146a5e1b8d8f42c6e9a5"}]},"branch":"refs/heads/master"},"08ff858256af379415deeb8bf7739fa5b900bfc9":{"kind":"REWORK","_number":3,"created":"2022-04-05 01:53:29.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/836416/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/836416/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/836416/3"}}},"commit":{"parents":[{"commit":"6f847aeb5a309eb6e70dee58057fef01886573f6","subject":"Fix delete zone transfer request with scoped token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f847aeb5a309eb6e70dee58057fef01886573f6"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-05 00:51:00.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-05 01:53:19.000000000","tz":-420},"subject":"Improved logging and minor cleanup of worker api","message":"Improved logging and minor cleanup of worker api\n\nUpdated existing logging to be a bit more consistent and\nadded some additional loggiong.\n\nIn addition to logging improvements the following changes\nwere also made.\n\n- Replaced legacy endpoint zone_api with worker_api as\n  the zone api no longer exists.\n\n- Removed some legacy mixing code no longer needed.\n\nChange-Id: Ie948931c1d31444c67829767f9ccb6c5eb6e6601\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08ff858256af379415deeb8bf7739fa5b900bfc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08ff858256af379415deeb8bf7739fa5b900bfc9"}]},"branch":"refs/heads/master"},"e5073859844a6b90931245dbad8e805630a385c6":{"kind":"REWORK","_number":4,"created":"2022-04-05 03:03:06.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/836416/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/836416/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/836416/4"}}},"commit":{"parents":[{"commit":"6f847aeb5a309eb6e70dee58057fef01886573f6","subject":"Fix delete zone transfer request with scoped token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f847aeb5a309eb6e70dee58057fef01886573f6"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-05 00:51:00.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-05 03:02:59.000000000","tz":-420},"subject":"Improved logging and minor cleanup of worker api","message":"Improved logging and minor cleanup of worker api\n\nUpdated existing logging to be a bit more consistent and\nadded some additional loggiong.\n\nIn addition to logging improvements the following changes\nwere also made.\n\n- Replaced legacy endpoint zone_api with worker_api as\n  the zone api no longer exists.\n\n- Removed some legacy mixing code no longer needed.\n\nChange-Id: Ie948931c1d31444c67829767f9ccb6c5eb6e6601\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e5073859844a6b90931245dbad8e805630a385c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e5073859844a6b90931245dbad8e805630a385c6"}]},"branch":"refs/heads/master"},"551e18ba88204c517ceec112cefd5608b3ac13a1":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-04-05 05:46:14.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/836416/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/836416/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/836416/5"}}},"commit":{"parents":[{"commit":"6f847aeb5a309eb6e70dee58057fef01886573f6","subject":"Fix delete zone transfer request with scoped token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f847aeb5a309eb6e70dee58057fef01886573f6"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-05 00:51:00.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-05 05:46:14.000000000","tz":0},"subject":"Improved logging and minor cleanup of worker api","message":"Improved logging and minor cleanup of worker api\n\nUpdated existing logging to be a bit more consistent and\nadded some additional logging.\n\nIn addition to logging improvements the following changes\nwere also made.\n\n- Replaced legacy endpoint zone_api with worker_api as\n  the zone api no longer exists.\n\n- Removed some legacy mixing code no longer needed.\n\nChange-Id: Ie948931c1d31444c67829767f9ccb6c5eb6e6601\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/551e18ba88204c517ceec112cefd5608b3ac13a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/551e18ba88204c517ceec112cefd5608b3ac13a1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e67fc23fdd0d77a71c2dd7f6e896ab7f1f6ee978":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-04-05 05:49:58.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/836416/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/836416/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/836416/6"}}},"commit":{"parents":[{"commit":"6f847aeb5a309eb6e70dee58057fef01886573f6","subject":"Fix delete zone transfer request with scoped token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f847aeb5a309eb6e70dee58057fef01886573f6"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-05 00:51:00.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-05 05:49:57.000000000","tz":0},"subject":"Improved logging and minor cleanup of worker api","message":"Improved logging and minor cleanup of worker api\n\nUpdated existing logging for the worker api to be a bit more\nconsistent and added some additional logging.\n\nIn addition to the logging improvements the following changes\nwere also made.\n- Replaced legacy endpoint zone_api with worker_api as\n  the zone api no longer exists.\n- Merged mixing code with base class to simplify code.\n\nChange-Id: Ie948931c1d31444c67829767f9ccb6c5eb6e6601\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e67fc23fdd0d77a71c2dd7f6e896ab7f1f6ee978"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e67fc23fdd0d77a71c2dd7f6e896ab7f1f6ee978"}]},"branch":"refs/heads/master","description":"Edit commit message"},"867ab1db6342e9d814d8863dd7d3ca749f48ab5b":{"kind":"REWORK","_number":7,"created":"2022-04-06 03:52:47.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/836416/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/836416/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/836416/7"}}},"commit":{"parents":[{"commit":"6f847aeb5a309eb6e70dee58057fef01886573f6","subject":"Fix delete zone transfer request with scoped token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f847aeb5a309eb6e70dee58057fef01886573f6"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-05 00:51:00.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-06 03:52:40.000000000","tz":-420},"subject":"Improved logging and minor cleanup of worker api","message":"Improved logging and minor cleanup of worker api\n\nUpdated existing logging for the worker api to be a bit more\nconsistent and added some additional logging.\n\nIn addition to the logging improvements the following changes\nwere also made.\n- Replaced legacy endpoint zone_api with worker_api as\n  the zone api no longer exists.\n- Merged mixing code with base class to simplify code.\n\nChange-Id: Ie948931c1d31444c67829767f9ccb6c5eb6e6601\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/867ab1db6342e9d814d8863dd7d3ca749f48ab5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/867ab1db6342e9d814d8863dd7d3ca749f48ab5b"}]},"branch":"refs/heads/master"},"ffb15201fda8c112afbcf54c28badea2a4d7ce2d":{"kind":"REWORK","_number":8,"created":"2022-04-06 04:35:42.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/836416/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/836416/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/836416/8"}}},"commit":{"parents":[{"commit":"6f847aeb5a309eb6e70dee58057fef01886573f6","subject":"Fix delete zone transfer request with scoped token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f847aeb5a309eb6e70dee58057fef01886573f6"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-05 00:51:00.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-06 04:35:29.000000000","tz":-420},"subject":"Improved logging and minor cleanup of worker api","message":"Improved logging and minor cleanup of worker api\n\nUpdated existing logging for the worker api to be a bit more\nconsistent and added some additional logging.\n\nIn addition to the logging improvements the following changes\nwere also made.\n- Replaced legacy endpoint zone_api with worker_api as\n  the zone api no longer exists.\n- Merged mixing code with base class to simplify code.\n\nChange-Id: Ie948931c1d31444c67829767f9ccb6c5eb6e6601\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffb15201fda8c112afbcf54c28badea2a4d7ce2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffb15201fda8c112afbcf54c28badea2a4d7ce2d"}]},"branch":"refs/heads/master"},"2427764d3fe549ebea40c24b4a80d5e9062d1840":{"kind":"REWORK","_number":9,"created":"2022-04-06 04:37:35.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/836416/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/836416/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/836416/9"}}},"commit":{"parents":[{"commit":"6f847aeb5a309eb6e70dee58057fef01886573f6","subject":"Fix delete zone transfer request with scoped token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f847aeb5a309eb6e70dee58057fef01886573f6"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-05 00:51:00.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-06 04:37:28.000000000","tz":-420},"subject":"Improved logging and minor cleanup of worker api","message":"Improved logging and minor cleanup of worker api\n\nUpdated existing logging for the worker api to be a bit more\nconsistent and added some additional logging.\n\nIn addition to the logging improvements the following changes\nwere also made.\n- Replaced legacy endpoint zone_api with worker_api as\n  the zone api no longer exists.\n- Merged mixing code with base class to simplify code.\n\nChange-Id: Ie948931c1d31444c67829767f9ccb6c5eb6e6601\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2427764d3fe549ebea40c24b4a80d5e9062d1840"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2427764d3fe549ebea40c24b4a80d5e9062d1840"}]},"branch":"refs/heads/master"},"02e6d87066dba89117d85cf47947b3e9b1aef5c9":{"kind":"REWORK","_number":10,"created":"2022-04-07 01:38:43.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/836416/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/836416/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/836416/10"}}},"commit":{"parents":[{"commit":"6f847aeb5a309eb6e70dee58057fef01886573f6","subject":"Fix delete zone transfer request with scoped token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f847aeb5a309eb6e70dee58057fef01886573f6"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-05 00:51:00.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-07 01:38:36.000000000","tz":-420},"subject":"Improved logging and minor cleanup of worker api","message":"Improved logging and minor cleanup of worker api\n\nUpdated existing logging for the worker api to be a bit more\nconsistent and added some additional logging.\n\nIn addition to the logging improvements the following changes\nwere also made.\n- Replaced legacy endpoint zone_api with worker_api as\n  the zone api no longer exists.\n- Merged mixing code with base class to simplify code.\n\nChange-Id: Ie948931c1d31444c67829767f9ccb6c5eb6e6601\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/02e6d87066dba89117d85cf47947b3e9b1aef5c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/02e6d87066dba89117d85cf47947b3e9b1aef5c9"}]},"branch":"refs/heads/master"},"d177f69812d8ba30e9f218daf4ee977dc582cd4b":{"kind":"REWORK","_number":11,"created":"2022-04-07 04:12:46.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/836416/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/836416/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/836416/11"}}},"commit":{"parents":[{"commit":"6f847aeb5a309eb6e70dee58057fef01886573f6","subject":"Fix delete zone transfer request with scoped token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f847aeb5a309eb6e70dee58057fef01886573f6"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-05 00:51:00.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-07 04:12:35.000000000","tz":-420},"subject":"Improved logging and minor cleanup of worker api","message":"Improved logging and minor cleanup of worker api\n\nUpdated existing logging for the worker api to be a bit more\nconsistent and added some additional logging.\n\nIn addition to the logging improvements the following changes\nwere also made.\n- Replaced legacy endpoint zone_api with worker_api as\n  the zone api no longer exists.\n- Merged mixing code with base class to simplify code.\n\nChange-Id: Ie948931c1d31444c67829767f9ccb6c5eb6e6601\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d177f69812d8ba30e9f218daf4ee977dc582cd4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d177f69812d8ba30e9f218daf4ee977dc582cd4b"}]},"branch":"refs/heads/master"},"698bb319183aad43241f557dc1d3fcbbc6633324":{"kind":"REWORK","_number":12,"created":"2022-04-07 04:16:47.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/836416/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/836416/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/836416/12"}}},"commit":{"parents":[{"commit":"6f847aeb5a309eb6e70dee58057fef01886573f6","subject":"Fix delete zone transfer request with scoped token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f847aeb5a309eb6e70dee58057fef01886573f6"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-05 00:51:00.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-07 04:16:37.000000000","tz":-420},"subject":"Improved logging and minor cleanup of worker api","message":"Improved logging and minor cleanup of worker api\n\nUpdated existing logging for the worker api to be a bit more\nconsistent and added some additional logging.\n\nIn addition to the logging improvements the following changes\nwere also made.\n- Replaced legacy endpoint zone_api with worker_api as\n  the zone api no longer exists.\n- Merged mixing code with base class to simplify code.\n\nChange-Id: Ie948931c1d31444c67829767f9ccb6c5eb6e6601\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/698bb319183aad43241f557dc1d3fcbbc6633324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/698bb319183aad43241f557dc1d3fcbbc6633324"}]},"branch":"refs/heads/master"},"263f13b0d6ea030a6adbde21b2cdc6353ee9b782":{"kind":"REWORK","_number":13,"created":"2022-04-07 05:42:55.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/836416/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/836416/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/836416/13"}}},"commit":{"parents":[{"commit":"6f847aeb5a309eb6e70dee58057fef01886573f6","subject":"Fix delete zone transfer request with scoped token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f847aeb5a309eb6e70dee58057fef01886573f6"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-05 00:51:00.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-07 05:42:49.000000000","tz":-420},"subject":"Improved logging and minor cleanup of worker api","message":"Improved logging and minor cleanup of worker api\n\nUpdated existing logging for the worker api to be a bit more\nconsistent and added some additional logging.\n\nIn addition to the logging improvements the following changes\nwere also made.\n- Replaced legacy endpoint zone_api with worker_api as\n  the zone api no longer exists.\n- Merged mixing code with base class to simplify code.\n\nChange-Id: Ie948931c1d31444c67829767f9ccb6c5eb6e6601\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/263f13b0d6ea030a6adbde21b2cdc6353ee9b782"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/263f13b0d6ea030a6adbde21b2cdc6353ee9b782"}]},"branch":"refs/heads/master"},"147e00f44febebf8089a93b58d739dbe174c9f4e":{"kind":"REWORK","_number":14,"created":"2022-04-07 17:16:21.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/16/836416/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/16/836416/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/16/836416/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/16/836416/14"}}},"commit":{"parents":[{"commit":"6f847aeb5a309eb6e70dee58057fef01886573f6","subject":"Fix delete zone transfer request with scoped token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f847aeb5a309eb6e70dee58057fef01886573f6"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-05 00:51:00.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-04-07 17:16:13.000000000","tz":-420},"subject":"Improved logging and minor cleanup of worker api","message":"Improved logging and minor cleanup of worker api\n\nUpdated existing logging for the worker api to be a bit more\nconsistent and added some additional logging.\n\nIn addition to the logging improvements the following changes\nwere also made.\n- Replaced legacy endpoint zone_api with worker_api as\n  the zone api no longer exists.\n- Merged mixing code with base class to simplify code.\n\nChange-Id: Ie948931c1d31444c67829767f9ccb6c5eb6e6601\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/147e00f44febebf8089a93b58d739dbe174c9f4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/147e00f44febebf8089a93b58d739dbe174c9f4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
