)]}'
{"id":"openstack%2Fdesignate~836874","triplet_id":"openstack%2Fdesignate~stable%2Fyoga~Ie49f99ba8e86913471ee6cd078f5afd28282c0f5","project":"openstack/designate","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-04-07 00:05:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie49f99ba8e86913471ee6cd078f5afd28282c0f5","subject":"Fix tox docs env to have the correct dependencies","status":"MERGED","created":"2022-04-06 18:08:26.000000000","updated":"2022-04-07 00:06:22.000000000","submitted":"2022-04-07 00:05:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836874","meta_rev_id":"81658b005a16a5da86dd31616a2b05b98bde412c","_number":836874,"virtual_id_number":836874,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-07 00:05:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-04-06 20:26:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2022-04-06 21:03:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-06 18:41:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2022-04-06 21:03:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2022-04-06 18:15:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-06 18:41:58.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-04-06 18:57:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-06 20:26:59.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"fe3623f10ff6fb7aa026c057b5a2102c5796c286","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-06 18:08:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e88900a63659256c95bb750f8924b0eda2cb5b63","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-06 18:15:35.000000000","message":"Patch Set 1: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30957c6259793a83ec2b373d00f942ff8b49d50c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-06 18:41:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4ecff3e9d9c6d40649fddba68f06d2c9aee44f95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 18:57:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0c713f00e3104a178b8b8f837a9c7813 : SUCCESS in 38m 33s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/1926e3ced0754e369867ae6d72471aed : FAILURE in 12m 57s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/79d0ca2142a648f5a01da8796f84aab1 : SUCCESS in 45m 47s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3a38326ded0d40328940f508dada84b4 : SUCCESS in 35m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f63c075b22fa48c18be43c86526098a1 : SUCCESS in 41m 22s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9d994dfb10de495eb426a1d424c1ac27 : SUCCESS in 46m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/709772df196d40e38e1a56e6f8e3e722 : SUCCESS in 39m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/688ac6d7054e4fb9baf7d92a7dd85d3b : SUCCESS in 34m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e8c6fa585f44e4d9dbc4aa593a51a32 : SUCCESS in 5m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/055bf4af587a448ba906254e2e34a7c3 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b840642dcaca46669e131fc339715e00 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3529bc2b4a834e8fb3a57b45c8658aed : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c8012b7b4c84a63ad6105705c9cf37c : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/767eff6252d4404b9ff87d70747b3699 : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a76bba57ee764c9186300bb0623a9063 : SUCCESS in 4m 30s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/66ca32c739534713afc1186fbe755533 : SUCCESS in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f389dc44bd7cc79723677b6fd3671ac036290086","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-06 20:26:59.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c1c8df4573a589c7f93f54500bbeb2fb3dde17ec","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-06 21:03:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d16d87e2832957bc0b35b4577c8c1837e12d5033","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 21:03:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b6af023e81cd9b60c5d68ce3aa356167662128fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 21:37:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/512ceedcd5f74d9e893716bd7392af54 : SUCCESS in 30m 57s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/b9b2e1a781bb462a8dd52a9d9ca609dd : SUCCESS in 33m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ece0220e6a7c4b2f8f47ccacc8dc809d : SUCCESS in 31m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6664196586704fceb86ebf80a17772a5 : FAILURE in 26m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4fd835da9e364cf4b3ac1f7d2464e2fd : SUCCESS in 30m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2c017557c9a341e7bb68d35c10115a30 : SUCCESS in 30m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/462395f293fa4b25b59bd0504e83115e : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7401a2de476e45cb974c1e71db6b43fd : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd5c31e03e6545a2ad48f1b7cdfb2540 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f7b76e7c5184835a4620d113a26f1d2 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e584ed7d12e4235aa014b9c3dfad5ca : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c483b4e929254b70bd9792df80abe46a : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c2e1d2c8817024ecf2c2d06c1fc2631c227f0430","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-06 22:09:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"20937f489f6e161a0559bf3be39ba503dd3bd6ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 22:54:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f125e435724f4facbc335ffb679f130c : SUCCESS in 40m 46s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/5180d7bee91f4eb0b5c57dc00eb595a1 : FAILURE in 25m 47s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2571619840cf44c895ca88f02bc07f59 : SUCCESS in 41m 22s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/39c0926190a64f2bbf1015d35ec0bf5b : SUCCESS in 30m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0c3ac322e1024aa5b854fc465e1f914a : SUCCESS in 37m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3ebb550e39784f47a477d3ff118f340a : SUCCESS in 40m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/743ad9145460460bbb9e3c6f2c9b9051 : SUCCESS in 29m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/08fb84a5de034df2a3d2ea6e425dc7a9 : SUCCESS in 31m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ffce453524f4c1e95d9d8a2c512fbea : SUCCESS in 4m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fba91ec3d2a74899a7d1bed1c90bcbe2 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97da9c61caec4889b0729e6b08e23732 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e594f3891e7c42cfb7386f6893e19821 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/547bc32f8fe94f76aac41829428d727d : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e562a0e4eba49e6abe822f7ea42cf3f : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56bca4de952841fd8accf28288aa2d1b : SUCCESS in 3m 37s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/65cf9a02cd854113bd91ae939634833b : SUCCESS in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3b9d1943cd98169f65a982ae6d2e6115a97e0510","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 22:55:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cba41f61fddcc0f279879d596e82174ce01a6f02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 00:05:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/60188bcb1b764a3391bb2fceae4898d0 : SUCCESS in 47m 01s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/5556362341064a678c4bc51bbf676d28 : SUCCESS in 45m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2b27798de66141e8b887febe5f779b56 : SUCCESS in 43m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/53b0b1be2264402a887d98fcee865ca6 : SUCCESS in 1h 03m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8926dc2af95947a89f0313283cd42f4b : SUCCESS in 49m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7dbf39df78d2431780a707e857f1424f : SUCCESS in 45m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/278ace380f9c4b49bd35cd95b28410c0 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86dcd748c89b415d890d8eb2f101c026 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f8afe9d79dd46af875c1e3e4ae2e648 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d044d064883d43ac94b56c2a247edbdd : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58c24b316ced4c3b8254a656e3992df8 : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2606b64cbd643088c578106a6614b6b : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"33ed5bea15a55189f56032f30e4abd843cdc783e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 00:05:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"81658b005a16a5da86dd31616a2b05b98bde412c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 00:06:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a2407fd2268406da8e6d526317bc725 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd8b44bf9c484d78b0e9050c3321eed1 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b40d3d908ca674a8bc112397f008ebce0e0661c","revisions":{"5b40d3d908ca674a8bc112397f008ebce0e0661c":{"kind":"REWORK","_number":1,"created":"2022-04-06 18:08:26.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/74/836874/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/836874/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/836874/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/836874/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/836874/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/836874/1"}}},"commit":{"parents":[{"commit":"0f25a8118ed86fe6951473a1129ef87a5052cb89","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0f25a8118ed86fe6951473a1129ef87a5052cb89"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-04 22:16:23.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-06 18:08:24.000000000","tz":0},"subject":"Fix tox docs env to have the correct dependencies","message":"Fix tox docs env to have the correct dependencies\n\nPip is pulling in unstrained dependencies during the \"sibblings\" package\ninstall step. Normally this is fine as the dependencies have already been\ninstalled and it\u0027s just installing the package under test. However, with\ndesignate there are a lot of overlapping dependencies and pip will end up\nreinstalling an unconstrained version. In this case it was Jinja2.\n\nChange-Id: Ie49f99ba8e86913471ee6cd078f5afd28282c0f5\n(cherry picked from commit 30a03e1092b29457e4b79c9c7e33f835f534c857)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5b40d3d908ca674a8bc112397f008ebce0e0661c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5b40d3d908ca674a8bc112397f008ebce0e0661c"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
