)]}'
{"id":"openstack%2Fdesignate~836878","triplet_id":"openstack%2Fdesignate~stable%2Fwallaby~Ie49f99ba8e86913471ee6cd078f5afd28282c0f5","project":"openstack/designate","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-04-06 21:03:37.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"Ie49f99ba8e86913471ee6cd078f5afd28282c0f5","subject":"Fix tox docs env to have the correct dependencies","status":"MERGED","created":"2022-04-06 18:13:34.000000000","updated":"2022-04-06 22:46:13.000000000","submitted":"2022-04-06 22:44:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836878","meta_rev_id":"35c805f1b4361bd646f93a218f4dedd6a0fa1892","_number":836878,"virtual_id_number":836878,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-06 22:44:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-04-06 20:28:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2022-04-06 21:03:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-06 18:42:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2022-04-06 21:03:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2022-04-06 18:15:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-06 18:42:04.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-04-06 19:27:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-06 20:28:08.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"e81db602b1aaf8bd594e9a5f3636824177f47ad2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-06 18:13:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebadb00dfe078766ea1167864fd00ddfe89ed0bc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-06 18:15:47.000000000","message":"Patch Set 1: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7074dc06e5f3ccd875371e299ef1fae2bf48a1c7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-06 18:42:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"783459a33052dd7db8f6dfd724dd3314f7911029","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 19:27:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dd209df83e4f4ea496b1b43c6a2854bb : SUCCESS in 46m 35s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/bec7387a5cd34c4682bc67129aa07ef0 : FAILURE in 23m 42s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/cf3bebdeb30a424abddf5dee11cd69ea : SUCCESS in 46m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0b8858dae1d04fa6ad4ed27b6128310a : SUCCESS in 50m 57s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2feffb509b1649a5936b490f5a557439 : SUCCESS in 1h 07m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/11d48f97a2cb42d09026e8326327e913 : SUCCESS in 45m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a97b2395cdcc4df9893c6d176369728d : SUCCESS in 42m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/983fa9bf057c410c8298d4cdb0854cbf : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce0fc1cd69ae47d09fe127744926bcdd : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df93f2b6a16d408da674e9ce01c096d6 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74ebba48f0d844888717cd03b277fbfa : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e9803fb8c0f4ab7b7c9f6fc47a84319 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/023d0c7f5c65459da4eb35230abdcdc6 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ed21bafd9624e10b9421e194bd51332 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"dba1ace58d38901604196f4b7535d0ebb858f630","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-06 20:28:08.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"38d71dd27283d8ca5a08bc9d92c4b16f7140813a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-06 21:03:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fde421daac3e54b335d23d70ffc44a778182e573","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 21:04:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a694a127d64bf5f10317a706249dad702a82effb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 22:44:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5d1eecd451e942569b9b0f91340480c4 : SUCCESS in 44m 41s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4122ddf941c54a0ca0ce5c7a87d49df3 : SUCCESS in 40m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8e6fb948d600461ca67ff799a3a7a175 : SUCCESS in 37m 59s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/086a68bef04642ecb0492c3394e45412 : SUCCESS in 1h 01m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4ce60da2d259459cbf75adc21b84ca16 : SUCCESS in 37m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5c0f82819f9e4ed0ad3bcf2af1ff9fa5 : SUCCESS in 37m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a70b7b4983bd4218b9b8b7df24c27fee : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3177db38ab0b4ed69958f61d8fc5e252 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4b2ba01236449ea81dc9bbc1284c949 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88610654bfa34bc5b366025512654225 : SUCCESS in 9m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a770fc673d5a4edeab26233242926d0a : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"de3c70daa5b3d3440029f343f0422d064aace4cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 22:44:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"35c805f1b4361bd646f93a218f4dedd6a0fa1892","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 22:46:13.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c64c8c2aa1444ad7a84bbe629278b05a : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c32c8f8a1b7e4529ab056b1ea1371d5d : FAILURE in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5d348de3e8d750c23d5345d1e6aafb32d5b074c4","revisions":{"5d348de3e8d750c23d5345d1e6aafb32d5b074c4":{"kind":"REWORK","_number":1,"created":"2022-04-06 18:13:34.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/78/836878/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/836878/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/836878/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/836878/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/836878/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/836878/1"}}},"commit":{"parents":[{"commit":"0323c4244929aaf199ecbb7c8d32138764029cc2","subject":"Allow TXT record over 255 characters if split","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0323c4244929aaf199ecbb7c8d32138764029cc2"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-04 22:16:23.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-06 18:13:32.000000000","tz":0},"subject":"Fix tox docs env to have the correct dependencies","message":"Fix tox docs env to have the correct dependencies\n\nPip is pulling in unstrained dependencies during the \"sibblings\" package\ninstall step. Normally this is fine as the dependencies have already been\ninstalled and it\u0027s just installing the package under test. However, with\ndesignate there are a lot of overlapping dependencies and pip will end up\nreinstalling an unconstrained version. In this case it was Jinja2.\n\nChange-Id: Ie49f99ba8e86913471ee6cd078f5afd28282c0f5\n(cherry picked from commit 08d56f871311464c7d8f3c1f0c56b2a7f9cd36e1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5d348de3e8d750c23d5345d1e6aafb32d5b074c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5d348de3e8d750c23d5345d1e6aafb32d5b074c4"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
