)]}'
{"id":"openstack%2Fdesignate~839446","triplet_id":"openstack%2Fdesignate~master~I1c8dfec7bd2b5f80b12d7229e47ef8ff70bd283b","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-05-05 14:57:46.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"I1c8dfec7bd2b5f80b12d7229e47ef8ff70bd283b","subject":"Switch CentOS jobs over to CentOS 9 Stream","status":"MERGED","created":"2022-04-26 23:12:34.000000000","updated":"2022-05-05 16:15:35.000000000","submitted":"2022-05-05 16:12:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839446","meta_rev_id":"e0e2ba2fe4a49d0b1c8dc34d9de543e6d9eb883b","_number":839446,"virtual_id_number":839446,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-05 16:12:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"all":[{"value":1,"date":"2022-04-30 13:43:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":2,"date":"2022-05-05 14:57:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-30 20:05:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":1,"date":"2022-05-05 14:57:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-27 00:01:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-30 13:43:27.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2022-04-30 20:05:58.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"0d90de6aac273933b7424c6b6549e6c99f62b06c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-26 23:12:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab5ad9ee3a1a4779d4a5470ab3c1490887c6033b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-26 23:14:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1d0788c9b451c16e09b36b7c6c6f280137b11daa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 00:01:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/32242eeebc7c49c0a500608cb049c774 : SUCCESS in 45m 27s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/06088b5fbe4e45208746d8cd51aec995 : POST_FAILURE in 11m 03s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5bd57acd87e342de89552fd71bd387c7 : SUCCESS in 37m 41s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/d7e2e50c2bcb44f89946e89009ba1671 : SUCCESS in 42m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d161bd09a8b743c6938f7ecdb4340d82 : SUCCESS in 44m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/999c388a3f854b9d9a70cbf56cfd70e3 : SUCCESS in 46m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fd9bd5baf9bc46708d896f38f2194453 : SUCCESS in 29m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0ba98e79fff545feb9cd72a22ebfd483 : SUCCESS in 41m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf49fb6bc5614abeb2e4ecace9be72e7 : SUCCESS in 5m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27fdf0e7d06a40e9b342c8e162533d36 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe034ae8f24b4b05b2323bdfad8c3a05 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6addd93d55b848c2938c68145ee0d890 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4a9e2d3777e4f7580dc50fd3192769d : SUCCESS in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f78a6cd18f4488c9d3a14f2cc50f5a6 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5353e4ea0cd41508893c854c3e51890 : SUCCESS in 6m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/43a91393446a41c4996f7b2ba918f12a : SUCCESS in 27m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1fc7284a6d8644d2be758f8c00faf28e : SUCCESS in 5m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f3da52987abf71c9ed253df1f609c9a4887cddc2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-28 17:34:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b69a24d8c1e9a6e31e1dddabc33e8a1684ae34e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 18:30:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d8bd227ac01746ba9f62d6186ab5c4f6 : SUCCESS in 40m 56s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b971c6a6d178468db990a4f5e053e613 : SUCCESS in 51m 25s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/eeeeb40fc5b04913affc200f0cd14633 : SUCCESS in 45m 04s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/227699302e62400ea051b7e59c07e804 : SUCCESS in 42m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/963386de3f2f46d393ae2063ee997508 : SUCCESS in 46m 23s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b004a947726f447c837a3b38f2023fba : SUCCESS in 45m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4cd99b86fb784357a7859b46a29cc1d0 : SUCCESS in 36m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/30178caabe51408a882c322d43e22e51 : SUCCESS in 38m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08e1c1af8959427fbe949e6cc1a03cdb : SUCCESS in 6m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e98235498614b768f172387332287d5 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b059b9ddc35447feb2461c79e7d4a4e7 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03768d03585942c2ba428e708c03ffe1 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9db0b6637e344ba390fd66f58eb565ee : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3aaaee662e4f4463855b1283d60fddc4 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7008a6b6cebf4eb481ed8949730d9aae : SUCCESS in 7m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/badf4938c4674e599bf6b8c4b943bd27 : SUCCESS in 29m 37s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/32aea5249c73436d814af03290c60ea8 : SUCCESS in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4b84304002a1929f3961498b151ad2f9a39969cc","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-04-30 13:43:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"57eab3a232c03abeee4eb744c5cd1c83c8d28c14","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-04-30 20:05:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"90da778adf06e4fadb6db66d3e846e17773e6158","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-05 14:57:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a0a0104588cd30c311c3fc1ad58bd4bd7b61bf65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 14:58:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d871236a47d7513058c4fd3d7f0f7354e80f739d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 16:12:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1cde901ab37d44c6a966fb6ce18a5ecc : SUCCESS in 38m 54s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2494b49a3a864115955697cc1c70697e : SUCCESS in 40m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6a274f36d01f4abc952f8a138ef5549f : SUCCESS in 42m 16s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/94e4adfc91db4f9897ac42e3d0ab939e : SUCCESS in 1h 07m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0c5159384c4e402388234e95553b2d60 : SUCCESS in 32m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/caad70f0fe2f4f5b832890a71880743c : SUCCESS in 53m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ad91187b9d5447897979eb036ea1fe6 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/659aacb79c654cb197e93f0ffea352bf : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2aaf0139b54944aa8418d88711daab96 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3395a7aee6c54edb93c888bbdd2beca5 : SUCCESS in 8m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a3fa15a9be9d4211a112784607aca2b4 : SUCCESS in 29m 17s","accounts_in_message":[],"_revision_number":2},{"id":"d9ffbac0dc91e516e73f9924cec313526ce49267","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 16:12:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e0e2ba2fe4a49d0b1c8dc34d9de543e6d9eb883b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 16:15:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6c32fc433d34211b9f067a35f0f3482 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6388f75049fa3b21879be1348a343afeb8f66714","revisions":{"85a780095cda17c3ac16634f2d79032ab7c2e11a":{"kind":"REWORK","_number":1,"created":"2022-04-26 23:12:34.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/46/839446/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/839446/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/839446/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/839446/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/839446/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/839446/1"}}},"commit":{"parents":[{"commit":"11b8389389db4cfcd7faa52a67a548b3e3fbb44d","subject":"Merge \"Clarifies the zone import error message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/11b8389389db4cfcd7faa52a67a548b3e3fbb44d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-26 23:09:45.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-26 23:12:32.000000000","tz":0},"subject":"Switch CentOS jobs over to CentOS 9 Stream","message":"Switch CentOS jobs over to CentOS 9 Stream\n\nSince Devstack is dropping CentOS 8 Stream[1], we need to move our jobs over to CentOS 9 Stream. Now that there is a \"devstack-single-node-centos-9-stream\" nodeset we can remove our local CentOS 9 Stream nodeset.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2022-April/028321.html\n\nChange-Id: I1c8dfec7bd2b5f80b12d7229e47ef8ff70bd283b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/85a780095cda17c3ac16634f2d79032ab7c2e11a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/85a780095cda17c3ac16634f2d79032ab7c2e11a"}]},"branch":"refs/heads/master"},"6388f75049fa3b21879be1348a343afeb8f66714":{"kind":"REWORK","_number":2,"created":"2022-04-26 23:14:08.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/46/839446/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/839446/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/839446/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/839446/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/839446/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/839446/2"}}},"commit":{"parents":[{"commit":"11b8389389db4cfcd7faa52a67a548b3e3fbb44d","subject":"Merge \"Clarifies the zone import error message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/11b8389389db4cfcd7faa52a67a548b3e3fbb44d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-26 23:09:45.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-26 23:14:00.000000000","tz":0},"subject":"Switch CentOS jobs over to CentOS 9 Stream","message":"Switch CentOS jobs over to CentOS 9 Stream\n\nSince Devstack is dropping CentOS 8 Stream[1], we need to move our jobs over to CentOS 9 Stream. Now that there is a \"devstack-single-node-centos-9-stream\" nodeset we can remove our local CentOS 9 Stream nodeset.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2022-April/028321.html\n\nChange-Id: I1c8dfec7bd2b5f80b12d7229e47ef8ff70bd283b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6388f75049fa3b21879be1348a343afeb8f66714"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6388f75049fa3b21879be1348a343afeb8f66714"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
