)]}'
{"id":"openstack%2Fdesignate~840110","triplet_id":"openstack%2Fdesignate~master~I1c863e98210aa74f7ecde03b619e971b3112e8d5","project":"openstack/designate","branch":"master","topic":"more_tests","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-05-05 17:11:44.000000000","reason":"Erik Olof Gunnar Andersson replied on the change"}},"hashtags":[],"change_id":"I1c863e98210aa74f7ecde03b619e971b3112e8d5","subject":"Added additional worker unit test coverage","status":"MERGED","created":"2022-05-01 05:35:50.000000000","updated":"2022-05-05 18:07:11.000000000","submitted":"2022-05-05 18:05:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840110-more_tests","meta_rev_id":"1b471ef68d139ef1b3029be2be5d49dfb7a711f4","_number":840110,"virtual_id_number":840110,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-05 18:05:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-05-05 18:07:11.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-05 17:11:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-05-05 17:11:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-01 06:35:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-03 21:56:33.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2022-05-05 17:11:12.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"58792079e5690afe2595d3274ef61aab28f4088b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-01 05:35:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7395c8757bab20859cc99e65c28e40253e3e74da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-01 05:40:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3f6a2a0e2a44a8440fc59dd7cd5945fbe52489bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-01 06:35:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/febbc2ab563042bb9b3cf678f9632cec : SUCCESS in 46m 10s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/0167c858e6b84d75b08d86c551f78703 : FAILURE in 15m 25s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/44dd8f788dd94b468173fda2557b0ad5 : SUCCESS in 39m 31s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/18bfe10ed8554863a6322a5cfc6a6f38 : SUCCESS in 36m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9aa3bb9d5036468ab0fee22114774090 : SUCCESS in 37m 07s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1033ffcc32d446128715106d17f1f722 : SUCCESS in 53m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/481436f9012245a9af73657963c7d190 : SUCCESS in 37m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9479a4d7d7f24da4875cf9e370d409ea : SUCCESS in 39m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db4acff1db0943178ab75e2798d99da4 : SUCCESS in 5m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6cd28058b00c46c38ae61cba509aff43 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/622bf2089b2e41d4adef4353b77755b9 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/166bd8ce5b07467db0e0249ce335ec5c : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec29394832c74b74ae0ff3ab287d26b7 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/832aa2f19b9c4b15b94fb548337aa410 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67d529d4fcce469093d8fe7605afc72a : SUCCESS in 7m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/66a45054dfa2448e939842264041cc33 : SUCCESS in 25m 05s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/11cdd038409a44cfb9520836a8b420d5 : SUCCESS in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ea916bea4dc80b0a12bcbe7b419461ea3c96b72d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-01 18:56:00.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f31623189c64b604e1699856af0c9f2a8d9b0526","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-01 19:05:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6eb71e1e77996155c93645370a1c8b669f2375ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-01 19:32:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1b8fc2b79285bcffcd0101eaf30855cb461814ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-01 20:27:38.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5f395f7cff47469aa82a7887f3f879bc : SUCCESS in 28m 55s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/852eea2feb1a4152a72653d584c9d84c : FAILURE in 14m 19s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a97797c112840f89b8976e7aaa31df7 : SUCCESS in 38m 53s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4cc1cfe4880644d6b6829a009d8696ef : SUCCESS in 39m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b9db3190ca58450eb679d975dfff5caa : SUCCESS in 36m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3bb32464df8b49d3b634d340ec24b20f : SUCCESS in 54m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ca03e30b6115420a8ca146113ed71b87 : SUCCESS in 37m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d2b9c120593742eca570de180579e4b9 : SUCCESS in 36m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01795b4f6068496480e08de87fe21bba : SUCCESS in 5m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36540673399c43da9c8b2bb1b8af22dc : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7829d39dc728441598f29f8a2af49936 : FAILURE in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bf474b0e54b479f92357dd2a6b42538 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43e6ea76f116437c9f15c2021c7c6e69 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b451eda47d7946dea6a643192dc56d5c : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ba4d331376c428cbfcceef12200091a : SUCCESS in 7m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/65ed564e8d944e6e8bd048cef1c46a12 : SUCCESS in 30m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/53f8a49c03b04e2d994981396ad9f5ab : SUCCESS in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"29f38025022e78ce5d5201a69d322a71a56c8928","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-01 20:42:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9d3665c89947c2e7e1b3de57b110729ff5fab867","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-01 21:52:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/20d667fa52a94ab5beba789ef2714f53 : SUCCESS in 44m 10s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/7a049b74df674efaa711d53540de41d0 : FAILURE in 23m 29s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0f0b4ea0a4d40e590bfd765a9ea4c5b : SUCCESS in 53m 02s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e8629473fa6d403c8c14b1b8556ecf1b : SUCCESS in 55m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6f8e395df61d4b398c65e958bc90304b : SUCCESS in 45m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/40710e05cd314e12819d7bd5c475d4fd : SUCCESS in 1h 03m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8bbd1e925b7945b9916efff2a34cb171 : SUCCESS in 45m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/638f413435ff49afbf670bad4e499473 : SUCCESS in 49m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a86b3e4986e409280027b643d049745 : SUCCESS in 5m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2c4fe7c7a234fd28e9b02a98ccd3a96 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbe1cd2cec8041ff9280ab98da50bae1 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/272dc63be3214d3aa4a1f18c207e34d8 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb0c1cbb9f954deaa76b5122425e68e0 : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/46f98fa9b6244269b5361c8795273b4a : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c775a108584f49e596189e7d5ebd3f7f : SUCCESS in 7m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f8bba8e4a5a44bb09794dbc02e00515c : SUCCESS in 34m 02s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/686747fa82244f19990dd2491574f513 : SUCCESS in 4m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ec8119c3de4f02e550dca86ebe51dee591c56baa","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-05-03 21:56:33.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"128a16f6cb773c21d6ad29bb5feb04b274afea09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-04 19:11:32.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"9b91f8a73574016276f15098976e38540104ca05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 21:29:43.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1b4f215084f844b79dfb1f44dbd7cd7b : SUCCESS in 43m 39s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/2276197ae2204686bdb3a85a20823e93 : FAILURE in 14m 41s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a754b044619f4e42afd5311b190d8a6d : SUCCESS in 57m 02s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/011126687e9c4efbad712d6d4c58c66b : SUCCESS in 44m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/318e673f8bc845d5ac0277e100d31c83 : SUCCESS in 44m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f575c50a8b3b4e70be44df6f6512b927 : SUCCESS in 1h 00m 59s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/03e0af0357ea4c0cbd5af91c9e2d7bba : SUCCESS in 1h 00m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/88b1db6f3ac348f8a49a1ec40c9de016 : SUCCESS in 36m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d7aaf7aede1b467a91c6ca6d30c2a66b : SUCCESS in 48m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43b9403d9b33421b8e2286095fd237e6 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a6b78f3452f421299071d1dad38908a : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abdf6c02b9a5468cae0bb745d7ccb2be : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13273479c3194190bb3b2cf73dd270c4 : SUCCESS in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27b3de3bbf97486a9d2df423d665e01c : SUCCESS in 9m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/784c07a9def6427fa1f30ee4e834a5bb : SUCCESS in 9m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d3dce52a59c64cf892902378586c90a9 : SUCCESS in 31m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e68b9e2e06b0465ab82524ac464859a8 : SUCCESS in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"46c6cfdfc9141cf529086b1a9447d36d11ef42d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-04 23:55:51.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"1f369a2b61616f9a666a1ddb41ddb667b0f7df78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 01:12:06.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d8dba56873a94129a9c3d3d4e0a11912 : SUCCESS in 47m 00s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/616db1ad546f480b93efb91f905cec1e : FAILURE in 26m 12s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0eb3234f94bb48f1afbc02dbaf09c810 : SUCCESS in 49m 39s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/09605a862a6e4a148c7cdeeb5c6ba06b : SUCCESS in 46m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/849f1cbc6b1446d6a9aa8d2962ba4c6a : SUCCESS in 50m 16s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/55ca55e26c644d77bddfa818610e93de : SUCCESS in 1h 10m 42s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4b31ce44f7ac450a9085eaf77698f560 : SUCCESS in 45m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e4156c4fc5b04a2999c8e25817cb1482 : SUCCESS in 28m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/156f75a0a6e545f78bf86fe7a85af2b0 : SUCCESS in 32m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34dd054f6b49471fac81d173082a46f3 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5bb209436fc4c61b66f45e7d11e9add : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f801c42c7cf44a885542427cc378637 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f6a6c34548f4ac399ab11fb32eb6267 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8be6de179f7f4bfb8b150a4843309cc5 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45370262995b4ff6b8bd42b7c32e5d58 : SUCCESS in 7m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/309ce374a478464ab83282f1a27f7940 : SUCCESS in 27m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6b5c545d8f124cd1a1f25fe1804a9dd6 : SUCCESS in 4m 20s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"322beb633d1434cc25a14eb31f01f4c0da7cb446","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-05 17:11:12.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f6a31282b2a39ba3895e7a3de616c21405b6116f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-05 17:11:44.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"fe0f4f7f8f14ca801b282945511b011834924e1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 17:12:30.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"66840e5e5ef8d0693e3fd40a06698c6e17503abc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 18:05:37.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6b6be106d057464a981470acfdd3c189 : SUCCESS in 45m 20s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2797d496451647c4865b1338534b2e9c : SUCCESS in 47m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/74344a8f5dd441118fb3b2b8b33e2366 : SUCCESS in 31m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1160504b04074424be8255041c11c772 : SUCCESS in 51m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f82ffacde56f46a787aef1db51fc3e6b : SUCCESS in 32m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f2c322ec9f4249cda4a2fb8df0d63c95 : SUCCESS in 36m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea71d961d8e54dd0ba8efd435bb2e1b6 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/960223c9878a494aaeba589f01942822 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89889400374b414ba4e053cae8cdbf57 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c6deef0dc544654aba0139a3fda068f : SUCCESS in 7m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ff1226b6146743cd8d1eb633e1e611c2 : SUCCESS in 30m 41s","accounts_in_message":[],"_revision_number":8},{"id":"d2dfbeea56148937e7254cd1900e863953144f61","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 18:05:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"6cc6d965dc6619fa6ba0538562cab646d6ee9d38","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 18:07:11.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b7a3758c2be406fac17798548befde3 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"13d98df2932f891957c04ea43a26934372b90d95","revisions":{"aad5e7ee77505f60dcd26a0417eed3b6f2bc1c2f":{"kind":"REWORK","_number":1,"created":"2022-05-01 05:35:50.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/840110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/840110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/840110/1"}}},"commit":{"parents":[{"commit":"77ca0c40ae46371af8a4f9d5a3ad48659dca2dbf","subject":"Validate worker actions before retrying poll","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/77ca0c40ae46371af8a4f9d5a3ad48659dca2dbf"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-01 05:35:31.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-01 05:35:31.000000000","tz":-420},"subject":"Added additional worker unit test coverage","message":"Added additional worker unit test coverage\n\nChange-Id: I1c863e98210aa74f7ecde03b619e971b3112e8d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aad5e7ee77505f60dcd26a0417eed3b6f2bc1c2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aad5e7ee77505f60dcd26a0417eed3b6f2bc1c2f"}]},"branch":"refs/heads/master"},"f2a51b471e6d14db1af6e189b18ab5188e21c969":{"kind":"REWORK","_number":2,"created":"2022-05-01 05:40:25.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/840110/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/840110/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/840110/2"}}},"commit":{"parents":[{"commit":"77ca0c40ae46371af8a4f9d5a3ad48659dca2dbf","subject":"Validate worker actions before retrying poll","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/77ca0c40ae46371af8a4f9d5a3ad48659dca2dbf"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-01 05:35:31.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-01 05:40:15.000000000","tz":-420},"subject":"Added additional worker unit test coverage","message":"Added additional worker unit test coverage\n\nChange-Id: I1c863e98210aa74f7ecde03b619e971b3112e8d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f2a51b471e6d14db1af6e189b18ab5188e21c969"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f2a51b471e6d14db1af6e189b18ab5188e21c969"}]},"branch":"refs/heads/master"},"4f730408994db9acadba921792db930a2fff719c":{"kind":"REWORK","_number":3,"created":"2022-05-01 18:56:00.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/840110/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/840110/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/840110/3"}}},"commit":{"parents":[{"commit":"77ca0c40ae46371af8a4f9d5a3ad48659dca2dbf","subject":"Validate worker actions before retrying poll","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/77ca0c40ae46371af8a4f9d5a3ad48659dca2dbf"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-01 05:35:31.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-01 18:56:00.000000000","tz":0},"subject":"Added additional worker unit test coverage","message":"Added additional worker unit test coverage\n\nChange-Id: I1c863e98210aa74f7ecde03b619e971b3112e8d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4f730408994db9acadba921792db930a2fff719c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4f730408994db9acadba921792db930a2fff719c"}]},"branch":"refs/heads/master"},"466856eeddd9140aee2f4db263f53786c09d5d37":{"kind":"REWORK","_number":4,"created":"2022-05-01 19:05:42.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/840110/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/840110/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/840110/4"}}},"commit":{"parents":[{"commit":"77ca0c40ae46371af8a4f9d5a3ad48659dca2dbf","subject":"Validate worker actions before retrying poll","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/77ca0c40ae46371af8a4f9d5a3ad48659dca2dbf"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-01 05:35:31.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-01 19:05:32.000000000","tz":-420},"subject":"Added additional worker unit test coverage","message":"Added additional worker unit test coverage\n\nChange-Id: I1c863e98210aa74f7ecde03b619e971b3112e8d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/466856eeddd9140aee2f4db263f53786c09d5d37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/466856eeddd9140aee2f4db263f53786c09d5d37"}]},"branch":"refs/heads/master"},"0c281b9bcfbe6553ef9b175cc24bd8e30373baa7":{"kind":"REWORK","_number":5,"created":"2022-05-01 19:32:50.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/840110/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/840110/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/840110/5"}}},"commit":{"parents":[{"commit":"77ca0c40ae46371af8a4f9d5a3ad48659dca2dbf","subject":"Validate worker actions before retrying poll","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/77ca0c40ae46371af8a4f9d5a3ad48659dca2dbf"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-01 05:35:31.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-01 19:32:37.000000000","tz":-420},"subject":"Added additional worker unit test coverage","message":"Added additional worker unit test coverage\n\nChange-Id: I1c863e98210aa74f7ecde03b619e971b3112e8d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0c281b9bcfbe6553ef9b175cc24bd8e30373baa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0c281b9bcfbe6553ef9b175cc24bd8e30373baa7"}]},"branch":"refs/heads/master"},"1f877723d5b47f95eccdc7c28a975a16b1e473de":{"kind":"REWORK","_number":6,"created":"2022-05-01 20:42:33.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/840110/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/840110/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/840110/6"}}},"commit":{"parents":[{"commit":"77ca0c40ae46371af8a4f9d5a3ad48659dca2dbf","subject":"Validate worker actions before retrying poll","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/77ca0c40ae46371af8a4f9d5a3ad48659dca2dbf"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-01 05:35:31.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-01 20:42:02.000000000","tz":-420},"subject":"Added additional worker unit test coverage","message":"Added additional worker unit test coverage\n\nChange-Id: I1c863e98210aa74f7ecde03b619e971b3112e8d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1f877723d5b47f95eccdc7c28a975a16b1e473de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1f877723d5b47f95eccdc7c28a975a16b1e473de"}]},"branch":"refs/heads/master"},"4381287f1d818c47a49193a0222c5b3b246d16b1":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-05-04 19:11:32.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/840110/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/840110/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/840110/7"}}},"commit":{"parents":[{"commit":"dc45a03563bc075242b4d27dfbc6d340166c1c0c","subject":"Validate worker actions before retrying poll","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc45a03563bc075242b4d27dfbc6d340166c1c0c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-01 05:35:31.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-04 19:11:32.000000000","tz":0},"subject":"Added additional worker unit test coverage","message":"Added additional worker unit test coverage\n\nChange-Id: I1c863e98210aa74f7ecde03b619e971b3112e8d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4381287f1d818c47a49193a0222c5b3b246d16b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4381287f1d818c47a49193a0222c5b3b246d16b1"}]},"branch":"refs/heads/master","description":"Rebase"},"13d98df2932f891957c04ea43a26934372b90d95":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-05-04 23:55:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/10/840110/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/840110/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/840110/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/840110/8"}}},"commit":{"parents":[{"commit":"fb152e21a51939ba9b3610b576ad9412008b87db","subject":"Merge \"Validate worker actions before retrying poll\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fb152e21a51939ba9b3610b576ad9412008b87db"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-01 05:35:31.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-04 23:55:51.000000000","tz":0},"subject":"Added additional worker unit test coverage","message":"Added additional worker unit test coverage\n\nChange-Id: I1c863e98210aa74f7ecde03b619e971b3112e8d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/13d98df2932f891957c04ea43a26934372b90d95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/13d98df2932f891957c04ea43a26934372b90d95"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
