)]}'
{"id":"openstack%2Fdesignate~840252","triplet_id":"openstack%2Fdesignate~master~I7876b4c1e2c800b654ca750211ee2e58f3ea4309","project":"openstack/designate","branch":"master","topic":"bug/1926429","attention_set":{},"removed_from_attention_set":{"5572":{"account":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"last_update":"2022-05-04 20:12:06.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-05-04 17:59:18.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"I7876b4c1e2c800b654ca750211ee2e58f3ea4309","subject":"Minimum TTL value allowed is zero","status":"MERGED","created":"2022-05-02 23:35:44.000000000","updated":"2022-05-04 20:13:59.000000000","submitted":"2022-05-04 20:12:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840252-bug/1926429","meta_rev_id":"92e62393c3d6ae56e42f899949827f8f9a338c88","_number":840252,"virtual_id_number":840252,"owner":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-04 20:12:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-05-04 17:59:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-04 19:10:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-04 19:10:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-03 00:01:25.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-05-03 00:33:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-04 19:10:27.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"6a274c43f9fc50e101718a8c62bc45ce02337a1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-05-02 23:35:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a90bdd03adb6b6138f2e91ebd2af17c2c3ed746","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-03 00:01:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f684921d336252e2e4fd654b6c396350e197592e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 00:33:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b76ab032de824e84a3346357f530287b : SUCCESS in 42m 54s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/b14f5ac444cb4a4c90bf5db1ff1f79e7 : FAILURE in 15m 35s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f03f939d6954481ba0d93cceba09fc2 : SUCCESS in 37m 42s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/790aa41bd3364b169f4755fe0c9790fc : SUCCESS in 40m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c0b5e4df78f74b619416ff8a0d097c34 : SUCCESS in 38m 15s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0eda9a136c0f4e8fb367be32ca5e19c4 : SUCCESS in 56m 07s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bca94031def84139bd56aa0d5a08c484 : SUCCESS in 55m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/532ec58db9534164bb4cd32e8addcb87 : SUCCESS in 37m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a43f3c41121e4d71879e68b9fbb013eb : SUCCESS in 35m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/242f7feaaa9e46a2ad29a0097942da25 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b1a497740f640b98ec1edbd5a50c4a0 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cce2b45acbda486ea0f7aa0964133ae6 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f85ccecc7c24d65a8292dfbcae44c4b : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68494bfaf100465b948166522235ccdc : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dee6ddb8db57438187c47604147a2874 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/477b18ec4d2f492bbf6e500284cb6130 : SUCCESS in 3m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9e29374a068c4c1c9e590212cfcdab72 : SUCCESS in 30m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b3bb145c476d447c8e14f80dd2049b9c : SUCCESS in 4m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1f2fdd4cc0a31d358099dc3b15101a1d70f0aec9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-03 17:21:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd17480820a44a7f7a0ac910b49c0694dea033fd","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-05-03 18:22:52.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"781bfd685248d63e708a7ce1957e4e25d715a6bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-05-03 19:30:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bf422a680e5b60d5f21c4d1ee7b09aadc5124833","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-05-03 20:02:34.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cc52fb413aae2d13ff5e60cf34cf2159887dfae6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 20:20:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a5cb7d49fe064e25a0b6f0706c92541d : SUCCESS in 37m 05s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/ac38b18671c04c4bb1fbe550a7fa965e : FAILURE in 13m 22s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1effb9a2d27d4b959353300303b3c847 : SUCCESS in 35m 48s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e58aa2b7305548979e32fc4646feb3de : SUCCESS in 36m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/24ca9569582146699b7200f2f609012b : SUCCESS in 36m 00s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ee294ae1ead24262a3226b8455bba1b0 : SUCCESS in 48m 51s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ed3f5e9661b64645b5876aaad8e42a85 : SUCCESS in 48m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2c1e2a2fc71b4fb9ba45ff62afcd511b : SUCCESS in 35m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7baba544f4fa476084c03b6aca9953a1 : SUCCESS in 32m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/653ce4d1a9b74e25a5f8c45214b74f6b : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0056daa24aa6437591c9a6e11a71c1ec : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e24aa9e84f5e4aff98ad32543a15bc4b : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9377e08a14284da8b40efe379118f1b6 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a594124585cb4884b2549c7c51b3a4b5 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/368725ac5c2d4d6085ffd9304d4fdf5c : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b43eb06c0d2d49b9abce2ef4b5139e82 : SUCCESS in 3m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1eb5e4a3a0c449248f34c38338ad0dbd : SUCCESS in 29m 10s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9585771b6e594bba98bc79dcd399470c : SUCCESS in 4m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"adbeeddf75c8cb023fc0b677bb429a06da5358b8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-04 17:59:18.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"26afaabb1ea647e70befa6e944fa6adca753a70d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-04 19:10:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"75baec9c263225c9e9b9298bf4712c03842fdb5e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-04 19:10:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"49dae54a99dc2985274c6be3763f14503130623b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 19:11:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fee0b045e5cb25a313121c74cec00aca3acd2f67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 20:12:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/47e2f9151e824d90bd1405cc20070845 : SUCCESS in 41m 58s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/0855b4a1f6694b5088856202624dc193 : SUCCESS in 43m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2153c7529b9b48078f96e7c8293b0899 : SUCCESS in 38m 19s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b465ff47e119490d881a2d1e71c05202 : SUCCESS in 1h 00m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1dfda5fad8444f1d9bf32ada33013c83 : SUCCESS in 41m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3b93aee29ee6452e9b546b4577602228 : SUCCESS in 42m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e0282cdce15480f94d52fd0cba28f63 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c4cf6cdf300432c9e2322b8ec2e9cad : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34008d8adb8a433c8e8f7680e01b2cbe : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f578cfd6be684f1091acec226398cf38 : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9edb10df93842608bd7c22b3133fa63 : SUCCESS in 3m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/629122c0131a4e908eaba13779b5b543 : SUCCESS in 29m 18s","accounts_in_message":[],"_revision_number":2},{"id":"a8098647f32c9bc16a56eadfb6b1836128752922","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 20:12:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"92e62393c3d6ae56e42f899949827f8f9a338c88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 20:13:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82bfd9cdd9a84f7fac2133c37990d2b8 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7a88a1e3ca14bf495e773df35974a09 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6ac659d241c65c245d8a090768ac7559c32a13f1","revisions":{"a807c52130e6d912a561b8ba110653630b2edbb6":{"kind":"REWORK","_number":1,"created":"2022-05-02 23:35:44.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/52/840252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/840252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/840252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/840252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/840252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/840252/1"}}},"commit":{"parents":[{"commit":"6e6a4537f2d0061dcbd833fa68b0dd4810e7d73e","subject":"Merge \"Drop lower-constraints.txt and its testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6e6a4537f2d0061dcbd833fa68b0dd4810e7d73e"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-05-02 23:03:49.000000000","tz":0},"committer":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-05-02 23:35:11.000000000","tz":0},"subject":"Minimum TTL value allowed is zero","message":"Minimum TTL value allowed is zero\n\nAccording to ITEF https://tools.ietf.org/html/rfc2181#section-8\nthe definition of the ttl value is unsigned integer and can have a\nminimum value of 0. This patch changes the minimum value of 1 to allow\nfor 0. Unit test have also been modified accordingly.\n\nCloses-Bug: #1926429\nChange-Id: I7876b4c1e2c800b654ca750211ee2e58f3ea4309\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a807c52130e6d912a561b8ba110653630b2edbb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a807c52130e6d912a561b8ba110653630b2edbb6"}]},"branch":"refs/heads/master"},"6ac659d241c65c245d8a090768ac7559c32a13f1":{"kind":"REWORK","_number":2,"created":"2022-05-03 19:30:36.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/52/840252/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/840252/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/840252/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/840252/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/840252/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/840252/2"}}},"commit":{"parents":[{"commit":"6e6a4537f2d0061dcbd833fa68b0dd4810e7d73e","subject":"Merge \"Drop lower-constraints.txt and its testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6e6a4537f2d0061dcbd833fa68b0dd4810e7d73e"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-05-02 23:03:49.000000000","tz":0},"committer":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-05-03 19:28:52.000000000","tz":0},"subject":"Minimum TTL value allowed is zero","message":"Minimum TTL value allowed is zero\n\nAccording to ITEF https://tools.ietf.org/html/rfc2181#section-8\nthe definition of the ttl value is unsigned integer and can have a\nminimum value of 0. This patch changes the minimum value of 1 to allow\nfor 0. Unit test have also been modified accordingly.\n\nCloses-Bug: #1926429\nChange-Id: I7876b4c1e2c800b654ca750211ee2e58f3ea4309\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ac659d241c65c245d8a090768ac7559c32a13f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ac659d241c65c245d8a090768ac7559c32a13f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
