)]}'
{"id":"openstack%2Fdesignate~840557","triplet_id":"openstack%2Fdesignate~master~I91a05bda874658d8fdc051866fb49a5a97a00fe6","project":"openstack/designate","branch":"master","topic":"mdns","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-05-07 01:30:14.000000000","reason":"Erik Olof Gunnar Andersson replied on the change"}},"hashtags":[],"change_id":"I91a05bda874658d8fdc051866fb49a5a97a00fe6","subject":"Remove deprecated and unused MDNS RPC calls","status":"MERGED","created":"2022-05-05 01:38:34.000000000","updated":"2022-05-07 01:30:14.000000000","submitted":"2022-05-06 07:07:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"840557-mdns","meta_rev_id":"dadd081466b6f9e5b7306120fd21a2492f151d73","_number":840557,"virtual_id_number":840557,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-06 07:07:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-05-05 21:12:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-06 06:20:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 03:40:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-05 21:12:55.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-05-06 16:58:54.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"ce7f99e4c5c49bdd75ec10d12c8ad21d449418c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-05 01:38:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"adcbb6e45e399d55eb545e38cae9f615e95372b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-05 01:47:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"05bdd956baeaf5abb0c2c367029c6eb6042222c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-05 02:04:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8b4da833337f96c29f8147448844f49688463961","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 03:40:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d4685f5519064323831c03c2335a9e28 : SUCCESS in 37m 35s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/8f5f185991834064a12ea429a17bb0b8 : FAILURE in 17m 41s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc5d92d7b3a84057bc0046a1f1b3e8e0 : SUCCESS in 54m 19s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f5bf3a7ca8154cd1b3f768f9ffc2e0ec : SUCCESS in 46m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ec6a9f397e794075ac181a851b5a1f6d : SUCCESS in 44m 20s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7a412eee4aa049ab988b179ae6c83d97 : SUCCESS in 1h 12m 18s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b5bd5b3d897a476db4b95e41bc50487a : SUCCESS in 1h 01m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b8d7584378544724bd1d4e22b7119596 : SUCCESS in 44m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e2801e3caafd4b3abdab0b43bc9daf6e : SUCCESS in 52m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3fcac3c196647d7803ab5c735ac75c0 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43be68207d644cc780bdff5c854f5d8f : FAILURE in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6056ced3d775441685d222a275f43ae9 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4fa487873c84405850cdaddde9eff6d : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd600c167fb34ecaa29546e1e97f7f6e : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea8d38ece69746adb07edfd6e6dfecd8 : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b49b4953550747d28daeba5bde8dd126 : FAILURE in 6m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/09a848587eeb48e0aa3a013a9798a22f : SUCCESS in 32m 02s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/44d2cb0d20524c48a63ca06a21e43642 : SUCCESS in 5m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"92ea01f7b64f4a1220d35d1c924cc773d90bbaca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-05 04:41:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a97540c88abd9a79705df219f2e417a5bef98555","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-05 04:49:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bfa83c2225838c2b34cecfc20288903fe80b77bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 05:57:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4ad662511c1742dca428391ae73da7de : SUCCESS in 50m 53s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/6ea780ead77c45b4bbdcf8f124424854 : FAILURE in 24m 42s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/97ce1788f5c04c15a79f7e55c5731b60 : SUCCESS in 45m 01s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e71a7e21dd84444a9caaf3d635d018bf : SUCCESS in 44m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6d205ffb8afe4e88a7bbcf312b4d38e6 : SUCCESS in 45m 13s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f36ebe32882d40e6bb1315bd562e20e7 : SUCCESS in 1h 02m 57s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/898a204489e54f279a29500f19fdc3de : SUCCESS in 57m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6fd1c92d0b994812ae92266f705355d2 : SUCCESS in 29m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ee4a22a38c5847a0abcc33f67e4c7527 : SUCCESS in 45m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad17475b7dca4bd39d5cca609859574c : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd2ab06576d449eaae63a490934fca8c : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6181a8ca9865452395cd2363cc0da57c : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1dcb270e453143ea834bc9a57cf0f2b5 : FAILURE in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd5e714beee145a781ed47c700b8e177 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02cbdee4b65f4493bd6e5a18dd7d18cc : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4b3725be78f49c7a3a3e00225c328a2 : SUCCESS in 3m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/460d9ee922cd49dcb937ecc6b6884b33 : SUCCESS in 31m 17s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/cf224db37abd4d73af92e4f70cc6349e : SUCCESS in 3m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a4ab0a4e589166fe4ee058cce3eb9d5dff1259ea","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-05 06:00:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8756fba0dd81aac663a403087ef43e9d37fe09f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 07:00:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/596777e1635b4e9bb2f7766ada27fe1c : SUCCESS in 30m 47s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/2f8bfc8ebce5487b9d49422184ee76ec : FAILURE in 15m 22s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/4509f419daa14e35928071a074887eb1 : SUCCESS in 38m 45s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2e07fd37c32b4140895d4c7da9fc3c65 : SUCCESS in 35m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/54226157279949148897a488949cda64 : SUCCESS in 37m 27s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/349c9ad24ca64522b5fbe1c1155c9c67 : SUCCESS in 43m 24s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/81df9e2458544753a3641f0a50957d4a : SUCCESS in 57m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8d6d0beac5c8432c9f500f1b7d977644 : SUCCESS in 41m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3415451b349f4db1a1f60105b051c006 : SUCCESS in 40m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c3bd03878204cdd93817b6608849169 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/238538171b75496f83b68918060d4e73 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2343fab3a9c4449383dd9831dfca5c48 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/917c86f515ec477fa52feb8ea6dea9b0 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45ed2717479548ffb642d36eee7ddb64 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48d0070af266402b83e63ba813c776a9 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d71a874bbeb42c28e026f6aa8e920eb : SUCCESS in 4m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/03cc48abc6984ff4ae04a037fbbfb49c : SUCCESS in 26m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/16283c071baf4901bc5cfdab75253f05 : SUCCESS in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d65212bfeb1bb79c75396b4cba9f5efdee5513ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-05 19:17:51.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"44c148722e03f8ad20d9b161553638f4011af980","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 20:14:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/09866074c9624663898d1e1e35d73024 : SUCCESS in 37m 25s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/dcfe763598b84318844b3961aa7747ec : FAILURE in 10m 26s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/76bb91dfe9d64b3587d382aa65588738 : SUCCESS in 34m 49s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/29b63ea93223439094ca15c4fb66f908 : SUCCESS in 32m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d0cf3095fe8d40e3bd76ce9cd09e6a30 : SUCCESS in 44m 28s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/53dd60b2017c4f1f9ddc82309c234a72 : SUCCESS in 44m 27s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/93e378ef5716466d86f69450f27cde52 : SUCCESS in 54m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2e2bd8ea5b2541d0944c4c6310eaf768 : SUCCESS in 33m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fdda7a9ae331477f9ccaa7629daefba5 : SUCCESS in 36m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c69baac4ff384881b291c9a7bb01f5a2 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ad0f27669434148997c598823a3ec30 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/995170e915bc452aa75fadcea820120b : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e3b7a43bcbf4519a59a16e28e368c69 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7a9163cee6742e09532a772f8569dc6 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/199b1a70f5734d5289f7880c13b77a84 : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7d7374c0caf4af7ace44b29fdb22f39 : SUCCESS in 4m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d4f54312fd194daa9e4800d8a863efef : SUCCESS in 28m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/15ac0e59bf9b4b878a71975bfb883afe : SUCCESS in 4m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"822a16e3c79334f1c50335921c93940aa0354459","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-05 21:12:55.000000000","message":"Patch Set 6: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"24fb5fe2be674ae4c6ea8b71f5775aefc0f48ce6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-06 06:20:54.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"819d65d9ef5924c378aa6857655b98cbc7bf1214","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 06:22:01.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"95e3af980bbe4f0c5554badad363b2053189a41c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 07:07:05.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fa905fe3efe94db9a14d5c97ce01bd74 : SUCCESS in 35m 31s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e7bf800838a24bc78442050cf75d3970 : SUCCESS in 33m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/63fb8b1ecda748aba6f70bb67e3a50b4 : SUCCESS in 33m 29s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4721643893d943c9bfb1ceabdee75823 : SUCCESS in 44m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/43e3e9f4cdc04695826dbe833b2c4f63 : SUCCESS in 30m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cc0729744756400ebe811020fc2cda68 : SUCCESS in 34m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c68eb1f11ee744e4986eee5358a008ac : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df67585ad9c74add8ce11f2c3626f92c : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fdfda403dc449e58e89ffe279eac612 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/824a6ac5cf214c9aa4b36e3fcac63b26 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5168d80039a45a2a342024535e68f64 : SUCCESS in 3m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/849e7f4dd4b245338fd8caa6f3a5b2a0 : SUCCESS in 25m 54s","accounts_in_message":[],"_revision_number":6},{"id":"d22a7214a99d2329561078c005cc7707a689bb80","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 07:07:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"35df4411491ae4c0013b01445e7cdba6880e5f93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 07:08:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7827604a39464a13b4b048e0a6db5049 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18cb98f8779041488ef7eddba93fade8 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":6},{"id":"5750398662e9e4237b295a0333f1219959b983fe","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-06 16:58:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"dadd081466b6f9e5b7306120fd21a2492f151d73","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-07 01:30:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"5d5d83e511acbf5d6f34e9a998ff16d96c5bb162","revisions":{"73c08aad67d7dd2195437dd7ac4b9ec5ef61e788":{"kind":"REWORK","_number":1,"created":"2022-05-05 01:38:34.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/57/840557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/57/840557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/840557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/840557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/840557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/57/840557/1"}}},"commit":{"parents":[{"commit":"e5d2fc0657ff71800283b0211b00a32e6e320b6f","subject":"Cleaned up agent backend code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e5d2fc0657ff71800283b0211b00a32e6e320b6f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-05 01:25:56.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-05 01:37:46.000000000","tz":-420},"subject":"[WIP] Remove unused MDNS RPC calls","message":"[WIP] Remove unused MDNS RPC calls\n\n- Remove deprecated mdns notify option.\n\nChange-Id: I91a05bda874658d8fdc051866fb49a5a97a00fe6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/73c08aad67d7dd2195437dd7ac4b9ec5ef61e788"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/73c08aad67d7dd2195437dd7ac4b9ec5ef61e788"}]},"branch":"refs/heads/master"},"5c5aced4223dab91c78216fc301aa9342b1ef3e8":{"kind":"REWORK","_number":2,"created":"2022-05-05 01:47:10.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/57/840557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/57/840557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/840557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/840557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/840557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/57/840557/2"}}},"commit":{"parents":[{"commit":"e5d2fc0657ff71800283b0211b00a32e6e320b6f","subject":"Cleaned up agent backend code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e5d2fc0657ff71800283b0211b00a32e6e320b6f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-05 01:25:56.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-05 01:46:59.000000000","tz":-420},"subject":"[WIP] Remove unused MDNS RPC calls","message":"[WIP] Remove unused MDNS RPC calls\n\n- Remove deprecated mdns notify option.\n\nChange-Id: I91a05bda874658d8fdc051866fb49a5a97a00fe6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5c5aced4223dab91c78216fc301aa9342b1ef3e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5c5aced4223dab91c78216fc301aa9342b1ef3e8"}]},"branch":"refs/heads/master"},"d0d95d0ba605493e7168472034291949d1414ef2":{"kind":"REWORK","_number":3,"created":"2022-05-05 02:04:56.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/57/840557/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/57/840557/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/840557/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/840557/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/840557/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/57/840557/3"}}},"commit":{"parents":[{"commit":"e5d2fc0657ff71800283b0211b00a32e6e320b6f","subject":"Cleaned up agent backend code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e5d2fc0657ff71800283b0211b00a32e6e320b6f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-05 01:25:56.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-05 02:04:28.000000000","tz":-420},"subject":"Remove deprecated and unused MDNS RPC calls","message":"Remove deprecated and unused MDNS RPC calls\n\n- Remove deprecated mdns notify option.\n\nChange-Id: I91a05bda874658d8fdc051866fb49a5a97a00fe6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d0d95d0ba605493e7168472034291949d1414ef2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d0d95d0ba605493e7168472034291949d1414ef2"}]},"branch":"refs/heads/master"},"7675f43aae626bfa6a0551c0c54134172db2e071":{"kind":"REWORK","_number":4,"created":"2022-05-05 04:41:04.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/57/840557/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/57/840557/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/840557/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/840557/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/840557/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/57/840557/4"}}},"commit":{"parents":[{"commit":"e5d2fc0657ff71800283b0211b00a32e6e320b6f","subject":"Cleaned up agent backend code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e5d2fc0657ff71800283b0211b00a32e6e320b6f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-05 01:25:56.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-05 04:40:50.000000000","tz":-420},"subject":"Remove deprecated and unused MDNS RPC calls","message":"Remove deprecated and unused MDNS RPC calls\n\n- Remove deprecated mdns notify option.\n\nChange-Id: I91a05bda874658d8fdc051866fb49a5a97a00fe6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7675f43aae626bfa6a0551c0c54134172db2e071"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7675f43aae626bfa6a0551c0c54134172db2e071"}]},"branch":"refs/heads/master"},"0557413007cbe9f7d7166a15ece02cc5459777f8":{"kind":"REWORK","_number":5,"created":"2022-05-05 04:49:39.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/57/840557/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/57/840557/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/840557/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/840557/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/840557/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/57/840557/5"}}},"commit":{"parents":[{"commit":"e5d2fc0657ff71800283b0211b00a32e6e320b6f","subject":"Cleaned up agent backend code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e5d2fc0657ff71800283b0211b00a32e6e320b6f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-05 01:25:56.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-05 04:49:29.000000000","tz":-420},"subject":"Remove deprecated and unused MDNS RPC calls","message":"Remove deprecated and unused MDNS RPC calls\n\n- Remove deprecated mdns notify option.\n\nChange-Id: I91a05bda874658d8fdc051866fb49a5a97a00fe6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0557413007cbe9f7d7166a15ece02cc5459777f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0557413007cbe9f7d7166a15ece02cc5459777f8"}]},"branch":"refs/heads/master"},"5d5d83e511acbf5d6f34e9a998ff16d96c5bb162":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-05-05 19:17:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/57/840557/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/57/840557/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/840557/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/840557/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/840557/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/57/840557/6"}}},"commit":{"parents":[{"commit":"95dbf944868ae9cdfe67f8dec0466c07fec98fb1","subject":"Merge \"Cleaned up agent backend code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/95dbf944868ae9cdfe67f8dec0466c07fec98fb1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-05 01:25:56.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-05 19:17:51.000000000","tz":0},"subject":"Remove deprecated and unused MDNS RPC calls","message":"Remove deprecated and unused MDNS RPC calls\n\n- Remove deprecated mdns notify option.\n\nChange-Id: I91a05bda874658d8fdc051866fb49a5a97a00fe6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5d5d83e511acbf5d6f34e9a998ff16d96c5bb162"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5d5d83e511acbf5d6f34e9a998ff16d96c5bb162"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
