)]}'
{"id":"openstack%2Fdesignate~841046","triplet_id":"openstack%2Fdesignate~master~I4fab6722d98dd3e188602c2808aefa7691810748","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-05-13 18:02:38.000000000","reason":"Erik Olof Gunnar Andersson replied on the change"}},"hashtags":[],"change_id":"I4fab6722d98dd3e188602c2808aefa7691810748","subject":"Move Zone XFR from mdns to worker","status":"ABANDONED","created":"2022-05-08 22:46:08.000000000","updated":"2022-06-13 23:50:46.000000000","total_comment_count":1,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"54c6415bbd5f5a28e01a2274afe89b5a2a4dcd3a","_number":841046,"virtual_id_number":841046,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-12 22:33:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-08 23:47:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a6f63c81c21fe47bee3f7f9821f218a4f474db20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-08 22:46:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"688215bc2b1f58fbf6b1981b900c8558d6ea3b94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-08 23:47:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/19b2257c3060458bbf6ab3a7b39484c4 : SUCCESS in 42m 25s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/eb592f399e504701bcc609a33b6dc81d : SUCCESS in 43m 55s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a82d2dbdb878482eaedd595fb679bcd1 : SUCCESS in 42m 43s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/924b97839535460e9d9a395ae90622c4 : SUCCESS in 39m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7dc47499df844cc0a9731d86382d75a0 : SUCCESS in 38m 19s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a8b20b2a12e74a409d057fbb78fed68c : SUCCESS in 54m 16s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9970f674f9a745e2bd7de734022a9a0b : SUCCESS in 52m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ae803007e570420c8d6695e07797c88c : SUCCESS in 26m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5384b05c07c64722905d1ddaf76e14a0 : SUCCESS in 41m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24386b2d8760462280e077e00b440e66 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e6a1c43788e41d6b928ff5067321cfe : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffbd85ebf2a44e859ebfab6d538a17b2 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afd7f9385a104490a1807f2c0cf0af9c : FAILURE in 8m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b001677f4f4b451eb1f5600c1fe667e8 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ff49afe921842a5a4185fd3edad79e4 : FAILURE in 6m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/472a0326d74a4b17abdc9fe6507ce83d : SUCCESS in 34m 04s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ce7800be73204be0af016f5b353cdc71 : SUCCESS in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed373e5f320f1b8caa8185c653567400800a1268","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-09 04:40:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"da5fbfef12bc98d9ba0bb409a6a7e9e38726bcfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-09 04:44:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3541690b709a0994203afa4abae5f5be3b40a92d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-09 04:54:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e61718480fb2df0367ab25bcfc9b5bb141ee2d33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 05:39:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2e67491eb66543969d43bdf695a29082 : SUCCESS in 31m 51s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2a7dec019bf24f4a8135393b047ccd4a : FAILURE in 9m 28s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d216e6e03974eb693dd2d520d94025c : SUCCESS in 34m 02s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2fce944b19ea4f7f821c0d001b5e7791 : SUCCESS in 30m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/774a168ea5514662adab27e99a0b36f3 : SUCCESS in 32m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/65b61b81ed244b7386af91dbe845da18 : SUCCESS in 44m 08s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1faba3e2138c46329ee7de5d7b8c4722 : SUCCESS in 42m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7881f28cce8641b6ba8bc9e845615c80 : SUCCESS in 28m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7e67cb5ed3d549a39d3ae2d0b6c4a537 : SUCCESS in 35m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ad98c1da0c1496e8b8a6197d2fdd4aa : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cd48f77149f498486238db93fadf715 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05bf943a7bf64b4987acfa15b98c5e40 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9ca94368faf4e46b181bb9a1b8cd419 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a35903d8eb4345e2bfe4be46fefddd65 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d62bc88859a46d488dcea79cb229f59 : SUCCESS in 7m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/edee2aa8118549d6b883712215166f68 : SUCCESS in 32m 17s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/514f8f45f8d54ae89f80e0440e48a60b : SUCCESS in 3m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8ddb866c4c8abd552c52d21bd2ea4683cb540ee6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-13 03:47:15.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"ca33c5347f09084097fb26019c178ece74529fbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-13 04:06:41.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"aa5fa0df97b65857a2ab216656ee2a2f99e6dc27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-13 04:09:45.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"596fa0fe4a232afce181bf829751a6bfe70a7a88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-13 04:21:36.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"cb5d5d0cd2dc23dcfa36ab411b78076c60ef8fe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-13 04:23:53.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e37d67d35c2198354aaf9c1ee842a37e9d964059","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-13 04:32:21.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"d48de24ae88afb293be86e194976759fabb80d4a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-13 04:33:33.000000000","message":"Topic fix removed","accounts_in_message":[],"_revision_number":10},{"id":"4c923519bad7417b140a6e00323f517409d46f01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 05:24:28.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d9ba204221874dfeadc0f8f23144ad74 : SUCCESS in 37m 44s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2ac8dc51334f4643a6d0eb0c123770fe : FAILURE in 13m 30s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/43b8029471524e4495a9a7522bf0bc51 : SUCCESS in 33m 39s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1f3f868d66354554afcaf580e8b5b249 : SUCCESS in 37m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b9e1fa85adb243a394ee732bc2611e9c : SUCCESS in 35m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ce72f56fc63642e9a5b7f0d54b7204d8 : SUCCESS in 50m 20s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f9bbaf60ad794ab3a80563771f26bc4b : FAILURE in 45m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2bf5846808f54f508895294b4238d782 : SUCCESS in 32m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a1f1992bbb8f488f94a34b4e11828507 : SUCCESS in 35m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4a0bcf0e46f4bce8b6572551b46aa21 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cab65cd4cff4c4d984a1b46c0a3f1b0 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35b8bbc22b9040f3b8e30af1d5933794 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b425058f8579485d9e2d273e63a2b68b : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2983fff30c2492f88a4e53a05d1437e : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e5f0d8302bf4ceeb47ad1662d21455f : SUCCESS in 8m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b393a7d78bd4430bb331f39a00fd5228 : SUCCESS in 32m 57s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/397eb1f819ae4ce5a6b20b3790912794 : SUCCESS in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"eb300e3315eb7f53d310550b147f1b4cb3c723db","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-13 18:02:38.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"daf8ef6d112edf2eb8fe1a38a6e61858be8489e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 19:18:10.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4083cc901f63442ab16cf58b745844a7 : SUCCESS in 40m 53s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3be993ce20444a57bad63b03814b460a : FAILURE in 12m 23s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/40088938edab48d084a23ba3c235b559 : SUCCESS in 44m 18s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4490b665f66042a8aa2a5e6807443ebd : SUCCESS in 41m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/069d05ef8eae45249f2896107aa7d22e : SUCCESS in 39m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/08f3881522fc4d928373264b73ed7047 : SUCCESS in 1h 04m 16s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2948a09812e541c6b2529f3f426ed623 : SUCCESS in 1h 09m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/eef109fba3b348dd9f7bb8bdd1cf8891 : SUCCESS in 38m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ecb36f36283740c987a4502e4b0f4bff : SUCCESS in 49m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff122ba7f0a34e5d80b5fd284bf0ee21 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2e7b64880054155b50fdcd34f7556d6 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82a792592c894e7188549d1000b947a2 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bd30a0f2e534d7d830dcb1f9d53fe67 : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7acfd7ab51544aeeb92e0123a0575564 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b2d8d4a55904b569aa2fc1e0992a33f : SUCCESS in 9m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d48860e4b2ed43d88b6ea58cbbd8ba36 : SUCCESS in 27m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/50cc5b7e2641448884371b2a50bf45d9 : SUCCESS in 5m 08s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d2bcd0e542fbefe8c42bd8e4e06c23016ba441db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-14 03:15:29.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"6f3897862494deed81695e5624855a8d6feae8fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-14 04:32:58.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5ed3d1731aa141eb8c8a53b5c9c7e359 : SUCCESS in 47m 51s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/aa58ae25a57649a197bb33626525d5c4 : FAILURE in 22m 05s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc964f5e7b8b46809bd9023cf64e3395 : SUCCESS in 54m 19s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/fc1cd7b49a2a4a8895f4b294345aa5cd : SUCCESS in 51m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4eeaf5ac260d47ae9dc99bc2967cf63e : SUCCESS in 52m 32s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/87f2a3167658402b8bc79e5cc35fcade : SUCCESS in 1h 12m 03s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c3cb9a63556247819c46c8d386066bf4 : SUCCESS in 55m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6c254c7855ba4461b83461323b09cf5b : SUCCESS in 45m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/689faa67a0d044f6ab2a5f4482d40b51 : SUCCESS in 43m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd5a7b5102754f10a57526cf190e8545 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf5ab5762b1f4dd5ba9cd3eea21a7194 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57b030319a95460aba346b79effa5226 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f6ef74f52904a12bb10629887ea0be5 : SUCCESS in 6m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c27011f1e08486ebacf7ac72df06d9c : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2be060e7eaf146c2ae962d130c20c47d : SUCCESS in 8m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9a6095f7084d4bc9ab34f059c6804a82 : SUCCESS in 36m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4547e5a488494850a093064a66d8536b : SUCCESS in 6m 15s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"beea930e405b809fd110ed05f0598514e0e6a74c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-15 05:38:12.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"f2de3cecbd241cf5b8f4abec2d49b3731bf47d52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-15 05:49:37.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"e0d75b96d5fe7788c0efee4ec3b55337b051bc89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-15 06:46:04.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4335d52bfb044f67bec8fc974c241dd7 : SUCCESS in 49m 51s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1ded7f79d7944212ad79f4100e3b7de9 : FAILURE in 10m 01s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc510a5942e74822af77662cb139c8c8 : SUCCESS in 35m 37s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/a39ad8166bd5437190a2aa49b7c7e67d : SUCCESS in 34m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/24cbb597179846b49a3f38d10067218d : SUCCESS in 37m 48s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/88e1a55138d74ebb98e2e57824a26bf4 : SUCCESS in 48m 19s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/edd314df8bde438f990cfd408a970444 : SUCCESS in 48m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/df512765f3584596b01341e5094a146c : SUCCESS in 35m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/42bc140381b24e34920a341985f312e4 : SUCCESS in 39m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/761fe765f67848e18d9934c4195990e5 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5805ffbe09f4998b30c674430ee4c55 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/928ebcbfc5c34acdbd30616075639637 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/656977b3054b4d8fbb88879ecc84c848 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a5461130892849668da52bb86768b399 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52bfa661d37546b4a355a04c055cc411 : SUCCESS in 7m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2b1cdb5a7bf844b8b38fc6024cbec7c5 : SUCCESS in 28m 58s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/be5e45b5ee594f1d987ac18e9aa7131c : SUCCESS in 4m 23s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"4aa5bc387952414f083100094bf790f6d308fe84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-15 06:57:29.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"64c38a66ee7985794074d6811774c5a06b16e987","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-15 07:57:36.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/833eb9b126b2461b814425f4889cafdd : SUCCESS in 47m 49s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3a11b8dd56d147628965caaf3119ec97 : FAILURE in 11m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c53e5510764e40c2bc0328c9198387a6 : SUCCESS in 43m 05s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/d3508c5234b042f8b31990db3d757a27 : SUCCESS in 42m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f5cb06ca9dd1420fb29a78772ac23c44 : SUCCESS in 46m 43s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/986bcca7f4cf4d86a12b9e73791668e8 : SUCCESS in 58m 38s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c7cb271472ba4720a15e2c378e6ce40f : SUCCESS in 44m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b893526681d347f9a2429f8ba32b7ba6 : SUCCESS in 39m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ea9be4f075ff4ddc97ca854657a2a436 : SUCCESS in 49m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/397c60268833408b858ce6c2240288e3 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a962fcf9f4e4085bb4607dc660b00f7 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aab7e922fdf14b3497c2797008cc4e38 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/194d491785ac4f57a9280fc922f38c6d : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/933dd96e66f84c38a3898c3c84925a75 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19e09f353830474cbfc783357bc5cc12 : SUCCESS in 8m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8070d50218df4e07be7ec06b446b3a6a : SUCCESS in 29m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/52c610dc69a9444da135acc3084a27f0 : SUCCESS in 4m 13s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"d9bece5f7f0973fdaf717b4ea9a0e5cd1f3efd73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-18 05:19:29.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"db258379082cfb6175b756d3d40ba96fb6c129e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 06:22:38.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9e40ec65f5b143b0aa21366f460bd804 : SUCCESS in 42m 14s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/349ebc9cfb35417da400c1cfb37e8eb0 : FAILURE in 22m 43s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2960d03cb4e426d8203e030919a42a4 : SUCCESS in 42m 48s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3a8bd3cb96dd4b48ae1e0016b2358357 : SUCCESS in 37m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/34560922d0f44b4fbb37e1bad31f7659 : SUCCESS in 49m 43s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/10285a557dda47df8656fc0bc353ec4c : SUCCESS in 1h 01m 33s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bf4206b7b05c40068760a96c5d945d49 : SUCCESS in 55m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0e2b0a009e954362a40fcf2bbac7a60f : SUCCESS in 50m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7d083328e891417e8f73ef4a5f9b03ec : SUCCESS in 40m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a17769d157b4b00b2e367d96c26024d : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d62af3fbfc547d78ea14bd609e53246 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7856aa72a5e443281b2f9b044e6c3e7 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16ed735c28c74d15af72106da1a34c83 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be5233f71ca14c3dad2fc52eb8923a19 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98154aeee9834d6fbab30c18864a4305 : SUCCESS in 9m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/abfa120d57bc47bd8687082802eae3b5 : SUCCESS in 30m 21s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/77934c96051146dc884c9ff09ce641cc : SUCCESS in 4m 31s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"05d4dcc9c2cae5abfe37e444921252ae5457ad3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-25 03:02:09.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"3e887cb275a4fb25bdfa67efe224cb6d0ab181d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 04:22:06.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b6f0f131e9af4150838eacdf0ec1e853 : SUCCESS in 54m 23s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/abe032d29f9b4ccabb1519a943ad67fa : FAILURE in 24m 10s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/512a754e48a3403b91a119c38b0b0181 : SUCCESS in 40m 32s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/34afc4c106054007b9eae4101796ab4d : SUCCESS in 36m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8c3513cffaa340faaace09c2b6479cca : SUCCESS in 40m 02s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a7853719f71342c6ba3fb84ede7a4e3b : SUCCESS in 1h 02m 38s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/028d72636a294fbc88b2a7a8f29702c1 : SUCCESS in 1h 08m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b0fb74e2726b44088304eca2e74b9365 : SUCCESS in 50m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/47003c474fd54735b8fe1d9e3a809d61 : SUCCESS in 52m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9c35d4004bd4ad4883c826e4e22c63f : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f04e67f1560481a8bcc9c46f1888a82 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7746f7155aae4623812aaaca0725a4b9 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a83f5d738124463acce6d8744f095b0 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97bb5ae584e34120a2e30df272f9b5bf : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2eefe4c86b640b4bb226f12750424f2 : SUCCESS in 7m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/865607ca50e04156b3d10133cc675ad9 : SUCCESS in 32m 25s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c7619ae9e66d4ecf968c4bc6bf616939 : SUCCESS in 5m 06s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"6814f2b931de5747037a511f23b46d1d21703f58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-27 17:40:07.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"4297f2e591b165478ca883f2d39075c23e7539f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 18:45:36.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8c139a7aae0f4065be02356a3e34713b : SUCCESS in 47m 39s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/078567eef584402683dbcbe092fbf541 : SUCCESS in 59m 52s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/58d66b84d4d94673a1e718ac5bf5a5d2 : SUCCESS in 37m 49s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2949868bf63940b3b87ef76ce32db73b : SUCCESS in 36m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e563ff0f93874e8995a7ec23e5ef63e8 : SUCCESS in 39m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/361ac6907cc349408241bd2e87162e89 : SUCCESS in 57m 37s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/387eeae4184a44979182d6444c6794e3 : SUCCESS in 56m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/87804151ce134857af4e6e8aeed9bde4 : SUCCESS in 40m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/516d386a4b3d43b9883f9d2d521e9741 : SUCCESS in 41m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/435553831c274dea8e074a90901b2d9d : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/058b60b37354494d86968cc0136abb39 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15ee7ca6e9864c6191d7347d1045caac : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/356c9ce49b8b48b89db2bb97d3627b93 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd9f3629ca984ec3bfc1e9a2134d9308 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbdc07f440ea4c82ab7851c5e0409720 : SUCCESS in 7m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/165f11eecced43eebc67c700d35350a5 : SUCCESS in 31m 07s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5c880bc4c4e64bc8ae74e2ff4a59ab12 : SUCCESS in 4m 18s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"dadce43d3ea10738d61da5a8ad48866e0d4c2892","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-12 21:12:21.000000000","message":"Patch Set 18: Patch Set 17 was rebased","accounts_in_message":[],"_revision_number":18},{"id":"eb41ed8949bd811b30705c88c04e3a656df479a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-12 22:33:15.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c8d1176b8559463380c2d235c4f3276a : SUCCESS in 46m 15s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/73ec10f835204b89a83e8674ffeca2f0 : FAILURE in 22m 36s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d076784c6f414f4fb4b7a1b6125f3479 : SUCCESS in 54m 19s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/595e71c065774eb9a0af2395499f1102 : SUCCESS in 45m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/25f9c801f9b3432f853d01ae9c201522 : SUCCESS in 42m 19s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/078bd87d4db14493a866c7cbd65853f1 : SUCCESS in 1h 13m 47s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7bb865e947154b58a79814c664e8e068 : SUCCESS in 1h 10m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a82d0da690634bef9e087f53324f171c : SUCCESS in 51m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e353b3d1ff4449bca8f95e1bb13cbaf7 : SUCCESS in 42m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2244b1886e664208af1c58d0da606b73 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a97346a9e3ba4d95a00841b8f953b067 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f07ea9e7418e4fcdb7dec7055ba5dff9 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf2aea1bc1c54b3f954986f60aae0edb : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/531c29b3fa5045f59355ddfce5166d69 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60a4178fec704efaa89d2085ad242592 : SUCCESS in 7m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/68e711eb1eb5413db6363018e60b70e4 : SUCCESS in 31m 18s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d8f15ca2968847aab711600b8422f4d9 : SUCCESS in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"917e4e15df56b4d5fbd9aa40db197584cad33523","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-12 22:37:51.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":18},{"id":"54c6415bbd5f5a28e01a2274afe89b5a2a4dcd3a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-13 23:50:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"99f55f04654c1f8ba7859af304962a86ecc5e8cf","revisions":{"ce0db6978cf2aaf1b9275af99058a9b532fdf783":{"kind":"REWORK","_number":1,"created":"2022-05-08 22:46:08.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/841046/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/841046/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/841046/1"}}},"commit":{"parents":[{"commit":"5773881dda6f4f27543b9c47cb3a19c7781adc6b","subject":"Fix issue with mdns rpc and make mdns safer to upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5773881dda6f4f27543b9c47cb3a19c7781adc6b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-08 22:45:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-08 22:45:39.000000000","tz":-420},"subject":"[WIP] Move Zone XFR from mdns to worker","message":"[WIP] Move Zone XFR from mdns to worker\n\nChange-Id: I4fab6722d98dd3e188602c2808aefa7691810748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ce0db6978cf2aaf1b9275af99058a9b532fdf783"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ce0db6978cf2aaf1b9275af99058a9b532fdf783"}]},"branch":"refs/heads/master"},"7bb54744710606e774160bf928165f32bc618392":{"kind":"REWORK","_number":2,"created":"2022-05-09 04:40:15.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/841046/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/841046/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/841046/2"}}},"commit":{"parents":[{"commit":"5773881dda6f4f27543b9c47cb3a19c7781adc6b","subject":"Fix issue with mdns rpc and make mdns safer to upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5773881dda6f4f27543b9c47cb3a19c7781adc6b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-08 22:45:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-09 04:40:06.000000000","tz":-420},"subject":"[WIP] Move Zone XFR from mdns to worker","message":"[WIP] Move Zone XFR from mdns to worker\n\nChange-Id: I4fab6722d98dd3e188602c2808aefa7691810748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7bb54744710606e774160bf928165f32bc618392"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7bb54744710606e774160bf928165f32bc618392"}]},"branch":"refs/heads/master"},"6b73c6f7087b1f8a72bd468e8ec39b422a524560":{"kind":"REWORK","_number":3,"created":"2022-05-09 04:44:22.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/841046/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/841046/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/841046/3"}}},"commit":{"parents":[{"commit":"5773881dda6f4f27543b9c47cb3a19c7781adc6b","subject":"Fix issue with mdns rpc and make mdns safer to upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5773881dda6f4f27543b9c47cb3a19c7781adc6b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-08 22:45:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-09 04:44:14.000000000","tz":-420},"subject":"[WIP] Move Zone XFR from mdns to worker","message":"[WIP] Move Zone XFR from mdns to worker\n\nChange-Id: I4fab6722d98dd3e188602c2808aefa7691810748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6b73c6f7087b1f8a72bd468e8ec39b422a524560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6b73c6f7087b1f8a72bd468e8ec39b422a524560"}]},"branch":"refs/heads/master"},"48282b16da7d87c80ef3210243a6f3e8603deb12":{"kind":"REWORK","_number":4,"created":"2022-05-09 04:54:18.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/841046/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/841046/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/841046/4"}}},"commit":{"parents":[{"commit":"5773881dda6f4f27543b9c47cb3a19c7781adc6b","subject":"Fix issue with mdns rpc and make mdns safer to upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5773881dda6f4f27543b9c47cb3a19c7781adc6b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-08 22:45:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-09 04:54:09.000000000","tz":-420},"subject":"[WIP] Move Zone XFR from mdns to worker","message":"[WIP] Move Zone XFR from mdns to worker\n\nChange-Id: I4fab6722d98dd3e188602c2808aefa7691810748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/48282b16da7d87c80ef3210243a6f3e8603deb12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/48282b16da7d87c80ef3210243a6f3e8603deb12"}]},"branch":"refs/heads/master"},"0f44418feed04436884bf2264a0ef6e95536447c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-05-13 03:47:15.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/841046/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/841046/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/841046/5"}}},"commit":{"parents":[{"commit":"f0cea06f1ad5f9e9a6558b215b982df9facb30fd","subject":"Fix issue with mdns rpc and make mdns safer to upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0cea06f1ad5f9e9a6558b215b982df9facb30fd"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-08 22:45:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-13 03:47:15.000000000","tz":0},"subject":"[WIP] Move Zone XFR from mdns to worker","message":"[WIP] Move Zone XFR from mdns to worker\n\nChange-Id: I4fab6722d98dd3e188602c2808aefa7691810748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0f44418feed04436884bf2264a0ef6e95536447c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0f44418feed04436884bf2264a0ef6e95536447c"}]},"branch":"refs/heads/master","description":"Rebase"},"ecc6af51bb16d94fd00e4aab58c038c867488778":{"kind":"REWORK","_number":6,"created":"2022-05-13 04:06:41.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/841046/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/841046/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/841046/6"}}},"commit":{"parents":[{"commit":"f0cea06f1ad5f9e9a6558b215b982df9facb30fd","subject":"Fix issue with mdns rpc and make mdns safer to upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0cea06f1ad5f9e9a6558b215b982df9facb30fd"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-08 22:45:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-13 04:05:05.000000000","tz":-420},"subject":"Move Zone XFR from mdns to worker","message":"Move Zone XFR from mdns to worker\n\nThis patch moves the perform_zone_xfr call from mdns\nto the worker. This helps simplify the design of mdns,\nwith the eventual goal that mdns won\u0027t need to handle\nany rpc calls.\n\nChange-Id: I4fab6722d98dd3e188602c2808aefa7691810748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ecc6af51bb16d94fd00e4aab58c038c867488778"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ecc6af51bb16d94fd00e4aab58c038c867488778"}]},"branch":"refs/heads/master"},"495e04ff97c15ceee5e1c1e2a70e321e447e88ef":{"kind":"REWORK","_number":7,"created":"2022-05-13 04:09:45.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/841046/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/841046/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/841046/7"}}},"commit":{"parents":[{"commit":"f0cea06f1ad5f9e9a6558b215b982df9facb30fd","subject":"Fix issue with mdns rpc and make mdns safer to upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0cea06f1ad5f9e9a6558b215b982df9facb30fd"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-08 22:45:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-13 04:09:30.000000000","tz":-420},"subject":"Move Zone XFR from mdns to worker","message":"Move Zone XFR from mdns to worker\n\nThis patch moves the perform_zone_xfr call from mdns\nto the worker. This helps simplify the design of mdns,\nwith the eventual goal that mdns won\u0027t need to handle\nany rpc calls.\n\nChange-Id: I4fab6722d98dd3e188602c2808aefa7691810748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/495e04ff97c15ceee5e1c1e2a70e321e447e88ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/495e04ff97c15ceee5e1c1e2a70e321e447e88ef"}]},"branch":"refs/heads/master"},"6d7e1173eddcffb94dbd6b3a96c19501b772cdb6":{"kind":"REWORK","_number":8,"created":"2022-05-13 04:21:36.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/841046/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/841046/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/841046/8"}}},"commit":{"parents":[{"commit":"f0cea06f1ad5f9e9a6558b215b982df9facb30fd","subject":"Fix issue with mdns rpc and make mdns safer to upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0cea06f1ad5f9e9a6558b215b982df9facb30fd"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-08 22:45:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-13 04:21:26.000000000","tz":-420},"subject":"Move Zone XFR from mdns to worker","message":"Move Zone XFR from mdns to worker\n\nThis patch moves the perform_zone_xfr call from mdns\nto the worker. This helps simplify the design of mdns,\nwith the eventual goal that mdns won\u0027t need to handle\nany rpc calls.\n\nChange-Id: I4fab6722d98dd3e188602c2808aefa7691810748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6d7e1173eddcffb94dbd6b3a96c19501b772cdb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6d7e1173eddcffb94dbd6b3a96c19501b772cdb6"}]},"branch":"refs/heads/master"},"fffbde92e5cda24e422656f50444049348eb46de":{"kind":"REWORK","_number":9,"created":"2022-05-13 04:23:53.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/841046/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/841046/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/841046/9"}}},"commit":{"parents":[{"commit":"f0cea06f1ad5f9e9a6558b215b982df9facb30fd","subject":"Fix issue with mdns rpc and make mdns safer to upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0cea06f1ad5f9e9a6558b215b982df9facb30fd"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-08 22:45:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-13 04:23:42.000000000","tz":-420},"subject":"Move Zone XFR from mdns to worker","message":"Move Zone XFR from mdns to worker\n\nThis patch moves the perform_zone_xfr call from mdns\nto the worker. This helps simplify the design of mdns,\nwith the eventual goal that mdns won\u0027t need to handle\nany rpc calls.\n\nChange-Id: I4fab6722d98dd3e188602c2808aefa7691810748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fffbde92e5cda24e422656f50444049348eb46de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fffbde92e5cda24e422656f50444049348eb46de"}]},"branch":"refs/heads/master"},"d418e36a3163b8f1556b00b3eae74afc8954c5a1":{"kind":"REWORK","_number":10,"created":"2022-05-13 04:32:21.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/841046/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/841046/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/841046/10"}}},"commit":{"parents":[{"commit":"f0cea06f1ad5f9e9a6558b215b982df9facb30fd","subject":"Fix issue with mdns rpc and make mdns safer to upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0cea06f1ad5f9e9a6558b215b982df9facb30fd"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-08 22:45:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-13 04:32:13.000000000","tz":-420},"subject":"Move Zone XFR from mdns to worker","message":"Move Zone XFR from mdns to worker\n\nThis patch moves the perform_zone_xfr call from mdns\nto the worker. This helps simplify the design of mdns,\nwith the eventual goal that mdns won\u0027t need to handle\nany rpc calls.\n\nChange-Id: I4fab6722d98dd3e188602c2808aefa7691810748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d418e36a3163b8f1556b00b3eae74afc8954c5a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d418e36a3163b8f1556b00b3eae74afc8954c5a1"}]},"branch":"refs/heads/master"},"f0644e8171992a73beb6b82cc238afe9e4473131":{"kind":"REWORK","_number":11,"created":"2022-05-14 03:15:29.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/841046/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/841046/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/841046/11"}}},"commit":{"parents":[{"commit":"f0cea06f1ad5f9e9a6558b215b982df9facb30fd","subject":"Fix issue with mdns rpc and make mdns safer to upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0cea06f1ad5f9e9a6558b215b982df9facb30fd"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-08 22:45:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-14 03:15:18.000000000","tz":-420},"subject":"Move Zone XFR from mdns to worker","message":"Move Zone XFR from mdns to worker\n\nThis patch moves the perform_zone_xfr call from mdns\nto the worker. This helps simplify the design of mdns,\nwith the eventual goal that mdns won\u0027t need to handle\nany rpc calls.\n\nChange-Id: I4fab6722d98dd3e188602c2808aefa7691810748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0644e8171992a73beb6b82cc238afe9e4473131"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0644e8171992a73beb6b82cc238afe9e4473131"}]},"branch":"refs/heads/master"},"75d5739aff642aef9e70c1d85f31ceded3a7319c":{"kind":"REWORK","_number":12,"created":"2022-05-15 05:38:12.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/841046/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/841046/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/841046/12"}}},"commit":{"parents":[{"commit":"f0cea06f1ad5f9e9a6558b215b982df9facb30fd","subject":"Fix issue with mdns rpc and make mdns safer to upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0cea06f1ad5f9e9a6558b215b982df9facb30fd"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-08 22:45:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-15 05:38:03.000000000","tz":-420},"subject":"Move Zone XFR from mdns to worker","message":"Move Zone XFR from mdns to worker\n\nThis patch moves the perform_zone_xfr call from mdns\nto the worker. This helps simplify the design of mdns,\nwith the eventual goal that mdns won\u0027t need to handle\nany rpc calls.\n\nChange-Id: I4fab6722d98dd3e188602c2808aefa7691810748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75d5739aff642aef9e70c1d85f31ceded3a7319c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75d5739aff642aef9e70c1d85f31ceded3a7319c"}]},"branch":"refs/heads/master"},"1a2f4565ce5706f18c806125b9cab11bef78628b":{"kind":"REWORK","_number":13,"created":"2022-05-15 05:49:37.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/841046/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/841046/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/841046/13"}}},"commit":{"parents":[{"commit":"f0cea06f1ad5f9e9a6558b215b982df9facb30fd","subject":"Fix issue with mdns rpc and make mdns safer to upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0cea06f1ad5f9e9a6558b215b982df9facb30fd"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-08 22:45:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-15 05:49:28.000000000","tz":-420},"subject":"Move Zone XFR from mdns to worker","message":"Move Zone XFR from mdns to worker\n\nThis patch moves the perform_zone_xfr call from mdns\nto the worker. This helps simplify the design of mdns,\nwith the eventual goal that mdns won\u0027t need to handle\nany rpc calls.\n\nChange-Id: I4fab6722d98dd3e188602c2808aefa7691810748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1a2f4565ce5706f18c806125b9cab11bef78628b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1a2f4565ce5706f18c806125b9cab11bef78628b"}]},"branch":"refs/heads/master"},"4c32778a1bfdaea8f199e48c0a9f4035f9b46760":{"kind":"REWORK","_number":14,"created":"2022-05-15 06:57:29.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/841046/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/841046/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/841046/14"}}},"commit":{"parents":[{"commit":"f0cea06f1ad5f9e9a6558b215b982df9facb30fd","subject":"Fix issue with mdns rpc and make mdns safer to upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0cea06f1ad5f9e9a6558b215b982df9facb30fd"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-08 22:45:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-15 06:57:17.000000000","tz":-420},"subject":"Move Zone XFR from mdns to worker","message":"Move Zone XFR from mdns to worker\n\nThis patch moves the perform_zone_xfr call from mdns\nto the worker. This helps simplify the design of mdns,\nwith the eventual goal that mdns won\u0027t need to handle\nany rpc calls.\n\nChange-Id: I4fab6722d98dd3e188602c2808aefa7691810748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4c32778a1bfdaea8f199e48c0a9f4035f9b46760"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4c32778a1bfdaea8f199e48c0a9f4035f9b46760"}]},"branch":"refs/heads/master"},"2ee087dac531002302c9217c67be9f6a50eafc4d":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2022-05-18 05:19:29.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/841046/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/841046/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/841046/15"}}},"commit":{"parents":[{"commit":"9a3b5556e7e0b742e8227dbe4415b9ff84078fb7","subject":"Fix issue with mdns rpc and make mdns safer to upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a3b5556e7e0b742e8227dbe4415b9ff84078fb7"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-08 22:45:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-18 05:19:29.000000000","tz":0},"subject":"Move Zone XFR from mdns to worker","message":"Move Zone XFR from mdns to worker\n\nThis patch moves the perform_zone_xfr call from mdns\nto the worker. This helps simplify the design of mdns,\nwith the eventual goal that mdns won\u0027t need to handle\nany rpc calls.\n\nChange-Id: I4fab6722d98dd3e188602c2808aefa7691810748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2ee087dac531002302c9217c67be9f6a50eafc4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2ee087dac531002302c9217c67be9f6a50eafc4d"}]},"branch":"refs/heads/master","description":"Rebase"},"cef332443fd776e55a7c4eafbb4df1f756b0b12c":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2022-05-25 03:02:09.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/841046/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/841046/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/841046/16"}}},"commit":{"parents":[{"commit":"56223bbd4fafaf62bc31aa0159219be49de6997f","subject":"Fix issue with mdns rpc and make mdns safer to upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/56223bbd4fafaf62bc31aa0159219be49de6997f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-08 22:45:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-25 03:02:09.000000000","tz":0},"subject":"Move Zone XFR from mdns to worker","message":"Move Zone XFR from mdns to worker\n\nThis patch moves the perform_zone_xfr call from mdns\nto the worker. This helps simplify the design of mdns,\nwith the eventual goal that mdns won\u0027t need to handle\nany rpc calls.\n\nChange-Id: I4fab6722d98dd3e188602c2808aefa7691810748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cef332443fd776e55a7c4eafbb4df1f756b0b12c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cef332443fd776e55a7c4eafbb4df1f756b0b12c"}]},"branch":"refs/heads/master","description":"Rebase"},"9a4782b3691f446c03258a474e57b3d288c02620":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2022-05-27 17:40:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/841046/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/841046/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/841046/17"}}},"commit":{"parents":[{"commit":"7857fada3b9340cb42f4508262f2b008c968f1d1","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7857fada3b9340cb42f4508262f2b008c968f1d1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-08 22:45:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-27 17:40:07.000000000","tz":0},"subject":"Move Zone XFR from mdns to worker","message":"Move Zone XFR from mdns to worker\n\nThis patch moves the perform_zone_xfr call from mdns\nto the worker. This helps simplify the design of mdns,\nwith the eventual goal that mdns won\u0027t need to handle\nany rpc calls.\n\nChange-Id: I4fab6722d98dd3e188602c2808aefa7691810748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a4782b3691f446c03258a474e57b3d288c02620"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a4782b3691f446c03258a474e57b3d288c02620"}]},"branch":"refs/heads/master","description":"Rebase"},"99f55f04654c1f8ba7859af304962a86ecc5e8cf":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2022-06-12 21:12:21.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/841046/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/841046/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/841046/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/841046/18"}}},"commit":{"parents":[{"commit":"e1f7b4d6e327f71bab4a991efdc534e60f01d9a0","subject":"Merge \"Check zone_id in notification_handler to avoid UUID format issue.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e1f7b4d6e327f71bab4a991efdc534e60f01d9a0"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-08 22:45:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-06-12 21:12:21.000000000","tz":0},"subject":"Move Zone XFR from mdns to worker","message":"Move Zone XFR from mdns to worker\n\nThis patch moves the perform_zone_xfr call from mdns\nto the worker. This helps simplify the design of mdns,\nwith the eventual goal that mdns won\u0027t need to handle\nany rpc calls.\n\nChange-Id: I4fab6722d98dd3e188602c2808aefa7691810748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/99f55f04654c1f8ba7859af304962a86ecc5e8cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/99f55f04654c1f8ba7859af304962a86ecc5e8cf"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
