)]}'
{"id":"openstack%2Fdesignate~841168","triplet_id":"openstack%2Fdesignate~master~I57f3d5a1d1f79186cc5b38e76d30f62e01b60482","project":"openstack/designate","branch":"master","topic":"bug/1971856","attention_set":{},"removed_from_attention_set":{"34579":{"account":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"last_update":"2022-05-18 04:31:06.000000000","reason":"Change was submitted"},"5572":{"account":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"last_update":"2022-05-18 04:31:06.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-05-17 21:47:26.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"I57f3d5a1d1f79186cc5b38e76d30f62e01b60482","subject":"Adds api_ca_cert configuration variable to pools.yaml","status":"MERGED","created":"2022-05-09 20:55:07.000000000","updated":"2022-05-18 04:32:33.000000000","submitted":"2022-05-18 04:31:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841168-bug/1971856","meta_rev_id":"f2aad4500e115bc7ceb0f51455c93f952eeb6781","_number":841168,"virtual_id_number":841168,"owner":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-18 04:31:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"all":[{"value":1,"date":"2022-05-13 15:23:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2022-05-17 21:47:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-18 03:24:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-18 03:24:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-09 22:23:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-10 00:08:20.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-05-10 15:27:13.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2022-05-18 03:24:29.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"2e07e6790889329925dc13a3d243edaf14409c61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"date":"2022-05-09 20:55:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6e6e7ddd79012ba59618e0f25de0a05430a4c5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"date":"2022-05-09 21:02:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ec5a2ecb48127b89e458a9f57f01f0a84905375e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 22:23:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fb6556e9affd4739be244025c91bdfb1 : SUCCESS in 39m 06s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/89c11a1af0644ea99e6d91bac3d00123 : SUCCESS in 42m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1c233af189342e5b89e65f7b0b99885 : SUCCESS in 39m 56s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/9493acf0b81a435c824434c53dcf8140 : SUCCESS in 43m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c52a1ecfb29d44318e9626ba324209d1 : SUCCESS in 47m 11s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/afba957fb57849799de27403ac38334a : SUCCESS in 56m 59s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/de74fc100c1a4334801e78e598a42f3a : FAILURE in 1h 14m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8a8e4a9b16b64fd2acea34f4c29da62e : SUCCESS in 37m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2199f4082a1143af83e6d235d6ce299b : SUCCESS in 39m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0933865565de4ccda767a25c917d7e8e : FAILURE in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/353bea3bdb6b4464b698967b4dbdc5fa : FAILURE in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ee3a61480754569bdcad27ea729969a : FAILURE in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/780e5eb1f6784258a6f25d818256b47c : FAILURE in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66e9a420d72c4de58b1fb2154829f8c2 : FAILURE in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d6af117472b4786b87d4322bdb4db39 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/caa3d486b2ae4601a0c27efcf6c5de75 : SUCCESS in 4m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/62f6d762887e459aa75256c787e95670 : SUCCESS in 29m 40s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1cf80fe33eb54524903ac1737894951b : FAILURE in 4m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1eae1b077ddc3812a87ecff34b0acc712541c093","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-10 00:08:20.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"822d651bc815ea255b2ded440c0389733412db5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"date":"2022-05-10 13:50:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9c311f5e31bd5c234f8af5c0467e8cb1e11b7d68","author":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"date":"2022-05-10 13:58:42.000000000","message":"Patch Set 3:\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"439b58311bfda8d3a4bbf97e43bd111f29827b8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 14:58:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0a8600472ba042f6bb240bd5a05c9f69 : SUCCESS in 36m 26s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/41eea82637294704a9ec9cbbc3d9b6a5 : SUCCESS in 49m 39s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9cde95e30d64d5e9a6fb69f40bf95c8 : SUCCESS in 38m 35s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/d266139e6fe240fc8c5f089cb1a0042f : SUCCESS in 34m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6e1f824dd3f7499ab17cc961783184a5 : SUCCESS in 36m 41s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/23f21cd101e34e3b9043cd109ebd04b7 : SUCCESS in 1h 02m 32s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3d89e84f86ff4e6fac4e1177758e4057 : FAILURE in 56m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/15bce24ef71f473a9a5fddc3bbfd00b9 : SUCCESS in 32m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d51c060e4c394210a7fa0222ba64b7d3 : SUCCESS in 47m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27a7c82101824aa893509028c9974478 : FAILURE in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c61b431bcc3748229ea5c0fba64bb3fe : FAILURE in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc23909a070a4c22b1bd53b7042e84fb : FAILURE in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e157114b1f374737acb53edfa1ba0ce5 : FAILURE in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c13bcd55e7c40a39c755affad2711b7 : FAILURE in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b61c663757be468d86abbec880056c0a : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dda3594e563f4222a5e2282d66012ec3 : SUCCESS in 3m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a7f5a06ce810465ea566bcc44570d1a7 : SUCCESS in 32m 31s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/bde78066a7c14641a6b0d9fa06dd4a66 : FAILURE in 5m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"177c985a66cf0c29f8167a1d69ea35c122013f11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-10 15:21:49.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"7f1da99eb9dfdd6ba7825179d9ff88a8b71c1217","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-10 15:23:14.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"dc97dc001db56b7127d9e22afc64f289527e6eff","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-05-10 15:27:13.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"141ab0423462ebff6e78d59b93be0b869a5c6ce9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"date":"2022-05-10 16:22:57.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"b6569aa7dce5012845b3d2692cc5df84f43bcc3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 17:21:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6e16ae6101eb4c358dd77df886ff86a6 : SUCCESS in 38m 51s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b426143b79f844f1affa3f79a7228f6d : SUCCESS in 48m 05s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7984ceed98d4889951703393ce0177c : SUCCESS in 40m 22s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3879885fc97a4522aad3a2f9ea3b6823 : SUCCESS in 37m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/223c2318568f49b39b0c7c353d7530c4 : SUCCESS in 37m 34s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8dd7b4d1977f4e6a8edf72c383c26091 : SUCCESS in 56m 47s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fe0b3e27a0e241a0bb1258b05ff465b0 : SUCCESS in 49m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/61dbc049327a4433adac2b11c76eeeda : SUCCESS in 35m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4d25f07339e84e04a7b292e21f4da44b : SUCCESS in 38m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c90365a5bd5d4f46bdef48150c80d051 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53309ad2ba2d4afabd65126066d0f7ea : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b350cc0f1c5a4ccfadd18c7eece3c086 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e2bd88663324ce2ba2cfc6e2fe97513 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d82e23e983504dd09e1f1e2f916fa063 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02ecc610851748119f7950de33d9162f : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a40756344564c6aac93a470bc750255 : SUCCESS in 4m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cfc468b06a6846a6b8ce634ec67e22b5 : SUCCESS in 27m 46s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0654325cdbe74f7d930c6be375b76b0e : SUCCESS in 4m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"260d135c3a8b6ac53d98bbe9cdea93f3198495b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"date":"2022-05-10 17:25:34.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"9c0771b20679a43e66b157ccf4bd2c7c87d40014","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 18:22:38.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f6942e571ede4792939aa2e914abe3c9 : SUCCESS in 44m 35s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f1d1c794a31046e0a8a0a41ec79f0cb2 : FAILURE in 22m 56s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/358c7dc840634400b30ce3c6ccfa5e35 : SUCCESS in 55m 25s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/38b44f104ffb49ba9ce2e3c97bc3b385 : SUCCESS in 44m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/530336adcf7742db9990a3e798f82746 : SUCCESS in 40m 47s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3eac0b4072894e80a91c82562f8bc3ee : SUCCESS in 54m 52s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/246c77f4e27c4ee6932471ec3c8d5f62 : SUCCESS in 53m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1b5bb9b3a2794e3390d4a40bd5ed595f : SUCCESS in 38m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f987a6b25e2d4b3cb19275f575feed73 : SUCCESS in 34m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b89a1bba241b4538a7dd2d532ed00e78 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10848574dbdf46b0998fc08f686da694 : FAILURE in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32502ff61a6346a9828af5666f7229cd : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33db7a9bf6f14f3aa79b046e22772c5d : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80c114e3e73545ef9c082a98a326d0ee : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6c64c5228cd4b699c696423b2819ada : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b84e527d8256473a959a9da38311fdf6 : SUCCESS in 4m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/431169a0895d4fdd9095edd58e8d31ad : SUCCESS in 37m 53s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/956cdd6c2dc24456822ba43936cea795 : SUCCESS in 4m 49s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5a875527d1c958d808457a5f498e56a75a6d8878","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"date":"2022-05-10 18:26:30.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c92013d08e0807f909d6cc0d4af85f84a3a5e987","author":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"date":"2022-05-10 18:28:57.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"1642554b741d015e04f4675a49f6d7f175c3c630","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 19:25:50.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c86de5516e2f4d56baebbcf481e6f0c8 : SUCCESS in 49m 23s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f7bf6669f5a84257b80bdf1da9ea4831 : FAILURE in 24m 21s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec52c893932a41acb5644952a56454ea : SUCCESS in 48m 06s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/da75e069e2f54f0fb919573bc74f5e64 : SUCCESS in 46m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/00dd76ecae45420eadebcb6471e7a979 : SUCCESS in 46m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6c02602a0bc94c9a9d24dd2666b26ee3 : FAILURE in 49m 04s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a3089b3eac56465195a676201e4afe77 : FAILURE in 52m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6e2901c3a4cb4ecf9926cb1014d46493 : SUCCESS in 52m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b5ebca351e814934ad5c102bfb7c119e : SUCCESS in 44m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa4c8b6787084e95a6678bbe208c4194 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24d186226ebb4665baef0a67f1361385 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1520d6651eb44196880204e8bdce698f : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0ea5dac21384780939058489a81c46d : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/838ed59fec274e249d2c2331312c6e3f : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98a37a0685ba44b8951a32f505b9b024 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6189eaeb5c0456390a58d7a6ccbf18b : SUCCESS in 3m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b086afc12121485fbd1979b928e85937 : SUCCESS in 30m 57s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/da745ccc66c146c2a6df79676951f5ef : SUCCESS in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0743ef97323ef95f325332d90dc2fcd31582ef17","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-10 22:19:21.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b102bde1af946d56fc0222a2c852c2b7b16ab749","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 23:36:49.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f3f23769b5034307bf675ff41b0b6afc : SUCCESS in 32m 48s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4293bd9af704494388894f1069c48579 : FAILURE in 20m 56s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d5cbab39df04710b8ab01dfc578a650 : SUCCESS in 52m 21s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4e28de70da8f4f92b2958a8f41e53477 : SUCCESS in 49m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/15b449b4557941268de159f3db88b5ba : SUCCESS in 45m 10s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8c386e655420489d8accb17bc1eafa97 : SUCCESS in 1h 11m 01s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/be0e1892b418410b917646e22fa755c0 : SUCCESS in 1h 08m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/47f91be9780e4f1f94fe51f5113ff52c : SUCCESS in 33m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e6a53eb867224b5e8b7bb9ee89b9084b : SUCCESS in 44m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22dd6ea588464f148f6e94a7f0107b32 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e4ec636f7274d36abf07855a1a84e7e : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/682fe26b30fa44aeb9d95a88a0f82ab6 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24f7dfe43325432e93ddf735e783874c : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efba5d21bc6341898a7cebe7530c85e5 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/485dfe998bfd49b1983b2285f78efc86 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2313e4de1a15491da7bd4eef55c92683 : SUCCESS in 3m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d1153e67e7a0423d9aacc31b099fcece : SUCCESS in 27m 39s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a9e5e6e9278447118b40e71e3d7f0c21 : SUCCESS in 4m 16s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"aabb2767caf84782ec9647c5b54936133c9d7efd","author":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"date":"2022-05-13 14:18:33.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"4de58b63a228352a848fbd45a852567e1d7a2925","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-05-13 15:02:33.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b933597ab544b014ff22fbc8263af24dc264e5c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"date":"2022-05-13 15:21:57.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d3cc97f55866148401b8ea7aa5690e0db87de839","author":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"date":"2022-05-13 15:23:07.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"39283800eebcdc4ba978b5ee1849e72d9c5864ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 16:24:05.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/67e0b8e5dc0f4558baab3c94451c2e65 : SUCCESS in 37m 17s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e3c78eb6b080456e81a5f920937a2356 : FAILURE in 13m 00s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/dca5c23ae62d47fd89d03c6809a3637c : SUCCESS in 41m 15s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/18da93edccc3494c8c6f3867aa32e7b6 : SUCCESS in 41m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/305cb368000c4a6a9db52e24f3079dd4 : SUCCESS in 38m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/667b7e14428f47149f5b11ee926373fc : SUCCESS in 1h 00m 37s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3cea803e370d426d96743a8e87304718 : SUCCESS in 58m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bc4cbac8348e4df8867aba35c3a676db : SUCCESS in 35m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b99773cce4ba43f3ab8f3bf819488dfc : SUCCESS in 38m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8d8048751204e089e79d291d05b3f22 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e271bc172fc94ea1b9807295f119555f : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db398510bd50461b82a361f34816aaa9 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c33ac29ff05b490da730327fa8d6970a : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5682793235ee4b29b64654e37c5aee5c : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30f4ac9b396f40cb91f9ff7a27228513 : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5f2ccdec7404d57b0380e9b1577286c : SUCCESS in 3m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5f91438669694ca8b4bc878979565bd5 : SUCCESS in 33m 30s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b9b46ff49ac2450cb0f03fefd56b1f1b : SUCCESS in 5m 51s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7dad52104fa5a3267fefd4a8b61f06987c486028","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-17 21:47:26.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"fa111acd68708c2b55d2c0c921a36d0b6e89c9e4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-18 03:24:29.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"f93ba5fa05c9c6bd7425d6728a8efab43e59f2d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 03:25:10.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"1dee734ec028ecf15a1dfe3b644b1b39eb9e631e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 04:31:06.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1154307e015e4b66b58edcd206560bbe : SUCCESS in 51m 19s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/486a2fae9ad444729483d5484aa61e7d : SUCCESS in 47m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a690ae81adf44c4994e86a2afb17a8a1 : SUCCESS in 45m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ff0790fc989145da868f11df900840a1 : SUCCESS in 1h 02m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c57c34894e604f70bbf755d553f3897e : SUCCESS in 44m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/79c64914614c4168842aecc73edc9aed : SUCCESS in 56m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/314db32e9b9e40b5a0c5b3214ac8b8bb : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/093585c088ad49fe89a32bff40d54ae6 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c395046f18e4d00ae3b7a0ad2c50a60 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb03826515d94abe9cdce6cde18a4f33 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f7410ad812f474ca89fc12fe9e618f5 : SUCCESS in 4m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/635c4d645d104187b0af5876002aee7d : SUCCESS in 37m 34s","accounts_in_message":[],"_revision_number":9},{"id":"c8593080755c30fb2f9ec4fd66ba901d4349fad8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 04:31:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"f2aad4500e115bc7ceb0f51455c93f952eeb6781","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 04:32:33.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a0732d769ed41d7830837a245695558 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8069cbc4a55e4b91988d576a1a979126 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"c230ddb8c5f71b51f2fa492d6ac076abbecc3841","revisions":{"b6bba49c979080f3483da751fdf1edfd113c55d8":{"kind":"REWORK","_number":1,"created":"2022-05-09 20:55:07.000000000","uploader":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"ref":"refs/changes/68/841168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/841168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/841168/1"}}},"commit":{"parents":[{"commit":"4b516eaccc782c328f7e7bb9d65a7511515e44ef","subject":"Merge \"Enable ptr unset tempest test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b516eaccc782c328f7e7bb9d65a7511515e44ef"}]}],"author":{"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","date":"2022-05-09 20:52:38.000000000","tz":-240},"committer":{"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","date":"2022-05-09 20:52:38.000000000","tz":-240},"subject":"Adds api_ca_cert configuration variable to pools.yaml","message":"Adds api_ca_cert configuration variable to pools.yaml\n\nAdss a configuration variable that allows a user to\ndeclare the CA certificate to be used to verify\ntraffic with a PowerDNS API endpoint.\n\nCloses-Bug: #1971856\nSigned-off-by: Juan Pablo Suazo \u003cjsuazo@whitestack.com\u003e\nChange-Id: I57f3d5a1d1f79186cc5b38e76d30f62e01b60482\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b6bba49c979080f3483da751fdf1edfd113c55d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b6bba49c979080f3483da751fdf1edfd113c55d8"}]},"branch":"refs/heads/master"},"0b202149fdb1f5841d04b8a8e46dbc8db292e798":{"kind":"REWORK","_number":2,"created":"2022-05-09 21:02:27.000000000","uploader":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"ref":"refs/changes/68/841168/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/841168/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/841168/2"}}},"commit":{"parents":[{"commit":"4b516eaccc782c328f7e7bb9d65a7511515e44ef","subject":"Merge \"Enable ptr unset tempest test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b516eaccc782c328f7e7bb9d65a7511515e44ef"}]}],"author":{"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","date":"2022-05-09 20:52:38.000000000","tz":-240},"committer":{"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","date":"2022-05-09 21:02:15.000000000","tz":-240},"subject":"Adds api_ca_cert configuration variable to pools.yaml","message":"Adds api_ca_cert configuration variable to pools.yaml\n\nAdss a configuration variable that allows a user to\ndeclare the CA certificate to be used to verify\ntraffic with a PowerDNS API endpoint.\n\nCloses-Bug: #1971856\nSigned-off-by: Juan Pablo Suazo \u003cjsuazo@whitestack.com\u003e\nChange-Id: I57f3d5a1d1f79186cc5b38e76d30f62e01b60482\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0b202149fdb1f5841d04b8a8e46dbc8db292e798"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0b202149fdb1f5841d04b8a8e46dbc8db292e798"}]},"branch":"refs/heads/master"},"9cf3e597f3346f9ae918879a1f38625852467d60":{"kind":"REWORK","_number":3,"created":"2022-05-10 13:50:43.000000000","uploader":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"ref":"refs/changes/68/841168/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/841168/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/841168/3"}}},"commit":{"parents":[{"commit":"4b516eaccc782c328f7e7bb9d65a7511515e44ef","subject":"Merge \"Enable ptr unset tempest test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b516eaccc782c328f7e7bb9d65a7511515e44ef"}]}],"author":{"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","date":"2022-05-09 20:52:38.000000000","tz":-240},"committer":{"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","date":"2022-05-10 13:50:31.000000000","tz":-240},"subject":"Adds api_ca_cert configuration variable to pools.yaml","message":"Adds api_ca_cert configuration variable to pools.yaml\n\nAdss a configuration variable that allows a user to\ndeclare the CA certificate to be used to verify\ntraffic with a PowerDNS API endpoint.\n\nCloses-Bug: #1971856\nSigned-off-by: Juan Pablo Suazo \u003cjsuazo@whitestack.com\u003e\nChange-Id: I57f3d5a1d1f79186cc5b38e76d30f62e01b60482\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9cf3e597f3346f9ae918879a1f38625852467d60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9cf3e597f3346f9ae918879a1f38625852467d60"}]},"branch":"refs/heads/master"},"873a02d14d8d2198df8e0566248d9da0ea6f439c":{"kind":"REWORK","_number":4,"created":"2022-05-10 15:21:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/841168/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/841168/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/841168/4"}}},"commit":{"parents":[{"commit":"4b516eaccc782c328f7e7bb9d65a7511515e44ef","subject":"Merge \"Enable ptr unset tempest test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b516eaccc782c328f7e7bb9d65a7511515e44ef"}]}],"author":{"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","date":"2022-05-09 20:52:38.000000000","tz":-240},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-10 15:21:49.000000000","tz":0},"subject":"Adds api_ca_cert configuration variable to pools.yaml","message":"Adds api_ca_cert configuration variable to pools.yaml\n\nAdss a configuration variable that allows a user to\ndeclare the CA certificate to be used to verify\ntraffic with a PowerDNS API endpoint.\n\nCloses-Bug: #1971856\nSigned-off-by: Juan Pablo Suazo \u003cjsuazo@whitestack.com\u003e\nChange-Id: I57f3d5a1d1f79186cc5b38e76d30f62e01b60482\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/873a02d14d8d2198df8e0566248d9da0ea6f439c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/873a02d14d8d2198df8e0566248d9da0ea6f439c"}]},"branch":"refs/heads/master"},"b5a883a4025bd2ca32d8ddf318d04bb97517dcd2":{"kind":"REWORK","_number":5,"created":"2022-05-10 15:23:14.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/841168/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/841168/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/841168/5"}}},"commit":{"parents":[{"commit":"4b516eaccc782c328f7e7bb9d65a7511515e44ef","subject":"Merge \"Enable ptr unset tempest test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b516eaccc782c328f7e7bb9d65a7511515e44ef"}]}],"author":{"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","date":"2022-05-09 20:52:38.000000000","tz":-240},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-10 15:23:14.000000000","tz":0},"subject":"Adds api_ca_cert configuration variable to pools.yaml","message":"Adds api_ca_cert configuration variable to pools.yaml\n\nAdss a configuration variable that allows a user to\ndeclare the CA certificate to be used to verify\ntraffic with a PowerDNS API endpoint.\n\nCloses-Bug: #1971856\nSigned-off-by: Juan Pablo Suazo \u003cjsuazo@whitestack.com\u003e\nChange-Id: I57f3d5a1d1f79186cc5b38e76d30f62e01b60482\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b5a883a4025bd2ca32d8ddf318d04bb97517dcd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b5a883a4025bd2ca32d8ddf318d04bb97517dcd2"}]},"branch":"refs/heads/master"},"9bb146e36c4ab29659b3f8128004bac7eebbc3bb":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-05-10 16:22:57.000000000","uploader":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"ref":"refs/changes/68/841168/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/841168/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/841168/6"}}},"commit":{"parents":[{"commit":"4b516eaccc782c328f7e7bb9d65a7511515e44ef","subject":"Merge \"Enable ptr unset tempest test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b516eaccc782c328f7e7bb9d65a7511515e44ef"}]}],"author":{"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","date":"2022-05-09 20:52:38.000000000","tz":-240},"committer":{"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","date":"2022-05-10 16:22:57.000000000","tz":0},"subject":"Adds api_ca_cert configuration variable to pools.yaml","message":"Adds api_ca_cert configuration variable to pools.yaml\n\nAdds a configuration variable that allows a user to\ndeclare the CA certificate to be used to verify\ntraffic with a PowerDNS API endpoint.\n\nCloses-Bug: #1971856\nSigned-off-by: Juan Pablo Suazo \u003cjsuazo@whitestack.com\u003e\nChange-Id: I57f3d5a1d1f79186cc5b38e76d30f62e01b60482\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9bb146e36c4ab29659b3f8128004bac7eebbc3bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9bb146e36c4ab29659b3f8128004bac7eebbc3bb"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c3074c092c82d4b4742df46c7b9538c241ed4968":{"kind":"REWORK","_number":7,"created":"2022-05-10 17:25:34.000000000","uploader":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"ref":"refs/changes/68/841168/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/841168/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/841168/7"}}},"commit":{"parents":[{"commit":"4b516eaccc782c328f7e7bb9d65a7511515e44ef","subject":"Merge \"Enable ptr unset tempest test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b516eaccc782c328f7e7bb9d65a7511515e44ef"}]}],"author":{"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","date":"2022-05-09 20:52:38.000000000","tz":-240},"committer":{"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","date":"2022-05-10 17:24:52.000000000","tz":-240},"subject":"Adds api_ca_cert configuration variable to pools.yaml","message":"Adds api_ca_cert configuration variable to pools.yaml\n\nAdds a configuration variable that allows a user to\ndeclare the CA certificate to be used to verify\ntraffic with a PowerDNS API endpoint.\n\nCloses-Bug: #1971856\nSigned-off-by: Juan Pablo Suazo \u003cjsuazo@whitestack.com\u003e\nChange-Id: I57f3d5a1d1f79186cc5b38e76d30f62e01b60482\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c3074c092c82d4b4742df46c7b9538c241ed4968"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c3074c092c82d4b4742df46c7b9538c241ed4968"}]},"branch":"refs/heads/master"},"eed98c93afde55728410bd9279238a9241ead313":{"kind":"REWORK","_number":8,"created":"2022-05-10 18:26:30.000000000","uploader":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"ref":"refs/changes/68/841168/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/841168/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/841168/8"}}},"commit":{"parents":[{"commit":"4b516eaccc782c328f7e7bb9d65a7511515e44ef","subject":"Merge \"Enable ptr unset tempest test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b516eaccc782c328f7e7bb9d65a7511515e44ef"}]}],"author":{"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","date":"2022-05-09 20:52:38.000000000","tz":-240},"committer":{"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","date":"2022-05-10 18:26:20.000000000","tz":-240},"subject":"Adds api_ca_cert configuration variable to pools.yaml","message":"Adds api_ca_cert configuration variable to pools.yaml\n\nAdds a configuration variable that allows a user to\ndeclare the CA certificate to be used to verify\ntraffic with a PowerDNS API endpoint.\n\nCloses-Bug: #1971856\nSigned-off-by: Juan Pablo Suazo \u003cjsuazo@whitestack.com\u003e\nChange-Id: I57f3d5a1d1f79186cc5b38e76d30f62e01b60482\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eed98c93afde55728410bd9279238a9241ead313"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eed98c93afde55728410bd9279238a9241ead313"}]},"branch":"refs/heads/master"},"c230ddb8c5f71b51f2fa492d6ac076abbecc3841":{"kind":"REWORK","_number":9,"created":"2022-05-13 15:21:57.000000000","uploader":{"_account_id":34579,"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","username":"jsuazo"},"ref":"refs/changes/68/841168/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/841168/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/841168/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/841168/9"}}},"commit":{"parents":[{"commit":"4b516eaccc782c328f7e7bb9d65a7511515e44ef","subject":"Merge \"Enable ptr unset tempest test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b516eaccc782c328f7e7bb9d65a7511515e44ef"}]}],"author":{"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","date":"2022-05-09 20:52:38.000000000","tz":-240},"committer":{"name":"Juan Pablo Suazo","email":"jsuazo@whitestack.com","date":"2022-05-13 15:21:47.000000000","tz":-240},"subject":"Adds api_ca_cert configuration variable to pools.yaml","message":"Adds api_ca_cert configuration variable to pools.yaml\n\nAdds a configuration variable that allows a user to\ndeclare the CA certificate to be used to verify\ntraffic with a PowerDNS API endpoint.\n\nCloses-Bug: #1971856\nSigned-off-by: Juan Pablo Suazo \u003cjsuazo@whitestack.com\u003e\nChange-Id: I57f3d5a1d1f79186cc5b38e76d30f62e01b60482\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c230ddb8c5f71b51f2fa492d6ac076abbecc3841"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c230ddb8c5f71b51f2fa492d6ac076abbecc3841"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
