)]}'
{"id":"openstack%2Fdesignate~841462","triplet_id":"openstack%2Fdesignate~master~Ied091e7acd0cb2c837f8ee7335b7881225da0126","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-05-11 23:31:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ied091e7acd0cb2c837f8ee7335b7881225da0126","subject":"Fix BIND9 log collection","status":"MERGED","created":"2022-05-11 17:56:14.000000000","updated":"2022-05-11 23:32:54.000000000","submitted":"2022-05-11 23:31:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841462","meta_rev_id":"2950ec5003c224db1a9f71b16c0694ed18ef2ae8","_number":841462,"virtual_id_number":841462,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-11 23:31:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-05-11 21:24:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-11 22:44:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-11 22:44:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-11 18:36:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-11 22:44:52.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"3fae9b4046a2b3d1f4afa0fe3769e80b843115e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-11 17:56:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f182b6ae744941c1245bd47532d27915879b0f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 18:36:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/463bf67f3aac4d77b1dfa4c80de55df6 : SUCCESS in 37m 52s","accounts_in_message":[],"_revision_number":1},{"id":"f4ef09c4a67eb456400f6eff76c8964f2d8c797d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-11 20:25:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f351e6519e524e4adafb8b3960038c176c7c3428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 21:22:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/370881d303de41f9b1ad4bed0d2fb9ae : SUCCESS in 39m 48s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/852ad011e24d4cf0b4bf95320decf9bb : SUCCESS in 43m 20s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab204b69821c4519ad3fe2b5356f2c77 : SUCCESS in 44m 15s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1bd05a38a181454b902235e81178b4eb : SUCCESS in 41m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/54297cca9e4b44129108619e15187d6c : SUCCESS in 35m 14s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9c32e5330d51471da5873cbe3a9b0d73 : SUCCESS in 53m 37s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e9f7004c191e421a989c6635b924f389 : SUCCESS in 53m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/231a224dd57b4d3aaa574b36d0a936ed : SUCCESS in 37m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1fcb5ba9e10944f283ee1c545088d5a4 : SUCCESS in 37m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f9e5a3a3a834b5aa9405667b0434064 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dabf6cdc9eb544fd84e232a104555cf7 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18ca646120754e96969c4cf344a4a8ce : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/218a14092ac3443f952624797dbff966 : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fbbac5b79df9410aa1172754260008a5 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e4616a97c0d4d78b77b88078a6f0539 : SUCCESS in 7m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3ae7767c8bb94a9d9850516e76a3639e : SUCCESS in 24m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e77b9cf94e9649ba9d2db281e3d2fc80 : SUCCESS in 5m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5473ec2111d2996782ea455536a34f9c1c323148","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-11 21:24:52.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"42dc6321be274df0b82b1db6b27e85cfbcb850a2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-11 22:44:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"15e86f8d42954e29dcb8afb5f0bcee96af43ccdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 22:45:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a3812eccbb4ab719c0abd707f0a044c0b09fab31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 23:31:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6dae7878f7464ebb9063e9142f20acbb : SUCCESS in 35m 19s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/05427047c28742a8b3ff9929324802f8 : SUCCESS in 32m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ff5b2cecaa924f32ad6c6cd88df7d8cf : SUCCESS in 30m 40s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b0bc8129db2c43bfb7d93db6acb8f7ec : SUCCESS in 43m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b5c095a7eea745d6b590e600fe14b115 : SUCCESS in 29m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/88706b2296024713b6c6d465de5d0cc4 : SUCCESS in 32m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3d3d5f8c2f14da4856352272193a370 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d193013034234ba2b7a18d60732a9851 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21fbc37ea4ff4da58dbd2ac8bff315f3 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8a967d2a96e4d91893b9c5848cba95e : SUCCESS in 6m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1ae294639ed041d397ad8d6065593f11 : SUCCESS in 29m 11s","accounts_in_message":[],"_revision_number":2},{"id":"bd599d3878bb6dfe55b96ecf58a3ff1d849893de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 23:31:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2950ec5003c224db1a9f71b16c0694ed18ef2ae8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 23:32:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d1ab4c55fa04932901f270156b14cc5 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1fd240f81b6850cb6db5eea91f7a91decf962757","revisions":{"578e9b454faca65bc982e1e50e63d58666f9791a":{"kind":"REWORK","_number":1,"created":"2022-05-11 17:56:14.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/62/841462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/62/841462/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/841462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/841462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/841462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/62/841462/1"}}},"commit":{"parents":[{"commit":"4b516eaccc782c328f7e7bb9d65a7511515e44ef","subject":"Merge \"Enable ptr unset tempest test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b516eaccc782c328f7e7bb9d65a7511515e44ef"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-11 17:54:40.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-11 17:56:13.000000000","tz":0},"subject":"DNM: Testing BIND9 log archive fix","message":"DNM: Testing BIND9 log archive fix\n\nCurrently the bind logs are not viewable in the zuul logs page, they prompt\nfor download. I think they are getting double gzip\u0027d by the designate role and the zuul-jobs role.\n\nChange-Id: Ied091e7acd0cb2c837f8ee7335b7881225da0126\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/578e9b454faca65bc982e1e50e63d58666f9791a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/578e9b454faca65bc982e1e50e63d58666f9791a"}]},"branch":"refs/heads/master"},"1fd240f81b6850cb6db5eea91f7a91decf962757":{"kind":"REWORK","_number":2,"created":"2022-05-11 20:25:05.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/62/841462/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/62/841462/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/841462/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/841462/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/841462/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/62/841462/2"}}},"commit":{"parents":[{"commit":"4b516eaccc782c328f7e7bb9d65a7511515e44ef","subject":"Merge \"Enable ptr unset tempest test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b516eaccc782c328f7e7bb9d65a7511515e44ef"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-11 17:54:40.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-11 20:23:15.000000000","tz":0},"subject":"Fix BIND9 log collection","message":"Fix BIND9 log collection\n\nCurrently the bind logs are not viewable in the zuul logs page, they prompt\nfor download. The wrong systemd unit logs are being exported and they were being double gziped. This patch corrects both of those issues such that the BIND9 logs will be properly collected in the zuul jobs.\n\nChange-Id: Ied091e7acd0cb2c837f8ee7335b7881225da0126\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fd240f81b6850cb6db5eea91f7a91decf962757"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fd240f81b6850cb6db5eea91f7a91decf962757"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
