)]}'
{"id":"openstack%2Fdesignate~841492","triplet_id":"openstack%2Fdesignate~master~Ic86b8637e00525037b1c7735f9909d83397593a2","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-05-12 19:00:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic86b8637e00525037b1c7735f9909d83397593a2","subject":"Stop double compressing pdns logs","status":"MERGED","created":"2022-05-11 21:29:10.000000000","updated":"2022-05-12 19:01:12.000000000","submitted":"2022-05-12 19:00:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841492","meta_rev_id":"dbfbcd950733994d3d067d498d7ecac0cc152d28","_number":841492,"virtual_id_number":841492,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-12 19:00:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-05-12 00:16:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-12 06:18:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-12 17:51:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-11 22:49:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-12 00:16:18.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2022-05-12 06:18:11.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"4587b476f37e100edbf49f8c978a9a6fef6d9e05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-11 21:29:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b41774cb4965af8118d80fe18fa8df834a02aec5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 22:49:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2af7a73a02c5438d8d23a25ebd78e3f5 : SUCCESS in 47m 35s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9abdff845b384f8d810aec3800c6fab8 : SUCCESS in 50m 38s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/aff7b85d51534521be17015359df8c9a : SUCCESS in 49m 18s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/b4c9bcd5c5ec4b0ea9b0401ea75ce57b : SUCCESS in 48m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1c689d318af54b4c8d26252c12bb7fd8 : SUCCESS in 48m 13s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3d153dbb5856442ba44759961734ab9c : SUCCESS in 1h 04m 06s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f42f95b9f8ea46fd86946626228572e3 : SUCCESS in 1h 13m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3ab99602eb3c47eb91af70fc6595d620 : SUCCESS in 43m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4784b304e87842e2b1f443a9b1eb9eb9 : SUCCESS in 42m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f46b88702a77446597c5f780d653a44b : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27943edfd9984d6f89001e62e5f12169 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/776bfa06e1a346a4b9236597ff578d04 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99a99a0a890e4167b72ca279d03c1d4c : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92e4fad65c8140de89add475ac04d767 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/694a7561383a4ed79460835c2a11ca0a : SUCCESS in 7m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d41a9ac0062c43139e6c3c32f718ac9d : SUCCESS in 31m 16s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4db6286a6b9141ec96999444ee0b2c00 : SUCCESS in 4m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b8bc95361a265e070508f7e17f2d173affb73891","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-05-12 00:16:18.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7bfbbe8c11b56799b17d03078aff934cc5a01406","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-12 06:18:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"32399519c0b68465ce4acdbbfa71f17d3ef5acff","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-12 17:51:40.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1299a11ba5b338695a682ec0d15328d5c6e09f87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 17:52:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc1acd064d19133f629d92d253d4edf5613248e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 19:00:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6e5644b647cc44fe9c84253c9fa55c4f : SUCCESS in 46m 50s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/c2ee69e080aa450da1af76c2aa787d0b : SUCCESS in 48m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0bded536adef4a199225a3328384b31b : SUCCESS in 49m 30s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a6ef1ed3788f402097a8f03562a4de59 : SUCCESS in 1h 03m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/40ac03732e704dc2898c52e303dfbb34 : SUCCESS in 50m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/25324cd8b4f5496eb65f473bf0687d49 : SUCCESS in 51m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aecaae76a61466d97b2572c2371efed : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5dd3a1170a14cfc8f13f99212fc0430 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79adfacef3784d2997f8a77008982f5b : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f83050a537a44748ace99d3c748fa9f : SUCCESS in 8m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/15620e99c6d74997b17b10e3feb841ca : SUCCESS in 36m 52s","accounts_in_message":[],"_revision_number":1},{"id":"289c8579c46391b41e0ad972567e674be01a7c85","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 19:00:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dbfbcd950733994d3d067d498d7ecac0cc152d28","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 19:01:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0553778006d34ec2a2ff6f5312416421 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1e34d020522581a3eeda9fa8032dfde86f2f516","revisions":{"d1e34d020522581a3eeda9fa8032dfde86f2f516":{"kind":"REWORK","_number":1,"created":"2022-05-11 21:29:10.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/92/841492/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/92/841492/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/92/841492/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/92/841492/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/92/841492/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/92/841492/1"}}},"commit":{"parents":[{"commit":"4b516eaccc782c328f7e7bb9d65a7511515e44ef","subject":"Merge \"Enable ptr unset tempest test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b516eaccc782c328f7e7bb9d65a7511515e44ef"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-11 21:27:23.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-11 21:29:08.000000000","tz":0},"subject":"Stop double compressing pdns logs","message":"Stop double compressing pdns logs\n\nThis patch changes the powerDNS log collection to stop double compressing the log archive. The designate role compressed the log file and the zuul role also compresses the file.\nThis will allow the powerDNS logs to be directly viewable via the zuul logs page.\n\nChange-Id: Ic86b8637e00525037b1c7735f9909d83397593a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d1e34d020522581a3eeda9fa8032dfde86f2f516"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d1e34d020522581a3eeda9fa8032dfde86f2f516"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
