)]}'
{"id":"openstack%2Fdesignate~841493","triplet_id":"openstack%2Fdesignate~master~I910cf21bf0e3fd8951b8a9ac5f2fcaafea80a35d","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-05-13 03:39:06.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"I910cf21bf0e3fd8951b8a9ac5f2fcaafea80a35d","subject":"DNM: Testing unbound logs on centos","status":"ABANDONED","created":"2022-05-11 21:45:11.000000000","updated":"2022-05-13 15:14:31.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b4dbc3e5a37ecb6b360ee4e8858f21c5454a08f3","_number":841493,"virtual_id_number":841493,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-12 22:24:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-11 22:21:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"60052e119275eed0d8e5186c2a61c2df8b9eac54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-11 21:45:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afd657455a36ef493562e0bf0cadc7f53eb35e39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 22:21:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9-centos9stream-unbound-debug https://zuul.opendev.org/t/openstack/build/709a5810b3d14d09adc64d116e2f1da3 : SUCCESS in 31m 33s","accounts_in_message":[],"_revision_number":1},{"id":"5fb5fd72f2346128ec3ddee569f4876bb23c353a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-11 22:41:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f8a600b88c16bfedcfd90b18f5fd13f21434b1de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 22:55:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9-centos9stream-unbound-debug https://zuul.opendev.org/t/openstack/build/fed8b38474ef4625a51a0e2eeada6949 : FAILURE in 10m 06s","accounts_in_message":[],"_revision_number":2},{"id":"a0d9af0233d2524f994eb22d6a0bf82416956b73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-11 23:32:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3b3ce623da373746f76c83b687fd2bb7223f5ddf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 23:58:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9-centos9stream-unbound-debug https://zuul.opendev.org/t/openstack/build/4d1d6982e3244428b088e9b3e7ad735e : POST_FAILURE in 21m 02s","accounts_in_message":[],"_revision_number":3},{"id":"a5d34094f57ca5aa58549bdeee2da7a5131544b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-12 00:23:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"872df067a10ae7bbf8859503ed3d937c15285380","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 00:33:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9-centos9stream-unbound-debug https://zuul.opendev.org/t/openstack/build/d0254d16d96e48b0ad8dfa7530708de0 : FAILURE in 9m 16s","accounts_in_message":[],"_revision_number":4},{"id":"7f042314930c891db45c7ce420985b0ed9b114f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-12 21:46:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c6dcd05e507ce95c50304cc8e37e63cb7f9a21ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 22:24:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9-centos9stream-unbound-debug https://zuul.opendev.org/t/openstack/build/36468e98e8314e348c5c8e5079734d44 : SUCCESS in 36m 49s","accounts_in_message":[],"_revision_number":5},{"id":"533f0dd5052093735ee78eee139870b83bbf23ae","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-13 03:39:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c033ae6a8aeb8a108de03b59e1e8755b777ccbe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 04:22:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- designate-bind9-centos9stream-unbound-debug https://zuul.opendev.org/t/openstack/build/f54b790856bf4575ab90f99e77750928 : SUCCESS in 42m 22s","accounts_in_message":[],"_revision_number":5},{"id":"b4dbc3e5a37ecb6b360ee4e8858f21c5454a08f3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-13 15:14:31.000000000","message":"Abandoned\n\nUnbound logging on centos is now fixed!","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"889515ab5f396037bb8451f0063a2dbe0d706b7a","revisions":{"c47f50d93b85cd0ca43d9e3afaee61053b94626f":{"kind":"REWORK","_number":1,"created":"2022-05-11 21:45:11.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/93/841493/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/93/841493/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/841493/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/841493/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/841493/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/93/841493/1"}}},"commit":{"parents":[{"commit":"4b516eaccc782c328f7e7bb9d65a7511515e44ef","subject":"Merge \"Enable ptr unset tempest test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b516eaccc782c328f7e7bb9d65a7511515e44ef"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-11 21:44:47.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-11 21:45:09.000000000","tz":0},"subject":"DNM: Testing unbound logs on centos","message":"DNM: Testing unbound logs on centos\n\nChange-Id: I910cf21bf0e3fd8951b8a9ac5f2fcaafea80a35d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c47f50d93b85cd0ca43d9e3afaee61053b94626f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c47f50d93b85cd0ca43d9e3afaee61053b94626f"}]},"branch":"refs/heads/master"},"7208a727fe261176dab8e9e11bc0050cfdf29549":{"kind":"REWORK","_number":2,"created":"2022-05-11 22:41:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/93/841493/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/93/841493/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/841493/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/841493/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/841493/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/93/841493/2"}}},"commit":{"parents":[{"commit":"4b516eaccc782c328f7e7bb9d65a7511515e44ef","subject":"Merge \"Enable ptr unset tempest test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b516eaccc782c328f7e7bb9d65a7511515e44ef"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-11 21:44:47.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-11 22:41:36.000000000","tz":0},"subject":"DNM: Testing unbound logs on centos","message":"DNM: Testing unbound logs on centos\n\nChange-Id: I910cf21bf0e3fd8951b8a9ac5f2fcaafea80a35d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7208a727fe261176dab8e9e11bc0050cfdf29549"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7208a727fe261176dab8e9e11bc0050cfdf29549"}]},"branch":"refs/heads/master"},"64e20aecb8c8a09c0b47f530b12e8a2071223a0d":{"kind":"REWORK","_number":3,"created":"2022-05-11 23:32:07.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/93/841493/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/93/841493/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/841493/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/841493/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/841493/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/93/841493/3"}}},"commit":{"parents":[{"commit":"4b516eaccc782c328f7e7bb9d65a7511515e44ef","subject":"Merge \"Enable ptr unset tempest test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b516eaccc782c328f7e7bb9d65a7511515e44ef"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-11 21:44:47.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-11 23:32:00.000000000","tz":0},"subject":"DNM: Testing unbound logs on centos","message":"DNM: Testing unbound logs on centos\n\nChange-Id: I910cf21bf0e3fd8951b8a9ac5f2fcaafea80a35d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/64e20aecb8c8a09c0b47f530b12e8a2071223a0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/64e20aecb8c8a09c0b47f530b12e8a2071223a0d"}]},"branch":"refs/heads/master"},"b8c18a7da1fde55fa29547fdf4dc0d338fd64914":{"kind":"REWORK","_number":4,"created":"2022-05-12 00:23:12.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/93/841493/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/93/841493/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/841493/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/841493/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/841493/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/93/841493/4"}}},"commit":{"parents":[{"commit":"4b516eaccc782c328f7e7bb9d65a7511515e44ef","subject":"Merge \"Enable ptr unset tempest test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b516eaccc782c328f7e7bb9d65a7511515e44ef"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-11 21:44:47.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-12 00:23:06.000000000","tz":0},"subject":"DNM: Testing unbound logs on centos","message":"DNM: Testing unbound logs on centos\n\nChange-Id: I910cf21bf0e3fd8951b8a9ac5f2fcaafea80a35d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b8c18a7da1fde55fa29547fdf4dc0d338fd64914"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b8c18a7da1fde55fa29547fdf4dc0d338fd64914"}]},"branch":"refs/heads/master"},"889515ab5f396037bb8451f0063a2dbe0d706b7a":{"kind":"REWORK","_number":5,"created":"2022-05-12 21:46:03.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/93/841493/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/93/841493/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/841493/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/841493/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/841493/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/93/841493/5"}}},"commit":{"parents":[{"commit":"4b516eaccc782c328f7e7bb9d65a7511515e44ef","subject":"Merge \"Enable ptr unset tempest test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b516eaccc782c328f7e7bb9d65a7511515e44ef"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-11 21:44:47.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-12 21:45:54.000000000","tz":0},"subject":"DNM: Testing unbound logs on centos","message":"DNM: Testing unbound logs on centos\n\nChange-Id: I910cf21bf0e3fd8951b8a9ac5f2fcaafea80a35d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/889515ab5f396037bb8451f0063a2dbe0d706b7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/889515ab5f396037bb8451f0063a2dbe0d706b7a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
