)]}'
{"id":"openstack%2Fdesignate~841677","triplet_id":"openstack%2Fdesignate~master~I6dda01fe002cc0170102624baa67925b04a04114","project":"openstack/designate","branch":"master","topic":"remove_metrics","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-06-12 22:37:53.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I6dda01fe002cc0170102624baa67925b04a04114","subject":"Removed unused metrics backend","status":"ABANDONED","created":"2022-05-13 04:25:41.000000000","updated":"2022-06-13 23:50:49.000000000","total_comment_count":4,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"bacbfac160a1e80faf468ee0f6125d2cb60655c0","_number":841677,"virtual_id_number":841677,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-12 22:25:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-06-13 23:50:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-13 05:32:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-13 15:54:51.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"0907210cef64e0c671ad225fb9edc21812ebb33b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-13 04:25:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"120921aca8f9890fbe5903bac5b2133bc5a3b999","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 05:32:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d574e1b82d1e491ebdd2a783a4c45e39 : SUCCESS in 44m 37s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c33e69d836fc4c708d0fdda2ccfcaaf7 : FAILURE in 23m 02s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ae8bf2755dd4d0b97e0962c4c60c81c : SUCCESS in 45m 33s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/f300bcd96a0a4244bed4dd3b68724e59 : SUCCESS in 48m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bea1455e4d434a6a968bc796afa83577 : SUCCESS in 44m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e68543c09c18483ba9aea7a617510a2f : SUCCESS in 1h 02m 30s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d828c4492bc140489b27b626eb85a374 : SUCCESS in 1h 00m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7c2a2df367c547318cea3b961e3c0541 : SUCCESS in 48m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/086570dfeb224630b9c589a83f9ce6db : SUCCESS in 42m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3aafb61574c94c7c87be5540f15d33b8 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93243984b9724914a9b12d5887b86229 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/113f390791744a619fdeff20f4ddbf26 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d844ec164f745e289de25a7b9c4981d : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/759265abef6c481e90c66b76ff74a1df : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da8c8e6b8b9746bcae91897d966d3464 : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/273f8653be854032b79e0f198b77259e : SUCCESS in 3m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/188292a8725a42b98ceb820191fdb7e1 : SUCCESS in 28m 30s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6fada6fcc6014aafaa285bd57868cd51 : SUCCESS in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ae4251454c584c5adb593a0b5be81d24fb0a103","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-05-13 15:54:51.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe19c209886e9d2fb66169643301e9633cf3dd55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-14 03:15:36.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"feff21820347a97f64100481de7ffb7f8160e45c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-14 03:17:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e238e082b4b986a81fe0d89bdf658bec28d71a61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-14 04:20:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/726fc4fa470c4857b129707741b24eb3 : SUCCESS in 40m 40s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7dccc8b46bde4998a107f934d077ebcc : SUCCESS in 47m 17s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2632b299bbaa4e5a84dd6638388064c7 : SUCCESS in 48m 20s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/edd5ff343a9a4747ae24ac84dc3e02ca : SUCCESS in 40m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9cae1a998f5d4619b317733cbaa42192 : SUCCESS in 44m 20s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/62b972ab6f7a494a9d14a15f44251a4d : SUCCESS in 59m 30s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fc9a24a2e592485ba42234a159f87fb3 : SUCCESS in 53m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b3e31f614816419597d794637f9edc32 : SUCCESS in 40m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0f93c25e61ad45b0bf893948d1e38c63 : SUCCESS in 38m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acd8e74207434e12a21838bf6608b94e : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/329861b2c43e4e35acf79d664f44c046 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/483f7f424ab14a169282a4add6979043 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98c1e4487da14c358063a12cd3c1640b : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/011e518029ee4932a03fc902edab1a76 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a29c1fe3c6c4746b72a91ae1d62bf92 : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/200064ead7d54916bc9bf548bca9f17e : SUCCESS in 4m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6ca840bd38d44fa2b35fb5ced59a4fba : SUCCESS in 31m 52s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ad7745bbe2374d18a24fb5f4f686f970 : SUCCESS in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"18d08e3c9222f4b05199f5c08b3546f7d2aaa103","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-15 05:32:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f591dceb51109e4a037811c29c3f7c02120accb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-15 05:38:26.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"e5a11c26eca6feba5ea173945b9185dbc2a9f617","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-15 05:49:47.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"709c2393f733a81cb0f14c3efad25c5a130336eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-15 06:01:45.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7934e137b5e90b81b2894b2164f4da7a5b260d67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-15 06:02:12.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"9a1100e828949fbe1a34045cd3d898dccd851b84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-15 06:52:15.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/232c7b4a57194b318a211a2e224c5416 : SUCCESS in 32m 27s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/dec2603210984be3802d23c0d74dd843 : SUCCESS in 37m 32s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5ddf01559654fb1983f6b94117d5dab : SUCCESS in 30m 21s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/c9b076d295224fa4b8ad0455e4dffe84 : SUCCESS in 37m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ee0e2002ad794d8192585fa41d872b06 : SUCCESS in 31m 02s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/20c1dc1511c24761977f600ba885a30f : SUCCESS in 49m 11s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8bd8c3ca4a014507acab08ff3d2ea92f : SUCCESS in 46m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0986306100f64c1982a5d56bf0a72471 : SUCCESS in 31m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/006a6791d9324016891b9d87cbca8d1d : SUCCESS in 32m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d195d22bcee44580be3bc1bbc3230392 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/473e6fbafbe046eaa87a7274e75e9b1d : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d1deb0076ff4c84bca5e815b0b88b1b : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0b3020d48d948f08d3d77433f09cb18 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0e6afbfc7ee4beca0433e9bb5f170b3 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cbe185e1af74e9092138489c53c1d2a : SUCCESS in 6m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/03a62e16b7ce4ad2b1b73ce71ab2b7c2 : SUCCESS in 1m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c88a52b5d68043a09a00e3b0240162d5 : SUCCESS in 3m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d57949af0e344f7b841420905b6b2da1 : SUCCESS in 28m 42s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/96be22b9149f475dbc8666acdd95a8b0 : SUCCESS in 4m 00s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1a5312aec95874aa0ad4fca6042ce85ae6621e24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-15 19:55:36.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"4a97294cdf6f590ac91afd2c18ebe20442704784","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-15 20:56:11.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6b9dff10cc3f44589bd709c0349d209b : SUCCESS in 39m 50s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c6ab0b93e2e6485c9e2eb97dbd74805d : FAILURE in 11m 37s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f48b194bf1674b9f886384e223df8013 : SUCCESS in 41m 49s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/e7fcbcbb39c04f8581432e96d3b5b5dd : SUCCESS in 39m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3944c4f0944e4948904e671e009b2fc2 : SUCCESS in 37m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0e7db067a2684ac09b47cfa836fccedc : SUCCESS in 59m 01s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/430719ec96234527a3e3555e58d55ba6 : SUCCESS in 58m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f2ce5d9c18ad4c4db4d4b675f171bd76 : SUCCESS in 39m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a78eda50901e4b169f5087d24b3c3dd4 : SUCCESS in 39m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e46ebf5448e4df9b096170330b5e1e2 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c91c04b7602d445d9b841fceca1cbea6 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6806df729391431ebee893014738e33f : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f964289fc8c420c92057f1b06d62eaf : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27064144170f46feaef6ac6c5f24da54 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74caee36977d459c9f8e10dab07860e3 : SUCCESS in 9m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e75e4f45f35c41c89edfd20760dacb58 : SUCCESS in 2m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e027ccddce644667b4fd6ccc6e133078 : SUCCESS in 3m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7ad1e887e8d341b6a46428b46a97ce57 : SUCCESS in 34m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/55173ff981344a21b265767c178d14b8 : SUCCESS in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4332daedf751fce32b16e76746d4cda92b075226","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-18 05:19:33.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"6c3329c15d7c4f181f7ce5339452ceef817bb659","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 06:34:39.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bb03dcf1b1e848938015a32f0d0f8b6a : SUCCESS in 38m 47s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/13c8ac2e50fb439abfc8ee128fa7801f : SUCCESS in 56m 30s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/739f86310de7457086cac4120f70b993 : SUCCESS in 43m 07s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/7f499415c58b4427acbf8be4f750673a : FAILURE in 37m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3faa58d7be4b470cb3220db3b9b28215 : SUCCESS in 39m 39s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d03490b5bc184335b92cb55bb51a8420 : SUCCESS in 1h 09m 47s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/01934a7a4c0c4047ae9d072ed93734e6 : SUCCESS in 1h 05m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5dcc9463c9d64b6188beda11511fc2d5 : SUCCESS in 40m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b02bc02528164fe093f7061fe1c55f38 : SUCCESS in 43m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bff0d50d786469a87ae352848373bd0 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0adbee31e9c947be99026ca9a24a00fb : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96cf4fbda9d44ad4a2545e7e3787eb3d : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6ae85fff87f4d14adf5e784adefc643 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/443bff5ae7594492bb12603a206b7cb1 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/affd53e0bdf54fe483109f2005785b5e : SUCCESS in 7m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/020d8a5b50f943e596333c0f7247040d : SUCCESS in 2m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6754d8a7e2a243e59c34409ece6e0a72 : SUCCESS in 4m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c6a6be290c2f44d2ade9d13c8279db42 : SUCCESS in 29m 52s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b459e30fba7d431599ee4807fbac958a : SUCCESS in 4m 26s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ccd6bfe297d6c19ddfc0cc4f2c44d41cfe27b4bd","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-18 21:37:50.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"cc3a106b0c152cf69b063bba735f13dcaa31a777","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 22:54:27.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a52643d70fea40daa854eac176fe5631 : SUCCESS in 50m 27s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/918c95429ef047e69debf768f2ec4628 : FAILURE in 22m 34s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/90879bc20a51472c8af0a1da479e9b22 : SUCCESS in 55m 40s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/76371a8dd7c542de9b14c542350a11bf : SUCCESS in 52m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1d7425b346954c3685b4336288176181 : SUCCESS in 45m 52s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/81f43b2b1aed433cb0f9121bd3da1feb : SUCCESS in 1h 08m 34s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4056b5ede1b5440e931e45116b7bcd0e : SUCCESS in 1h 07m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3294cd24f6b6444c96ac7736b2f73ce1 : SUCCESS in 50m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e284df89fa3842339f4452140cb78d10 : SUCCESS in 51m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4844d7dca8ff4dd2881017b53bf3b1f7 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bf730163fbc4b998a29910be602af55 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bed5e3dbe7814942a41042fe498875d3 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/945073674600483aa0c8cc3715977075 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a78afd947a345878329571baeb33381 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd49bef16a09426cbbfa50ca2eb17d70 : SUCCESS in 7m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b209f3f0382a4b45b8bf7321cbe946b9 : SUCCESS in 1m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/353db84fc5f848ba8534e7a978556c7d : SUCCESS in 3m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fc9707033a6941c6b24e77fae3c26376 : SUCCESS in 36m 26s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6568c837a09042f886ab2721978b131e : SUCCESS in 4m 54s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3ab6fd937399fd62108ee1f0e3fa93dfa1552f14","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-05-24 21:03:00.000000000","message":"Patch Set 10: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"af438282b406424fff0299698da8eab252ca4dd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-25 03:02:20.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"c65286274f5502bfb0bdb08a1555be7dae9654c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 04:16:52.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f5be5dadc54f4d9397b11f438be62d80 : SUCCESS in 46m 11s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/887c5e5fca4d42ab8fc53c18f63e9d09 : FAILURE in 25m 39s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f5c7af1d3b942bb9295a8df20a1a5dd : SUCCESS in 56m 50s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ddbe8f6a34464f9a9661417f825be431 : SUCCESS in 57m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/859290c678444659b4a915b7a48dc489 : SUCCESS in 39m 23s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ad83fa456d9744a5b60afad877946f4b : SUCCESS in 58m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/98e73e09a8aa44f78bc31295735c77a1 : SUCCESS in 54m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ca73ebf06b54483ebf10a625b6932417 : SUCCESS in 41m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fd7dc752ea1a4477a8983881c4f90082 : SUCCESS in 37m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/484124e45d604be593d1aa41a90c5caa : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d865b40fe4d4725b70539f195a4e0d9 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0a52ce24cc64ee4b42c786f3fb8accb : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65c91b93d1f647b18df04ee36d6b4e35 : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb7b600ffed84153b04aa28ebe2c5284 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1c3cfaee11540e9ae9c7e2cdd60def3 : SUCCESS in 9m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3d480a0fb0a84cc39699ac34da8d7378 : SUCCESS in 2m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/111f82a1388146799a9d557593ba899d : SUCCESS in 4m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/89b19c6c0185472189efc7744796ac65 : SUCCESS in 31m 32s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9f10e37db5e747038ed93da6b3ebd8db : SUCCESS in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d0f59454a5caf8035d65ebee1e79d25922125da8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-27 17:40:20.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"06379931dee1540cde59bfa1a47670b24f135ed9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 18:38:10.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f24c7c79a65040f5a3bca7479c725e54 : SUCCESS in 51m 02s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/be8bf4de59cb4e79b1441e26c82cb8c3 : FAILURE in 10m 36s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9342787239c4ed7b90ecc85be5a890e : SUCCESS in 50m 15s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ec798e60eb9845559b217d23b0a9cfbd : SUCCESS in 43m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0297966d235440aba07f0a69090ce609 : SUCCESS in 51m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6169ca772ef54676b73f73ce0433249e : SUCCESS in 55m 54s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/56e5fff1e56b4b49bb763a15d163620f : SUCCESS in 55m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/59349ff8404140e09b36e42531224537 : SUCCESS in 46m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/53d4c74134e24d38902e128da4bf87c6 : SUCCESS in 39m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d871694919d94137a7706551e2008816 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c7001dca34748019d4f5467dcba044d : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b516ae8f480d48b88c3fa9eb0ad41c2b : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6477f3ff51644106a93426fff792b51f : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56ad85f86e884c5eb52951c6edcba6a7 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3608e53cf58743ddbdf1f1fa66b8a51f : SUCCESS in 8m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d4ec0beeb04b4a81881c7b4d0bd196cd : SUCCESS in 2m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3263dbb8f3e4f7b8433b754e7cce263 : SUCCESS in 4m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b03e56f1a13041468fb61f89aecf889d : SUCCESS in 30m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/acf3d655a93d4276a2ab56f421b8964a : SUCCESS in 4m 18s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"1ab05a3af70cb96d2e58b111221d748e0416e582","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-12 21:12:36.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"e5f264e33394d0f7bb8a9356665054d3a3cab3cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-12 22:25:35.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bcb3ddc1a5764388863fa447370277f0 : SUCCESS in 47m 43s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b1b5a9d7269a40688ef76de8543dea3c : FAILURE in 17m 00s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/87e75fd976fa4d3cb3fe62a7e0eb6a81 : SUCCESS in 58m 15s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/07830613e9d54a36af0b20d257c70d59 : SUCCESS in 43m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4b1d7290fde7465eb852e2aa21f1205c : SUCCESS in 46m 45s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/481db1c3f6794f3f9e4247d0126cb815 : SUCCESS in 1h 06m 16s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/06f98074d74a4755be1dc15beba8088b : SUCCESS in 59m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/679de5140e48424192cbda2e6ef5755b : SUCCESS in 43m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a0fdb48cb9b8462b82cc1bfd4c92fc1c : SUCCESS in 46m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bb48dc5530440ffb1258742152403b9 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea4d4fdb5a16465188221b177c080be6 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5869109dfa9d459987d165b45d5f666f : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/603d7c706d5a44e5a148850ca9eb3eda : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ade9ceea523a43799e5aabe95cb8c450 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b77bf038d89f4e558ac6e2b1bbbb00be : SUCCESS in 7m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b1f80334d66e47cb80cd200403198e54 : SUCCESS in 2m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ddff6f70f6e43a1abb5726cf503bf3e : SUCCESS in 3m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9669d2ee4a854fe9802658608cac1a53 : SUCCESS in 25m 09s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5a40c7862f57485bbad5c1c0b668fadc : SUCCESS in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"a27093e37ef5d787c7e3ca8a28603f33cbc44fbd","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-12 22:37:53.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":13},{"id":"94a0d7a7b733998f88c403ba9363fb208a59d01b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-13 23:50:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"d7579909dbc01cbf7f5b2e291a312649399b67a0","revisions":{"c72b1b435b1be069266c1780d2ce49f03bb1497a":{"kind":"REWORK","_number":1,"created":"2022-05-13 04:25:41.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/77/841677/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/841677/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/841677/1"}}},"commit":{"parents":[{"commit":"fffbde92e5cda24e422656f50444049348eb46de","subject":"Move Zone XFR from mdns to worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fffbde92e5cda24e422656f50444049348eb46de"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-13 04:24:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-13 04:24:36.000000000","tz":-420},"subject":"Removed unused metrics backend","message":"Removed unused metrics backend\n\nAll calls using the metrics backend has been removed and the\nmonascastatsd no longer serves a purpose.\n\nChange-Id: I6dda01fe002cc0170102624baa67925b04a04114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c72b1b435b1be069266c1780d2ce49f03bb1497a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c72b1b435b1be069266c1780d2ce49f03bb1497a"}]},"branch":"refs/heads/master"},"ae3f1842a01af2405f5a662cf160a2e85f75fba8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-05-14 03:15:36.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/77/841677/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/841677/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/841677/2"}}},"commit":{"parents":[{"commit":"f0644e8171992a73beb6b82cc238afe9e4473131","subject":"Move Zone XFR from mdns to worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0644e8171992a73beb6b82cc238afe9e4473131"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-13 04:24:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-14 03:15:36.000000000","tz":0},"subject":"Removed unused metrics backend","message":"Removed unused metrics backend\n\nAll calls using the metrics backend has been removed and the\nmonascastatsd no longer serves a purpose.\n\nChange-Id: I6dda01fe002cc0170102624baa67925b04a04114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ae3f1842a01af2405f5a662cf160a2e85f75fba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ae3f1842a01af2405f5a662cf160a2e85f75fba8"}]},"branch":"refs/heads/master","description":"Rebase"},"4cba7a721d771b717570d900731df953e2a00ffe":{"kind":"REWORK","_number":3,"created":"2022-05-14 03:17:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/77/841677/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/841677/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/841677/3"}}},"commit":{"parents":[{"commit":"f0644e8171992a73beb6b82cc238afe9e4473131","subject":"Move Zone XFR from mdns to worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0644e8171992a73beb6b82cc238afe9e4473131"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-13 04:24:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-14 03:17:41.000000000","tz":-420},"subject":"Removed unused metrics backend","message":"Removed unused metrics backend\n\nAll calls using the metrics backend has been removed and the\nmonascastatsd no longer serves a purpose.\n\nChange-Id: I6dda01fe002cc0170102624baa67925b04a04114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4cba7a721d771b717570d900731df953e2a00ffe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4cba7a721d771b717570d900731df953e2a00ffe"}]},"branch":"refs/heads/master"},"07655853b1cfbde25323810b1567e37ec273578b":{"kind":"REWORK","_number":4,"created":"2022-05-15 05:32:38.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/77/841677/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/841677/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/841677/4"}}},"commit":{"parents":[{"commit":"f0644e8171992a73beb6b82cc238afe9e4473131","subject":"Move Zone XFR from mdns to worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0644e8171992a73beb6b82cc238afe9e4473131"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-13 04:24:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-15 05:32:21.000000000","tz":-420},"subject":"Removed unused metrics backend","message":"Removed unused metrics backend\n\nAll calls using the metrics backend has been removed and the\nmonascastatsd no longer serves a purpose.\n\nChange-Id: I6dda01fe002cc0170102624baa67925b04a04114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/07655853b1cfbde25323810b1567e37ec273578b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/07655853b1cfbde25323810b1567e37ec273578b"}]},"branch":"refs/heads/master"},"6d0728ef4dadb2f8dd84d4c8c28639f99006b8d0":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-05-15 05:38:26.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/77/841677/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/841677/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/841677/5"}}},"commit":{"parents":[{"commit":"75d5739aff642aef9e70c1d85f31ceded3a7319c","subject":"Move Zone XFR from mdns to worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75d5739aff642aef9e70c1d85f31ceded3a7319c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-13 04:24:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-15 05:38:26.000000000","tz":0},"subject":"Removed unused metrics backend","message":"Removed unused metrics backend\n\nAll calls using the metrics backend has been removed and the\nmonascastatsd no longer serves a purpose.\n\nChange-Id: I6dda01fe002cc0170102624baa67925b04a04114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6d0728ef4dadb2f8dd84d4c8c28639f99006b8d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6d0728ef4dadb2f8dd84d4c8c28639f99006b8d0"}]},"branch":"refs/heads/master","description":"Rebase"},"8b40b48bc9e9381a75ad2df0bbe787d6b9104e9c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-05-15 05:49:47.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/77/841677/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/841677/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/841677/6"}}},"commit":{"parents":[{"commit":"e1b92ab37c8abb60a6fed4f42779ed4f74be6ded","subject":"Remove unused render_template_to_file method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e1b92ab37c8abb60a6fed4f42779ed4f74be6ded"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-13 04:24:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-15 05:49:47.000000000","tz":0},"subject":"Removed unused metrics backend","message":"Removed unused metrics backend\n\nAll calls using the metrics backend has been removed and the\nmonascastatsd no longer serves a purpose.\n\nChange-Id: I6dda01fe002cc0170102624baa67925b04a04114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8b40b48bc9e9381a75ad2df0bbe787d6b9104e9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8b40b48bc9e9381a75ad2df0bbe787d6b9104e9c"}]},"branch":"refs/heads/master","description":"Rebase"},"69a814193b17720ac5f10ca63e2f48355c534f51":{"kind":"REWORK","_number":7,"created":"2022-05-15 06:01:45.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/77/841677/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/841677/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/841677/7"}}},"commit":{"parents":[{"commit":"e1b92ab37c8abb60a6fed4f42779ed4f74be6ded","subject":"Remove unused render_template_to_file method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e1b92ab37c8abb60a6fed4f42779ed4f74be6ded"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-13 04:24:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-15 06:01:36.000000000","tz":-420},"subject":"Removed unused metrics backend","message":"Removed unused metrics backend\n\nAll calls using the metrics backend has been removed and the\nmonascastatsd no longer serves a purpose.\n\nChange-Id: I6dda01fe002cc0170102624baa67925b04a04114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/69a814193b17720ac5f10ca63e2f48355c534f51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/69a814193b17720ac5f10ca63e2f48355c534f51"}]},"branch":"refs/heads/master"},"50aa5a47132e14a8e5eef0710d763932b8681664":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-05-15 06:02:12.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/77/841677/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/841677/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/841677/8"}}},"commit":{"parents":[{"commit":"1a2f4565ce5706f18c806125b9cab11bef78628b","subject":"Move Zone XFR from mdns to worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1a2f4565ce5706f18c806125b9cab11bef78628b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-13 04:24:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-15 06:02:12.000000000","tz":0},"subject":"Removed unused metrics backend","message":"Removed unused metrics backend\n\nAll calls using the metrics backend has been removed and the\nmonascastatsd no longer serves a purpose.\n\nChange-Id: I6dda01fe002cc0170102624baa67925b04a04114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/50aa5a47132e14a8e5eef0710d763932b8681664"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/50aa5a47132e14a8e5eef0710d763932b8681664"}]},"branch":"refs/heads/master","description":"Rebase"},"4a356ccc8b7b93890a00c277214b3f1e8c15a347":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2022-05-15 19:55:36.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/77/841677/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/841677/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/841677/9"}}},"commit":{"parents":[{"commit":"4c32778a1bfdaea8f199e48c0a9f4035f9b46760","subject":"Move Zone XFR from mdns to worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4c32778a1bfdaea8f199e48c0a9f4035f9b46760"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-13 04:24:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-15 19:55:36.000000000","tz":0},"subject":"Removed unused metrics backend","message":"Removed unused metrics backend\n\nAll calls using the metrics backend has been removed and the\nmonascastatsd no longer serves a purpose.\n\nChange-Id: I6dda01fe002cc0170102624baa67925b04a04114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4a356ccc8b7b93890a00c277214b3f1e8c15a347"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4a356ccc8b7b93890a00c277214b3f1e8c15a347"}]},"branch":"refs/heads/master","description":"Rebase"},"cf8f197fa40b13429cd7985a66d095f482ca0d97":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2022-05-18 05:19:33.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/77/841677/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/841677/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/841677/10"}}},"commit":{"parents":[{"commit":"2ee087dac531002302c9217c67be9f6a50eafc4d","subject":"Move Zone XFR from mdns to worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2ee087dac531002302c9217c67be9f6a50eafc4d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-13 04:24:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-18 05:19:33.000000000","tz":0},"subject":"Removed unused metrics backend","message":"Removed unused metrics backend\n\nAll calls using the metrics backend has been removed and the\nmonascastatsd no longer serves a purpose.\n\nChange-Id: I6dda01fe002cc0170102624baa67925b04a04114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cf8f197fa40b13429cd7985a66d095f482ca0d97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cf8f197fa40b13429cd7985a66d095f482ca0d97"}]},"branch":"refs/heads/master","description":"Rebase"},"66df67620bab754c3c6460b380f7fa77905b2e42":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-05-25 03:02:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/77/841677/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/841677/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/841677/11"}}},"commit":{"parents":[{"commit":"cef332443fd776e55a7c4eafbb4df1f756b0b12c","subject":"Move Zone XFR from mdns to worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cef332443fd776e55a7c4eafbb4df1f756b0b12c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-13 04:24:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-25 03:02:20.000000000","tz":0},"subject":"Removed unused metrics backend","message":"Removed unused metrics backend\n\nAll calls using the metrics backend has been removed and the\nmonascastatsd no longer serves a purpose.\n\nChange-Id: I6dda01fe002cc0170102624baa67925b04a04114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/66df67620bab754c3c6460b380f7fa77905b2e42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/66df67620bab754c3c6460b380f7fa77905b2e42"}]},"branch":"refs/heads/master","description":"Rebase"},"b532e706b119a6e0b1a5e08c6a59576d7aea1166":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2022-05-27 17:40:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/77/841677/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/841677/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/841677/12"}}},"commit":{"parents":[{"commit":"9a4782b3691f446c03258a474e57b3d288c02620","subject":"Move Zone XFR from mdns to worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a4782b3691f446c03258a474e57b3d288c02620"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-13 04:24:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-27 17:40:20.000000000","tz":0},"subject":"Removed unused metrics backend","message":"Removed unused metrics backend\n\nAll calls using the metrics backend has been removed and the\nmonascastatsd no longer serves a purpose.\n\nChange-Id: I6dda01fe002cc0170102624baa67925b04a04114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b532e706b119a6e0b1a5e08c6a59576d7aea1166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b532e706b119a6e0b1a5e08c6a59576d7aea1166"}]},"branch":"refs/heads/master","description":"Rebase"},"d7579909dbc01cbf7f5b2e291a312649399b67a0":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2022-06-12 21:12:36.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/77/841677/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/841677/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/841677/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/841677/13"}}},"commit":{"parents":[{"commit":"99f55f04654c1f8ba7859af304962a86ecc5e8cf","subject":"Move Zone XFR from mdns to worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/99f55f04654c1f8ba7859af304962a86ecc5e8cf"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-13 04:24:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-06-12 21:12:36.000000000","tz":0},"subject":"Removed unused metrics backend","message":"Removed unused metrics backend\n\nAll calls using the metrics backend has been removed and the\nmonascastatsd no longer serves a purpose.\n\nChange-Id: I6dda01fe002cc0170102624baa67925b04a04114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d7579909dbc01cbf7f5b2e291a312649399b67a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d7579909dbc01cbf7f5b2e291a312649399b67a0"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
