)]}'
{"id":"openstack%2Fdesignate~842560","triplet_id":"openstack%2Fdesignate~master~I94324104a7ed89e757d48bb250d8b1132026ce46","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-05-27 18:39:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I94324104a7ed89e757d48bb250d8b1132026ce46","subject":"Update zuul queue configuration","status":"MERGED","created":"2022-05-19 16:11:35.000000000","updated":"2022-05-27 18:42:46.000000000","submitted":"2022-05-27 18:39:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842560","meta_rev_id":"14cf78118deb25babba7e11b4fdbe93d8db36cfc","_number":842560,"virtual_id_number":842560,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-27 18:39:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-05-24 20:58:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":2,"date":"2022-05-26 14:44:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-27 17:41:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-27 17:41:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":1,"date":"2022-05-19 16:23:45.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-19 17:16:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-24 20:58:31.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2022-05-27 17:41:03.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"06cef5299a2dac26f0e57c34199e0491109faee2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-19 16:11:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5283dea69f47239237ad06a5d11cef495b05396c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-19 16:23:45.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"221593af101bb4ae61d9a849da168bb19bcad4f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 17:16:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bf847f4aed4e4bf3b43132ad27420f82 : SUCCESS in 41m 47s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5e830cb56678425daa1db51cf13652f3 : SUCCESS in 52m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5096f96bd8414ec58ca2cda0896e24a8 : SUCCESS in 42m 45s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/180aa40c0c5a48f0ae6847d20397b1bd : SUCCESS in 42m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/972ad576b8de4469b4c2778d350ab8a6 : SUCCESS in 39m 44s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/bcdd487899d643918d53e7b1e78a37df : SUCCESS in 1h 01m 19s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d6ab45731510466cb6cbe0bb525981b2 : SUCCESS in 58m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ef53c59a0d1e45b09d979bf8170669f8 : FAILURE in 40m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f46c252b144743dd99bcf082521c20f7 : SUCCESS in 39m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38742d6f9bb44ac397d1d7c1d4ca5b72 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7022561ad99341308d01733958ddde28 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cf69fbdde554bc9be1fc1fdd179e9b6 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ef0309c405b40b4ac15441012078c99 : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2fab7006b6644b56b379340d3a20a4db : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e58b11cfe4b44ba0918b7c50117d18ef : SUCCESS in 7m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/82f00674be0148bebce299dff96e9153 : SUCCESS in 38m 43s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c394b592f24f4872ad8461e0333dbe18 : SUCCESS in 4m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eeee350812d477d6faa3ed6d09cf550d7e5de263","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-24 17:37:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb2bb721a46f82ee58e3c5e44e86a9a320241e78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 18:43:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/70c777c670d24ef9ac1f3f468fec9a49 : SUCCESS in 34m 02s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c8d7e6992ed2488aba087378e98241c3 : FAILURE in 11m 16s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8788b4c0b2140ca9e289729bfad41f5 : SUCCESS in 39m 43s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2dd6fa185e7544388ac47614b5fd02cb : SUCCESS in 47m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ce6b5871dd744a41a33dd7d7c2eddad1 : SUCCESS in 50m 20s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/24bd48e3625748b0bee6919918b72b31 : SUCCESS in 1h 03m 34s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/987974a7ef124c86b3288b06a3d1b6ae : SUCCESS in 56m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8818bee071d94f0a8f054afaedd193ba : SUCCESS in 45m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4782d9ffd1134a52add6884149fc0370 : SUCCESS in 46m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c83e86220b3e4a7ea1af9d3e820c62cf : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4c2ae9c8a8342b18891a7a62d093072 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31a777a6979646d5b52cf9275a085a80 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0ac7e65eb784ae0a57b12b80560a4a1 : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/207c2f826ba14195828d084d0f0d3287 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/618e0c1b54d0407c88c46d1977dc8703 : SUCCESS in 7m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/af9714d473044549896e92bb1b2974c1 : SUCCESS in 37m 00s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6641f0ecee524f19a4aba75f92c8c876 : SUCCESS in 5m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e314b5fb0f04791d15788a85859657cf3974e52c","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-05-24 20:58:31.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d3fd78a2663a43029b7b3d48624c921930ada8d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-26 14:44:36.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7d9051ab5052be5f3baf448824b34934ca765c15","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-27 17:41:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"667d0ae1162bf7bfe3489f7722c32671ec129ff9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 17:41:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ea767665bd78c0716c40419d8929b6435f408c96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 18:39:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/27505fc788ac468093ab871e86f98b97 : SUCCESS in 46m 07s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1ccb80ac232e482e954f907026fe9df6 : SUCCESS in 47m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/804e9b73053742f3ae33d17d23e9efba : SUCCESS in 36m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a9d449cb130546da8d606c9b7a4b3b22 : SUCCESS in 56m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6b023e98b67c40fda305e7abfd3ddeea : SUCCESS in 47m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bcc8bbcb82d34ef681057912e02a0cd9 : SUCCESS in 42m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f1a46bf90eb4a0eb56dfe5fb3cf0247 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70554b0ce8ef42668aae37b53b0099ca : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6cc70dc5f4447bc8a62e21d0cebbf1e : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c11d1793db1d414bab2246a522def583 : SUCCESS in 8m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1127bd72b1814cf9a0574e724e80f8dc : SUCCESS in 30m 22s","accounts_in_message":[],"_revision_number":1},{"id":"6b4c3063e35a9616fabe20eaaef067bb31716b45","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 18:39:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"14cf78118deb25babba7e11b4fdbe93d8db36cfc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 18:42:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaa6b5213b2b43d7abc1d20ea197e422 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fe9fb45c21a7be2f88600e1f29fd37f56244cc1a","revisions":{"fe9fb45c21a7be2f88600e1f29fd37f56244cc1a":{"kind":"REWORK","_number":1,"created":"2022-05-19 16:11:35.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/60/842560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/842560/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/842560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/842560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/842560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/842560/1"}}},"commit":{"parents":[{"commit":"4e4300ce14e7da6d19dca135616ce4427ae7adee","subject":"Merge \"Adds api_ca_cert configuration variable to pools.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4e4300ce14e7da6d19dca135616ce4427ae7adee"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-19 16:10:12.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-05-19 16:11:33.000000000","tz":0},"subject":"Update zuul queue configuration","message":"Update zuul queue configuration\n\nZuul is changing how queues are declared[1]. This patch updates Designate for the new way to define the queue.\n\n[1] https://lists.zuul-ci.org/pipermail/zuul-discuss/2022-May/001801.html\n\nChange-Id: I94324104a7ed89e757d48bb250d8b1132026ce46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fe9fb45c21a7be2f88600e1f29fd37f56244cc1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fe9fb45c21a7be2f88600e1f29fd37f56244cc1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
