)]}'
{"id":"openstack%2Fdesignate~843707","triplet_id":"openstack%2Fdesignate~master~I9c2616153c4a7431c04e8d19780429477d828ddc","project":"openstack/designate","branch":"master","topic":"ussuri-pin-tempest","hashtags":[],"change_id":"I9c2616153c4a7431c04e8d19780429477d828ddc","subject":"DNM: Testing tempest pin on stable/ussuri","status":"ABANDONED","created":"2022-05-28 06:22:01.000000000","updated":"2022-05-31 21:27:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bd5c20cea498375562c71db21cd4b64e499b4757","_number":843707,"virtual_id_number":843707,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-28 07:26:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-28 07:26:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc7ab38bf8a0ffa7fb716ccbaf345a9956fc5c00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-28 06:22:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4eda498fe408cd3a0bdd39a2ce960e2f3c270ba5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 07:26:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ce85b5fc05a549cc9965a4e0eb232081 : SUCCESS in 45m 09s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f757e3e1bac94821a2f8ac8c05af1a8d : FAILURE in 12m 31s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7fabd537c384bfbb69c96c0c2a276e2 : SUCCESS in 43m 16s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/39272967e56d4ceca9512189abfd3bf1 : SUCCESS in 49m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e01dd64534b349ffbc95c0d345948060 : SUCCESS in 37m 17s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/64aee35d2db3418eaf24c671f4b67b26 : SUCCESS in 58m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8ccf9aa4c1794bdcae31f13fa1be7551 : SUCCESS in 59m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/249f50e7cd1b445d9ac042107c016a95 : SUCCESS in 38m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/093e677a94f041b38a5e19fef5097734 : SUCCESS in 39m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6da8e81a7c944e5f8a0ebbe187ee3c4e : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bd560546af34e37b79feee350c57d8b : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0858d1ee3e7476abe96c30ca9a98bb8 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ca64c44d3aa48d0abd6efdb0c322a09 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80c76f9089054b3684f42043e875f1e5 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d9c9b7e9f524702a24f0c78f77247bc : SUCCESS in 7m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/240e8e5dfbe04af8b7029b85270c4179 : SUCCESS in 42m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8109a9aabc3a4e3286c67c4accdef4b6 : SUCCESS in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd5c20cea498375562c71db21cd4b64e499b4757","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-31 21:27:58.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"281bce53037df3a4b0cf99cc610313770d571c46","revisions":{"281bce53037df3a4b0cf99cc610313770d571c46":{"kind":"REWORK","_number":1,"created":"2022-05-28 06:22:01.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/07/843707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/07/843707/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/843707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/843707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/843707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/07/843707/1"}}},"commit":{"parents":[{"commit":"ca74e78687dc69479a7f051b0040c9767ae5ca0c","subject":"Merge \"Update zuul queue configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ca74e78687dc69479a7f051b0040c9767ae5ca0c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-28 06:21:49.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-28 06:21:49.000000000","tz":-300},"subject":"DNM: Testing tempest pin on stable/ussuri","message":"DNM: Testing tempest pin on stable/ussuri\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/838051\nChange-Id: I9c2616153c4a7431c04e8d19780429477d828ddc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/281bce53037df3a4b0cf99cc610313770d571c46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/281bce53037df3a4b0cf99cc610313770d571c46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
