)]}'
{"id":"openstack%2Fdesignate~843709","triplet_id":"openstack%2Fdesignate~stable%2Fussuri~I9c2616153c4a7431c04e8d19780429477d828ddc","project":"openstack/designate","branch":"stable/ussuri","topic":"ussuri-pin-tempest","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-05-31 17:24:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9c2616153c4a7431c04e8d19780429477d828ddc","subject":"Pin designate-tempest-plugin for stable/ussuri jobs","status":"MERGED","created":"2022-05-28 06:27:25.000000000","updated":"2022-05-31 17:28:13.000000000","submitted":"2022-05-31 17:24:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843709-ussuri-pin-tempest","meta_rev_id":"236c9a6aba56cd54ce6dfd6755a60aa57f65c4c9","_number":843709,"virtual_id_number":843709,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-31 17:24:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-05-31 16:20:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-05-31 15:44:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-05-31 16:20:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-28 07:30:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-31 15:44:33.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-05-31 16:20:33.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"ca79088164c0035ea5ba1ec9afac602f079cbac7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-28 06:27:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1c43eea9da82222fb535eee2374ac6b4f2287ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 07:30:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/425695efaa70412f8f67ccff9cdbb062 : FAILURE in 36m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2929fb6cdbe149adb9c29e7fffcfaedc : FAILURE in 35m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6a5e6a6b277a4da69d0fee3ca38199b1 : FAILURE in 52m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f79989cde2e34f8eb721fd23c5f950cf : FAILURE in 39m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/dbeb28acff0f4d70a4c1f152e85d8b56 : FAILURE in 37m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0184c75fc962415b854eb7eefdb6d918 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb5d18708c5a4cd2a55c8bc66fcb05d7 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab12988d44dc444590a172eb1436cb70 : SUCCESS in 6m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f5c1db4303d46a4b7ff348faa6c8471 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fea2d2a245c841dc9eee8188ee48b2d5 : SUCCESS in 7m 30s","accounts_in_message":[],"_revision_number":1},{"id":"80a83ba5506b929ad2bb5cade266b3693f52d5e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-28 18:43:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"03e8e0d6e3f75c5342e3ad7ad8890765f7759c0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-28 18:44:53.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f33eb68bb1c6b89b004ceb3295ff4bd376806805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 19:34:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/988429eb8a674481bec76162103e141e : SUCCESS in 28m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d595a9140a3f40829e0a904560a9b38c : SUCCESS in 28m 29s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2e3c7db486e1451c8d7a18b0266ee1a6 : FAILURE in 43m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6a398d0accf24f2cb92cd9cb291a8f8c : SUCCESS in 23m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a3ced94879214567ba203e42e99a5c27 : SUCCESS in 40m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ef936b28b4a45b1bb2573a7c71b6a2f : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/269c137661bf4d1a9d7b49bbcd3460b2 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d827443d6da24ebabf486a706f26d7ba : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9f56a336bc445c0b884fcb73d9a9ce2 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff4fd20d4b734537b136776bbaab4c37 : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":3},{"id":"bf2ed7d779cf22c3bc210eade271fcefb92d174a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-28 19:48:42.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"1ce53b75bda52a94739227f90e68b9b070aeedfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 21:00:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d2c8f565d9ea4b6fab07284fd320f9b0 : SUCCESS in 39m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a2f640fd7dd846a6be2cbb42707ec811 : SUCCESS in 41m 11s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/69b80b3d289a4f179e40fb558efedf28 : SUCCESS in 1h 06m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1749fd24033e4a3b98b2bc66cac909fc : SUCCESS in 39m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e7a6eac648a5466e8e6bcd9df3a1722e : SUCCESS in 44m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/280c941a119c410da7af4c26c2c3ce85 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55b59c9d3fd64b1e837c947318ea5356 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe37f974ee944bb79bfe404ee69fa553 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/31a26c4d3685499b8b3c7fb37e2734ba : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66c7d18020cd49bf8b4d60da9d8de470 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":4},{"id":"77b63c8f51c3df89be30d52fe5751c64c7a28479","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-31 15:44:33.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"db5503fe031b37fbae7c0687c8a5e743fb45f251","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-05-31 16:20:33.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"930d87aa7f822d3b7d3300174eaa4eade4f4b6a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 16:22:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f346effbbcdd10deba5c143791d2ad74eacc4460","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 17:24:59.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f57842571339437381af6db10dabae8c : SUCCESS in 47m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4b3a391fb220443c96b6ad55058c245b : SUCCESS in 41m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4e48981f17ec4181bd011f03bbc9ea5e : SUCCESS in 1h 01m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/05603c9bdc6040cc86553e5e82171416 : SUCCESS in 41m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/dfe2df8239ca4b0480e47ddcf6cf4ed0 : SUCCESS in 29m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c88a539719514b8583bad437c8b3369a : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c17e59bfed64359bd56d40120e55cff : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/44ffc6ef2f3744f1809bd49659fa8734 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d5c5bf358044f8ab1db7f4db04ea2b9 : SUCCESS in 8m 10s","accounts_in_message":[],"_revision_number":4},{"id":"dcf22e0c5ed63a98a730a49f0b93d1ab92781db9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 17:24:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"236c9a6aba56cd54ce6dfd6755a60aa57f65c4c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 17:28:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0421ad2584b24a50a6e1fe5fcef629a7 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8e9176c8769a35f797cc1f23d3cc49a2e77c6233","revisions":{"94d9020acbde513641248037e3defa10f6a6389e":{"kind":"REWORK","_number":1,"created":"2022-05-28 06:27:25.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/09/843709/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/843709/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/843709/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/843709/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/843709/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/843709/1"}}},"commit":{"parents":[{"commit":"a9ab61bc1d98248af108d895ce2cb7696818dcda","subject":"Fix pool update with duplicate masters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a9ab61bc1d98248af108d895ce2cb7696818dcda"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-28 06:26:45.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-28 06:26:45.000000000","tz":-300},"subject":"DNM: Testing tempest pin on stable/ussuri","message":"DNM: Testing tempest pin on stable/ussuri\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/838051\nChange-Id: I9c2616153c4a7431c04e8d19780429477d828ddc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/94d9020acbde513641248037e3defa10f6a6389e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/94d9020acbde513641248037e3defa10f6a6389e"}]},"branch":"refs/heads/stable/ussuri"},"864826892e17ff686ce78b8b972e26b2b8855cc5":{"kind":"REWORK","_number":2,"created":"2022-05-28 18:43:08.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/09/843709/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/843709/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/843709/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/843709/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/843709/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/843709/2"}}},"commit":{"parents":[{"commit":"a9ab61bc1d98248af108d895ce2cb7696818dcda","subject":"Fix pool update with duplicate masters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a9ab61bc1d98248af108d895ce2cb7696818dcda"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-28 06:26:45.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-28 18:42:59.000000000","tz":-300},"subject":"Pin designate-tempest-plugin for stable/ussuri jobs","message":"Pin designate-tempest-plugin for stable/ussuri jobs\n\nAs stable/victoria are in EM phase, and QA team pinning\ntempest 26.1.0 for stable/ussuri\n\n- https://review.opendev.org/c/openstack/devstack/+/838051\n\nTo have compatible designate-tempest-plugin version with Tempest\n26.1.0, we need to pin designate-tempest-plugin 0.8.0 version.\n\nmaster designate-tempest-plugin is not comaptbile with Tempest 26.1.0\n- https://zuul.opendev.org/t/openstack/build/2929fb6cdbe149adb9c29e7fffcfaedc\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/838051\nChange-Id: I9c2616153c4a7431c04e8d19780429477d828ddc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/864826892e17ff686ce78b8b972e26b2b8855cc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/864826892e17ff686ce78b8b972e26b2b8855cc5"}]},"branch":"refs/heads/stable/ussuri"},"96e0412f2677c8806cb9ed821932aa3186816ba1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-05-28 18:44:53.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/09/843709/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/843709/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/843709/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/843709/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/843709/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/843709/3"}}},"commit":{"parents":[{"commit":"a9ab61bc1d98248af108d895ce2cb7696818dcda","subject":"Fix pool update with duplicate masters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a9ab61bc1d98248af108d895ce2cb7696818dcda"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-28 06:26:45.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-05-28 18:44:53.000000000","tz":0},"subject":"Pin designate-tempest-plugin for stable/ussuri jobs","message":"Pin designate-tempest-plugin for stable/ussuri jobs\n\nAs stable/victoria are in EM phase, and QA team pinning\ntempest 26.1.0 for stable/ussuri\n\n- https://review.opendev.org/c/openstack/devstack/+/838051\n\nTo have compatible designate-tempest-plugin version with Tempest\n26.1.0, we need to pin designate-tempest-plugin 0.8.0 version.\n\nmaster designate-tempest-plugin is not comaptbile with Tempest 26.1.0\n- https://zuul.opendev.org/t/openstack/build/2929fb6cdbe149adb9c29e7fffcfaedc\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/838051\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/838053\nChange-Id: I9c2616153c4a7431c04e8d19780429477d828ddc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/96e0412f2677c8806cb9ed821932aa3186816ba1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/96e0412f2677c8806cb9ed821932aa3186816ba1"}]},"branch":"refs/heads/stable/ussuri","description":"Edit commit message"},"8e9176c8769a35f797cc1f23d3cc49a2e77c6233":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-05-28 19:48:42.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/09/843709/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/843709/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/843709/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/843709/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/843709/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/843709/4"}}},"commit":{"parents":[{"commit":"a9ab61bc1d98248af108d895ce2cb7696818dcda","subject":"Fix pool update with duplicate masters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a9ab61bc1d98248af108d895ce2cb7696818dcda"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-28 06:26:45.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-05-28 19:48:41.000000000","tz":0},"subject":"Pin designate-tempest-plugin for stable/ussuri jobs","message":"Pin designate-tempest-plugin for stable/ussuri jobs\n\nAs stable/victoria are in EM phase, and QA team pinning\ntempest 26.1.0 for stable/ussuri\n\n- https://review.opendev.org/c/openstack/devstack/+/838051\n\nTo have compatible designate-tempest-plugin version with Tempest\n26.1.0, we need to pin designate-tempest-plugin 0.8.0 version.\n\nmaster designate-tempest-plugin is not comaptbile with Tempest 26.1.0\n- https://zuul.opendev.org/t/openstack/build/2929fb6cdbe149adb9c29e7fffcfaedc\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/838051\nDepends-On: https://review.opendev.org/c/openstack/devstack-gate/+/843689\nChange-Id: I9c2616153c4a7431c04e8d19780429477d828ddc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8e9176c8769a35f797cc1f23d3cc49a2e77c6233"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8e9176c8769a35f797cc1f23d3cc49a2e77c6233"}]},"branch":"refs/heads/stable/ussuri","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
