)]}'
{"id":"openstack%2Fdesignate~843710","triplet_id":"openstack%2Fdesignate~stable%2Fvictoria~If7a5311291ec9729c9fdee4dc88aedbf30537330","project":"openstack/designate","branch":"stable/victoria","topic":"victoria-pin-tempest","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-05-31 21:36:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If7a5311291ec9729c9fdee4dc88aedbf30537330","subject":"Pin designate-tempest-plugin for stable/victoria jobs","status":"MERGED","created":"2022-05-28 06:29:22.000000000","updated":"2022-05-31 21:41:18.000000000","submitted":"2022-05-31 21:36:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843710-victoria-pin-tempest","meta_rev_id":"c33da5b934157c2855f2180f6947f60bc9a9a5a0","_number":843710,"virtual_id_number":843710,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-31 21:36:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-05-31 16:19:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-05-31 15:44:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-05-31 16:19:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-28 07:27:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-31 15:44:47.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-05-31 16:19:58.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-05-31 19:42:15.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"}],"messages":[{"id":"3c06c798b134fef65fc0dd74731d1c973c85aa03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-28 06:29:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7d219c021ee4ccc768b476f90ca15a53dd79d2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 07:27:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/01dd10e9ef6c4e8385434ad2239fd2ca : FAILURE in 50m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1dfe9a3fd58c4300bb841d5fe1db6d6e : FAILURE in 47m 52s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f9cf794f4fa044a2a9cb60cf1fb82b4a : FAILURE in 22m 54s (non-voting)\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/860f9b2b057b42bf8da336c130d5a651 : FAILURE in 46m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e87089dca2604810967b1dd1191fa4e7 : FAILURE in 45m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0cbdcd3f08e48c2a006b0ed4677aab0 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d4c63b6106442059c787e5c107f3e78 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0218c6c25ff413c87bc03087267c6e1 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/274a2946e9654984aa890d1e80f8f7e7 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f1abf03655642bcb4c118d19ff60ecd : SUCCESS in 7m 39s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/f356808e03134ccc917798bbdc49bf21 : FAILURE in 23m 47s","accounts_in_message":[],"_revision_number":1},{"id":"9e82e961267520b030ac353d15d35c47104f1a08","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-28 16:28:53.000000000","message":"Topic victotria-pin-tempest removed","accounts_in_message":[],"_revision_number":1},{"id":"5092478af02bb82475312992755105f283fcd077","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-28 16:28:56.000000000","message":"Topic set to victoria-pin-tempest","accounts_in_message":[],"_revision_number":1},{"id":"6843719e8bd5418662e37a15827e83b4c9a682f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-28 18:24:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9e4602986f11cfa2d0d4e26e44f58a1d83effe26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-28 18:39:32.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"ad2eb8b8b6880e6b06d941cb8361a8ebcca2de95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-28 18:44:36.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"21b1cc4b730a9cbe9af9c45f4e9992d8dcd5ae0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 19:32:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bd18920d996a42d7abdeb0cbb4bbb0f0 : SUCCESS in 27m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d4473c6eb33244c88657369bef22c183 : SUCCESS in 26m 13s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4ac67d19299545cf90edac29f894919c : SUCCESS in 41m 42s (non-voting)\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/38b063086f644f18bd90ee7f390b662a : SUCCESS in 27m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/47515568778a42228657cf69de2de7dc : SUCCESS in 29m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a167d6322254f32a0da383448476d0b : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/979e163f15a64e2fac9dfafa9c76cb9f : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebafb2de896349bc80a4fd5ce1ebc156 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08c5dcc3b3454dff9ef152a1940d21b9 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9051b36500db4065bc9dda7994e83501 : SUCCESS in 7m 22s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/eb40e9ebc6c249e7be19da927474b637 : SUCCESS in 21m 56s","accounts_in_message":[],"_revision_number":4},{"id":"efaca4a055004a04b524991aa08000d20e1331fe","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-30 20:56:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d522d31ef14a8aa82fe42212602015b09f198cbe","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-31 15:44:47.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5f270ae5ddca47d0e0f0b184298d54065c222865","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-05-31 16:19:58.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"51515b4f9577c821116d7d9b978f4074ab10871a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 16:20:18.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n\nWarning:\n  Change 838053 in project openstack/neutron-tempest-plugin does not share a change queue with 843710 in project openstack/designate","accounts_in_message":[],"_revision_number":4},{"id":"ff4bab36ca2b4104155c3fdd527ed9f012ca1836","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-31 19:42:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"989a928eb9f28613c34beb3c5a2b435c231a52ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 20:40:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bb903c8402144afd899bc140831ed753 : SUCCESS in 28m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b60d37493fb241fab6a26cf8569f94f0 : SUCCESS in 27m 59s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0128fe3654e94b2aab9d53e12c806e54 : SUCCESS in 57m 09s (non-voting)\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5e55782ec77e4e4e9e9b4c3a6c542139 : SUCCESS in 42m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ad5b976a1eb14022b0e19439f4839fd1 : SUCCESS in 30m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/016a938ebf4e469eaa1a0598fb672a03 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d04d6188539e45619e0754662340900e : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6294e524264148e6a642257fb5e4406a : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02cd300112474896bc839cab6414b461 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d9c9679b1574d41b242a54ab0416039 : SUCCESS in 7m 22s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/be35187bf26546d79cbb75537bef72d9 : SUCCESS in 23m 27s","accounts_in_message":[],"_revision_number":4},{"id":"11a21f708ed26dffc586256e3963ddce44ebb0c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 20:41:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8ceeaa457a3b7156b10d4d9d599c21b9772294d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 21:36:08.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b88ddf2042d3476da294a030492b4884 : SUCCESS in 36m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a12232042dfa4819b4a7b512da4a7013 : SUCCESS in 38m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8d7395f409e24118a69793c3be5c6af0 : SUCCESS in 48m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/30bfec5dd305492abcfc937acc7537da : SUCCESS in 42m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91891df77d394559a9141cd49c733a2b : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4269c3d2af4b4e15a3ee8d7356337df1 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1d1d02d68bb4d54ad912fbc9ada70fe : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a40bc2f381341e79102eaa6c2837c92 : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":4},{"id":"295cfdf63a4bc85e148e8e3f64512b03accb1995","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 21:36:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"c33da5b934157c2855f2180f6947f60bc9a9a5a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 21:41:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57a5ae8f8da74fe4adae57fa3c81b9c4 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"61097e4652d13e7b6d305ae029ee3baf00a31a51","revisions":{"af16ed40b46a5fc287e3dd063a6d1254d9093bf2":{"kind":"REWORK","_number":1,"created":"2022-05-28 06:29:22.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/10/843710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/843710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/843710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/843710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/843710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/843710/1"}}},"commit":{"parents":[{"commit":"eee0fcbee357c7f09b16de31cc68f23b765f4462","subject":"Checks for invalid denylist regex patterns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eee0fcbee357c7f09b16de31cc68f23b765f4462"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-28 06:28:35.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-28 06:28:35.000000000","tz":-300},"subject":"DNM: testing Tempest pin in stable/victoria","message":"DNM: testing Tempest pin in stable/victoria\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/843295\nChange-Id: If7a5311291ec9729c9fdee4dc88aedbf30537330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af16ed40b46a5fc287e3dd063a6d1254d9093bf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af16ed40b46a5fc287e3dd063a6d1254d9093bf2"}]},"branch":"refs/heads/stable/victoria"},"5bafd485ec35548ea7aec0e08c3ba65d8a02bcb3":{"kind":"REWORK","_number":2,"created":"2022-05-28 18:24:31.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/10/843710/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/843710/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/843710/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/843710/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/843710/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/843710/2"}}},"commit":{"parents":[{"commit":"eee0fcbee357c7f09b16de31cc68f23b765f4462","subject":"Checks for invalid denylist regex patterns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eee0fcbee357c7f09b16de31cc68f23b765f4462"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-28 06:28:35.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-28 18:23:37.000000000","tz":-300},"subject":"Pin designate-tempest-plugin for stable/victoria jobs","message":"Pin designate-tempest-plugin for stable/victoria jobs\n\nAs stable/victoria are in EM phase, and QA team pinning\ntempest 26.1.0 for stable/victoria\n\n- https://review.opendev.org/c/openstack/devstack/+/843295\n\nTo have compatible designate-tempest-plugin version with Tempest\n26.1.0, we need to pin designate-tempest-plugin 0.9.0 version.\n\nmaster designate-tempest-plugin is not comaptbile with Tempest 26.1.0\n- https://zuul.opendev.org/t/openstack/build/01dd10e9ef6c4e8385434ad2239fd2ca/logs\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/843295\n\nChange-Id: If7a5311291ec9729c9fdee4dc88aedbf30537330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5bafd485ec35548ea7aec0e08c3ba65d8a02bcb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5bafd485ec35548ea7aec0e08c3ba65d8a02bcb3"}]},"branch":"refs/heads/stable/victoria"},"eb1e7f00fc41c9539585519abec4a7cf66bfd5bd":{"kind":"REWORK","_number":3,"created":"2022-05-28 18:39:32.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/10/843710/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/843710/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/843710/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/843710/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/843710/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/843710/3"}}},"commit":{"parents":[{"commit":"eee0fcbee357c7f09b16de31cc68f23b765f4462","subject":"Checks for invalid denylist regex patterns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eee0fcbee357c7f09b16de31cc68f23b765f4462"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-28 06:28:35.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-05-28 18:39:31.000000000","tz":0},"subject":"Pin designate-tempest-plugin for stable/victoria jobs","message":"Pin designate-tempest-plugin for stable/victoria jobs\n\nAs stable/victoria are in EM phase, and QA team pinning\ntempest 26.1.0 for stable/victoria\n\n- https://review.opendev.org/c/openstack/devstack/+/843295\n\nTo have compatible designate-tempest-plugin version with Tempest\n26.1.0, we need to pin designate-tempest-plugin 0.9.0 version.\n\nmaster designate-tempest-plugin is not comaptbile with Tempest 26.1.0\n- https://zuul.opendev.org/t/openstack/build/01dd10e9ef6c4e8385434ad2239fd2ca/logs\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/843295\n\nChange-Id: If7a5311291ec9729c9fdee4dc88aedbf30537330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eb1e7f00fc41c9539585519abec4a7cf66bfd5bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eb1e7f00fc41c9539585519abec4a7cf66bfd5bd"}]},"branch":"refs/heads/stable/victoria"},"61097e4652d13e7b6d305ae029ee3baf00a31a51":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-05-28 18:44:36.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/10/843710/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/843710/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/843710/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/843710/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/843710/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/843710/4"}}},"commit":{"parents":[{"commit":"eee0fcbee357c7f09b16de31cc68f23b765f4462","subject":"Checks for invalid denylist regex patterns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eee0fcbee357c7f09b16de31cc68f23b765f4462"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-28 06:28:35.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-05-28 18:44:36.000000000","tz":0},"subject":"Pin designate-tempest-plugin for stable/victoria jobs","message":"Pin designate-tempest-plugin for stable/victoria jobs\n\nAs stable/victoria are in EM phase, and QA team pinning\ntempest 26.1.0 for stable/victoria\n\n- https://review.opendev.org/c/openstack/devstack/+/843295\n\nTo have compatible designate-tempest-plugin version with Tempest\n26.1.0, we need to pin designate-tempest-plugin 0.9.0 version.\n\nmaster designate-tempest-plugin is not comaptbile with Tempest 26.1.0\n- https://zuul.opendev.org/t/openstack/build/01dd10e9ef6c4e8385434ad2239fd2ca/logs\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/843295\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/838053\n\nChange-Id: If7a5311291ec9729c9fdee4dc88aedbf30537330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/61097e4652d13e7b6d305ae029ee3baf00a31a51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/61097e4652d13e7b6d305ae029ee3baf00a31a51"}]},"branch":"refs/heads/stable/victoria","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
