)]}'
{"id":"openstack%2Fdesignate~843738","triplet_id":"openstack%2Fdesignate~master~I38efa22bb8b0de6fd9a70339e8044fea70080004","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-05-31 03:38:28.000000000","reason":"Erik Olof Gunnar Andersson replied on the change"}},"hashtags":[],"change_id":"I38efa22bb8b0de6fd9a70339e8044fea70080004","subject":"Fixed incorrect quota exception message","status":"MERGED","created":"2022-05-29 03:45:29.000000000","updated":"2022-05-31 04:31:17.000000000","submitted":"2022-05-31 04:29:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843738","meta_rev_id":"449913dc0839620fd52da88985575c678a8395cd","_number":843738,"virtual_id_number":843738,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-31 04:29:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-05-31 03:14:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-31 03:38:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-29 04:39:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-31 03:14:00.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"ea233233f1f96ca1bf7bca9f615f491f6c8496d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-29 03:45:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89699af3add4b142d5827cd3c2037998862ed351","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-29 04:39:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/666f4e870bdb4335b40d890d350b6fa5 : SUCCESS in 32m 22s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/142c0873dda64cb3990d47e45b0aa022 : SUCCESS in 40m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f701164b8e443af850c9a4299de180f : SUCCESS in 33m 37s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/5a8eeef34b024c1b8ad06cd9db4de5df : SUCCESS in 33m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8f5cc8f4896f431a97c145d1826b3829 : SUCCESS in 29m 56s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/528dc2147d0c4588b5588e51ca08e96d : SUCCESS in 50m 33s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bce91e37394145eb85952696db296807 : SUCCESS in 46m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0c07f973d9494daa943085ecd597706a : SUCCESS in 29m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c338143079804559bb97207fadd8a970 : SUCCESS in 30m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ec8fc467d7147c9a05787b16d1f161f : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c3129c42bfc41d7a57997e0e2b3403d : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ee2c816efb54bc4949f7b8baeaa4b79 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8296826bffee4138b7f6a002f591dcad : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8658eeea0b50425a852f432265cce9ad : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58b8bb4a39b343169c21add09b734b3f : SUCCESS in 6m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/edbed797ec5b4e3abf8e50b82a12705f : SUCCESS in 34m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/db481f841fec40f6a242dfcfa51456c9 : SUCCESS in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2d3d73363aa1ab3786994d91d82da22fb7b64b77","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-05-31 03:14:00.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"522e365a7fd6a5aa716e685cc2da3d8854330a2e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-31 03:38:28.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8a5c65be405d4709cef7767a32e16fca457d857b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-05-31 03:38:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"717a21a8cf81054eb63b576e00cc4167b9d97d9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 03:39:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3baf079a560cf6cde02b7523e1c59296effce9a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 04:29:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/20e00a7b60e545eb8fc3a923c24da033 : SUCCESS in 34m 20s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/564f7b9a5e2f42198775112eb5e80355 : SUCCESS in 33m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8210f56654724cd786ed3a41811a35b6 : SUCCESS in 30m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/74555a152c59482fa7b961028e83e306 : SUCCESS in 48m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/65f9cc77b9a54b9ab9fce54d47ac65d9 : SUCCESS in 29m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3604269a90b94a0f9404199bbe40ac1a : SUCCESS in 34m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe48b4d9f5dc4f97bd6dd3895c1d192b : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97d334b674fa4f578c70a9af079701dd : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/299b9dd793694881a266d4428fc76baf : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c9df44313f5456cbc51517eb9708b32 : SUCCESS in 6m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/30669b603c184b8bb8d3d4c083f81ad6 : SUCCESS in 31m 26s","accounts_in_message":[],"_revision_number":1},{"id":"23b3c0e15c8367b78e354e6fdf3fc22a0ea5baa7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 04:29:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"449913dc0839620fd52da88985575c678a8395cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 04:31:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faeb64d4cadb4db98c7f6d999cfc99be : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"291558c380232b726be0253fdee1f88902522beb","revisions":{"291558c380232b726be0253fdee1f88902522beb":{"kind":"REWORK","_number":1,"created":"2022-05-29 03:45:29.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/38/843738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/843738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/843738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/843738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/843738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/843738/1"}}},"commit":{"parents":[{"commit":"ca74e78687dc69479a7f051b0040c9767ae5ca0c","subject":"Merge \"Update zuul queue configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ca74e78687dc69479a7f051b0040c9767ae5ca0c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-29 03:44:50.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-05-29 03:44:50.000000000","tz":-420},"subject":"Fixed incorrect quota exception message","message":"Fixed incorrect quota exception message\n\nChange-Id: I38efa22bb8b0de6fd9a70339e8044fea70080004\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/291558c380232b726be0253fdee1f88902522beb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/291558c380232b726be0253fdee1f88902522beb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
