)]}'
{"id":"openstack%2Fdesignate~844088","triplet_id":"openstack%2Fdesignate~stable%2Fxena~I7876b4c1e2c800b654ca750211ee2e58f3ea4309","project":"openstack/designate","branch":"stable/xena","topic":"bug/1926429-stable/yoga-stable/xena","attention_set":{},"removed_from_attention_set":{"5572":{"account":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"last_update":"2022-06-06 22:40:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7876b4c1e2c800b654ca750211ee2e58f3ea4309","subject":"Minimum TTL value allowed is zero","status":"MERGED","created":"2022-05-31 23:03:47.000000000","updated":"2022-06-06 22:41:53.000000000","submitted":"2022-06-06 22:40:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844088-bug/1926429-stable/yoga-stable/xena","cherry_pick_of_change":844082,"cherry_pick_of_patch_set":1,"meta_rev_id":"0625af46fa26884b329e48d3ed6707666d1ea234","_number":844088,"virtual_id_number":844088,"owner":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-06 22:40:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-06-04 20:46:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-06 21:40:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-06 21:40:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-31 23:55:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-04 20:46:55.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-06-06 21:40:55.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"51bc360e272463387f01adbfa5da2a8eb6644887","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-05-31 23:03:47.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"615ec46f37e03ec5e3daa1a9f666f5b3fc055bbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 23:55:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/86f36b126aea453c9e794b3d6731e038 : SUCCESS in 44m 10s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/1cee592268a3411aae66bed71955696c : FAILURE in 24m 33s (non-voting)\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b427269c6ac7457e92acb1968e31eb16 : SUCCESS in 31m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/15ac3f5c818446bebe9c4e22ed49ee73 : SUCCESS in 46m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/367f8db610034b7b8e4f69c37303079c : SUCCESS in 30m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fe0f0a0ecd0d4db0999c8e3c77cdd92e : SUCCESS in 32m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39842c0fd7064b79ac39042fd1f7c582 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4236327fd41d41f0858055176c0f3eda : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/333981c291904056990309884f9c1cce : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8a33a72f323480aab055e12c7bf79c3 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e0fdc27979c49b98c15fa5a5c42477a : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/694e2d6d90f3405a818041a1086e15da : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8893e0f731af433fac0f41d5b551f535 : SUCCESS in 3m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6fd325f8642041d9b80b4199471e5095 : SUCCESS in 27m 41s","accounts_in_message":[],"_revision_number":1},{"id":"0f3f85fa2b0aa2ca40dbd6d82aaadf20263cc659","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-04 20:46:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3028d4d57f090ece16128fcf6e6f0f29eaa5d08","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-06 21:40:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4c7ce20102927a779e6045a13bc6220bdc339f77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 21:41:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c277599bd53360c19273776167fddbf2f28e01d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 22:40:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0d60a468f75e4b84bfdb187a239c7207 : SUCCESS in 32m 21s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/7ac8d06376f244b88c6a37e6029f36da : SUCCESS in 36m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3ff7108c909043d983954cbc4189ee8b : SUCCESS in 37m 48s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a1e6340e45bf41edb84a5287d86e2460 : SUCCESS in 57m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a49d02ec419845fdae14ea6f11586707 : SUCCESS in 36m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3f177b6d858c4e0c9394e9f5166dc1d4 : SUCCESS in 39m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0de1e3b990c34c25b1b126f2e7b0be76 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5d352353fb04b9dbdebe3522ee3929d : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75794f670bda4e3daf453e3847e7368b : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4da3242e4af54fdd9c1fdb149be7eedf : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0145ecb0c2c844159bbed4c79177ef37 : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":1},{"id":"08a5faed6aec1b5462a164c0773d6da200b15d13","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 22:40:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0625af46fa26884b329e48d3ed6707666d1ea234","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 22:41:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e49fa2a6ad7746c097f127305e61915f : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31f712f6877a4923a572cd5418f3a963 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"264ac571d5ccb484e8071d2bec575992296dfb2a","revisions":{"264ac571d5ccb484e8071d2bec575992296dfb2a":{"kind":"REWORK","_number":1,"created":"2022-05-31 23:03:47.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/88/844088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/88/844088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/844088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/844088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/844088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/88/844088/1"}}},"commit":{"parents":[{"commit":"c78db47c746cab2fc505bd60f11f05a08eb94120","subject":"Update zuul queue configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c78db47c746cab2fc505bd60f11f05a08eb94120"}]}],"author":{"name":"dekehn","email":"dekehn@gmail.com","date":"2022-05-02 23:03:49.000000000","tz":0},"committer":{"name":"Don Kehn","email":"dekehn@gmail.com","date":"2022-05-31 23:03:47.000000000","tz":0},"subject":"Minimum TTL value allowed is zero","message":"Minimum TTL value allowed is zero\n\nAccording to ITEF https://tools.ietf.org/html/rfc2181#section-8\nthe definition of the ttl value is unsigned integer and can have a\nminimum value of 0. This patch changes the minimum value of 1 to allow\nfor 0. Unit test have also been modified accordingly.\n\nCloses-Bug: #1926429\nChange-Id: I7876b4c1e2c800b654ca750211ee2e58f3ea4309\n(cherry picked from commit 6ac659d241c65c245d8a090768ac7559c32a13f1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/264ac571d5ccb484e8071d2bec575992296dfb2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/264ac571d5ccb484e8071d2bec575992296dfb2a"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
