)]}'
{"id":"openstack%2Fdesignate~844208","triplet_id":"openstack%2Fdesignate~stable%2Fxena~I4522796ac056630c63798830e8f8d92628f14a37","project":"openstack/designate","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-06-03 15:22:02.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"I4522796ac056630c63798830e8f8d92628f14a37","subject":"Fix delete zone transfer request with scoped token","status":"MERGED","created":"2022-06-01 22:22:31.000000000","updated":"2022-06-03 17:23:06.000000000","submitted":"2022-06-03 17:21:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844208","cherry_pick_of_change":836340,"cherry_pick_of_patch_set":1,"meta_rev_id":"ee0709e891999058b64edf78e03292e7b286b248","_number":844208,"virtual_id_number":844208,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-03 17:21:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-06-02 19:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-02 18:25:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2022-06-02 19:39:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-01 23:13:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-02 18:25:42.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"8a0605bf51a84b1688e72a6b44905c1f44058e6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-01 22:22:31.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"9b745e1287fec8cd24bce3aa24661f14e5beab68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 23:13:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2f50efae84014db48cfe96c96ae2a809 : SUCCESS in 30m 00s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/32bc6ae891874d34a6ebd99756f437a7 : FAILURE in 35m 09s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/cad87ddfec5944c0accd7faa30278b21 : SUCCESS in 33m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4bf14d29eb2f43a4b5ff59c0a5637106 : SUCCESS in 39m 11s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3e91a19412a74138acf0bf8e69499ed9 : SUCCESS in 49m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7b96d99fe65b42f6b4d97276b29c879c : FAILURE in 35m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ec7053a3cf6b4590bd2e0c14a3291aa0 : SUCCESS in 33m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac0e4ea691024f18a51aee291a0c7ea4 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82c66f4a14574866b3b066a7971c3ca7 : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7a2b54cce554d3fa7e6daad4d8a9953 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f517d7e01485449a9858fb4309526832 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0aa09dfa36174c509c2d7ff586ba4b84 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c0be2990bd04e60a39abf2ff3451b2e : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9f554ddbdad442fb5c6ff84de73037b : SUCCESS in 3m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b1837d92a9b74f319af72330908fb661 : SUCCESS in 29m 48s","accounts_in_message":[],"_revision_number":1},{"id":"8ebe6eadbbeaaf447ca536a13a0e2cd75359bee6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-02 18:25:42.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"00cf28c43a9eb3c9fafb45bd6c6e4a417642e12b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 19:21:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/84db498a4c484e5db669b815dd715de4 : SUCCESS in 43m 43s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/9934bd26c4fd470f8ebf0b7c069f6d73 : FAILURE in 30m 17s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/cb0528a0840c4fd9809e4e919a3b8f0b : SUCCESS in 34m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/06e6c7d49eab407aba809c5738afccc5 : SUCCESS in 30m 08s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ea0837c72bd34e7ca5f238425cc7071c : SUCCESS in 52m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7617503477394fa5991974039c431786 : SUCCESS in 30m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3c1d5973b6134bd6b2163ab635949cb6 : SUCCESS in 28m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c111ccc47b74847b0d88ebaaf93fa7f : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fe14a49958940d1956aec43d1d17cc8 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79331443c1da4a63bc732b3977faf362 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bfd0f88a393457085d7e06ea0529802 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6583109ab0ab4ff6a016e5cd21fb46a2 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff68ec59c5b440b38e010a6c1b01763c : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb55d68f577349dc9b821e8607c2a009 : SUCCESS in 4m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5068dbb379094f0a800f139a46105efd : SUCCESS in 33m 58s","accounts_in_message":[],"_revision_number":1},{"id":"094f7c0cc1436a7c02fc465fd177281bc2f721fc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-02 19:39:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ce6766cba944a4407527ccb55a64777d0b8fa49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 19:39:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01a2a1a1d887ea86363815d4ea5c83e9df0a0d3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 20:44:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/38cf2a68b3a2416a926e56568b8df7f2 : SUCCESS in 30m 27s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/be70dad737b048a58932f0ea80e1ca89 : SUCCESS in 33m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/64bbc6647f4045b19cbe0f77401336af : SUCCESS in 29m 53s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2a7a30dda2b74667933745df63b64c55 : SUCCESS in 1h 03m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/218bed57c7e84412a3a870edc07ec476 : SUCCESS in 32m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/52e1741361e64ad3a5a63a8a2a699527 : FAILURE in 29m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/205b6670118e4ad885f228e7cf2a5b19 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04af645a40fe40d9a3e6ed44558ef051 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/873ca3168169463bb988dc02777bb48f : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f35377479c94404cbdf5cd3d99ef86c9 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f469ebf40f314724bcf3542241fcb6b9 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"4b8d2aaa3f89fd9123f4187ecf4a7a53d5d975e1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-03 04:40:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f34b9a43058ac3b20f0beb02ab51c9349087aa2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 05:26:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cda053bf73f24a3ea1322ae4ba95b4d9 : SUCCESS in 34m 32s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/f8caf95bdd304a98a6b2bc5bd192926a : FAILURE in 30m 19s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/dc6525f85106484e85c2720213da2f62 : SUCCESS in 30m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cb75a20dc948425fa853e07170fed7c4 : SUCCESS in 29m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/85f80b8a4ed14ba789f2339522846e57 : SUCCESS in 44m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ef1c0bf4eafd4ba68c63e244a79d6838 : SUCCESS in 28m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7a42457084da4e7a87372658d3d04b30 : SUCCESS in 29m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdccde9af0a44b069d9bc46f3e64dd27 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77200632783d41bfb32e27945f6c172b : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52e38e5e7d7e4d168f1f7d048691da85 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f409f03df704c48a67a467e082c3a2b : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/985789db85bb484e96085384cdfbe3ff : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29a26a2c7b64483fb42b6191aa08bc71 : SUCCESS in 8m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cfcb7a49dde490a9dcde88ec640904d : SUCCESS in 3m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/82d6881892774ccd87c1260e9acce3ae : SUCCESS in 34m 55s","accounts_in_message":[],"_revision_number":1},{"id":"8b59fe602aea047a591c83dbfc3ca68a37d8a14c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 05:27:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c7fe057492ec4e07735e1f3de4c4eea9d5dee6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 06:45:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d914f4e6444e437e8d9b54866a187cb5 : SUCCESS in 53m 16s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/1d3f496f48b743669d01e519fa9d4cd0 : SUCCESS in 43m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/614e8ef1e240410a8aed19505c74ec6d : SUCCESS in 43m 16s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6bd02424dcbe4b93a50bea405ed7e5d4 : SUCCESS in 1h 13m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/aa59b2bf3bae4150ba104379e92dbc4a : SUCCESS in 45m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/99e3ea4e5d8b4344a4f810884c4c96ee : FAILURE in 48m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e90176c67d74246b6bcf9b0170bc5a8 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/121f29898b8d4745882483ce3633b721 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f8dccf97ee54a79937b058da03f0aec : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9ff8f02b5554043bcba4d9120c35ef5 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e5d391032284c98bd38cb11d2b354d9 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1},{"id":"1d51d819cbef045f360f52841b9832ba127f9994","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-06-03 06:48:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bab7df44a1c5337f242710132a355da60cdd9a44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 07:48:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/42f07488020e4a3daa2474dae0749626 : FAILURE in 39m 13s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/cb250164d1604d44a335f6b5f1ad1dc6 : FAILURE in 28m 34s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/35a044c371504ccbbf8cbee961cce33e : SUCCESS in 32m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/30897b9f7b6344dd9063697c9a2d1d00 : SUCCESS in 37m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/82e800d8df76413f9526014426bd20cf : SUCCESS in 58m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e8f04b7ffbcd4600879404476ce5c03b : SUCCESS in 33m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9725545e99a346f982ace9f4976f3808 : SUCCESS in 35m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/469ea4a558844577844cf18ce5d47041 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec954c02f2624d85975460c78638535d : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c70c90839a934a3a890e9e127d55fe19 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fd103e4b4fb4a9d921c0a3f1b768c72 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f84a4c5648c495097f2f40e5cc4a1c5 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dd1b9f6b4f84d2da6607e5dd972c000 : SUCCESS in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3ddd27192c342da99b444293e4f909a : SUCCESS in 5m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/48eb053f0d8344bdb63c9331f99563e9 : SUCCESS in 35m 39s","accounts_in_message":[],"_revision_number":1},{"id":"e752f0c08a8c847dd90a547227656e38d26bcee8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-03 15:22:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cbd97e1d456413fce7c63f09932b09b5d0ec3c42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 16:20:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d04d31a335914360bd9283b2eddd7fd6 : SUCCESS in 49m 19s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/8b5dc8264ca541a58b586139fb1ef5fa : FAILURE in 32m 02s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/599b8f08e43043659c2a502e7f7e5c44 : SUCCESS in 43m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d201a23d0a244d4c9565affc3fa3be98 : SUCCESS in 38m 08s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/86146f2b3c224355af76d50a6fa61064 : SUCCESS in 57m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/63d5882fa6fd48dba22e5588b2fb186c : SUCCESS in 47m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/eff1c2d8a190487890e794655ece0295 : SUCCESS in 38m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74ce6a5fe25d43708cff2e42ede2f114 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22bd20ca038440d9a9d18ccca9f4012c : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f1eaf24bcad457480f3a62edfaa01a0 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfe77c0333df429a9b50b525f36a7d13 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f44e6057aac47d394f6c90d89551547 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e66a8dd45e8948fab320f656d1109ef7 : SUCCESS in 8m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7424777e132b4055b044c58b04442443 : SUCCESS in 6m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2f9c89083939427aba78022d16891577 : SUCCESS in 29m 01s","accounts_in_message":[],"_revision_number":1},{"id":"28ad26fc73e1db5d86dbd440cae245c0e4dd67f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 16:21:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b1ff34ea94a64b92c1dff4634d661038bd4d0a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 17:21:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/812d4de9d0274cadbafb1f7687f1241f : SUCCESS in 36m 05s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/9ed5bba2c4644e489cedc282097b6ac4 : SUCCESS in 37m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c900df98b1fd46ddba18e2655fdfa4a5 : SUCCESS in 38m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1351ae3af5bd461dbe03492f2ead65dc : SUCCESS in 58m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/10837b4055854a3eb83eb9a1828cee2b : SUCCESS in 38m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/473a89396cae47f6ab47ecefd5df006f : SUCCESS in 39m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78d4a555d26d45abb4e28b9bcf38eebc : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9679fd24589f4fb996d7498990d4c556 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6658ead0da14045afc23beb07285e2b : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bf4313f7f7545e299476f857837850b : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/528ad8c3e2774a4ea02c42d6c964b012 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"9ef4e7e4094d89707a998188f486faab44f85568","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 17:21:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ee0709e891999058b64edf78e03292e7b286b248","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 17:23:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aab42909d514493baee3cca11adf2336 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c3e0aa888dc4ea2b4352551deaa7f42 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"63bc0c1db93899273627462cb177c1481502e9fc","revisions":{"63bc0c1db93899273627462cb177c1481502e9fc":{"kind":"REWORK","_number":1,"created":"2022-06-01 22:22:31.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/08/844208/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/844208/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/844208/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/844208/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/844208/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/844208/1"}}},"commit":{"parents":[{"commit":"6a4b524a7fbf9c94bcb1c192e4304bcecd8693ae","subject":"Merge \"Fix support for scoped tokens and default roles\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6a4b524a7fbf9c94bcb1c192e4304bcecd8693ae"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-01 22:25:07.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-06-01 22:22:31.000000000","tz":0},"subject":"Fix delete zone transfer request with scoped token","message":"Fix delete zone transfer request with scoped token\n\nThis patch fixes an issue were deleting a zone transfer request with a scoped\ntoken may fail.\n\nChange-Id: I4522796ac056630c63798830e8f8d92628f14a37\n(cherry picked from commit 6f847aeb5a309eb6e70dee58057fef01886573f6)\n(cherry picked from commit 276a9fbe89f84dd43ed0495a5445eb9fd4cc56ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/63bc0c1db93899273627462cb177c1481502e9fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/63bc0c1db93899273627462cb177c1481502e9fc"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
